)]}'
{"id":"openstack%2Fbarbican~705686","triplet_id":"openstack%2Fbarbican~master~Ic1088556c95ff122d422f06a9cfd0549303217eb","project":"openstack/barbican","branch":"master","topic":"add-containers-undercloud-gate","hashtags":[],"change_id":"Ic1088556c95ff122d422f06a9cfd0549303217eb","subject":"Add undercloud-containers to gate","status":"MERGED","created":"2020-02-04 14:01:27.000000000","updated":"2020-04-28 18:15:02.000000000","submitted":"2020-04-28 18:11:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705686-1588097466594-bde3a662","meta_rev_id":"8ab228a167c6ce8e27e274dd12b6c79042412f15","_number":705686,"virtual_id_number":705686,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-28 18:11:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"recommended":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2020-04-28 04:04:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2020-04-28 17:32:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2020-04-28 17:32:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 13:08:10.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-04-28 17:32:49.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2020-04-28 18:11:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ef147789e555852576ba6b82a0cbb17f0a3b30fc","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-02-04 14:01:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffa68ffe8d8ea4cf36f4ad36f07817d6978a25c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 16:14:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb840c19a9ba4544ba20cc933fd4c1a7 : SUCCESS in 7m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4e7e3bf28a9414dacbcc4699308ebe9 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01beb418bbc34f93ac82729c1cef8395 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f17068e6bdf4632a79c2f519261c695 : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf20a28fab7546be9f21cfbf56535ec4 : SUCCESS in 9m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04bc06467dd8449b94934df6cdb85496 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e118f2d0e54f4aff9c92e6583a53d0b2 : SUCCESS in 6m 54s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/5830bf8f18774119b655f6fef11739d9 : SUCCESS in 44m 37s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/51e047f52c884e70a18b5661aa192a34 : FAILURE in 15m 17s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/6760fdc4c0d1417e86fd7c928eb050f2 : FAILURE in 17m 30s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/a31ed127ffd14fed996f9d8a33c16c79 : FAILURE in 22m 46s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/579d7fff3d674db88bde5a5aac4c4f14 : FAILURE in 17m 19s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/d8069b7632f748d39cdbe6d4cc4e06a0 : FAILURE in 52m 28s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51317cddd802408d9c7d08727a3e64f3 : SUCCESS in 44m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/80d8d1b92eb248129937988d2d5f5704 : SUCCESS in 1h 09m 35s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/43348e43a70146c5af25991be9dbebd2 : SUCCESS in 2h 07m 45s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cf580dae2fa546c3a9a53c1565a6ca71 : SUCCESS in 2h 08m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62425b6f2279479bc512f839b4c5b9a818b2f498","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-02-06 13:41:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"83bda37964ffc518b2288fe41f407a91d60da277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 15:25:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3db719fabf9432d920d6f6c8acc1b4e : SUCCESS in 7m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb03c6b80d9e40d09425205655b47e1b : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/372fa0e8b9c7496eb350acc1d164e8db : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fce747e0be704e2495f6d3883f13535d : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab9c73f9f8474538830b646d541ea075 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a800f96f9294bb7aa56ff71258a7d86 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf65f60a46a459091f5e93b6ced2171 : SUCCESS in 8m 33s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/8f72da5abc7141bdb0fb80db31f6d1b1 : SUCCESS in 50m 02s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/4fd50047d8e541dfbd770fbc70bf04de : FAILURE in 19m 10s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/3564705499994d9fa98f9624f1ae6729 : FAILURE in 17m 15s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/2c9f9941135142358036b38c2a2d817f : FAILURE in 22m 45s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/e4baff2059ab4ae39d13adb4035f7731 : FAILURE in 15m 29s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/5e5c058a3707457f9832e82ee59f113d : FAILURE in 51m 18s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9faf13f98e649248aedeb92d6aa5da0 : SUCCESS in 48m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/52f9f10ed16548f4ad938edc431bd666 : SUCCESS in 1h 06m 56s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/613a9cf6767940cc9a922c1eeb0e707c : FAILURE in 1h 39m 36s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/64f23426b5e9434184b8b018aff79658 : FAILURE in 1h 30m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04329345db0d199fe249c2805a32777615f8c79b","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-02-11 13:08:10.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fb6616f741907fb8cdb09872dfba1e31068b3672","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-02-11 13:08:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a1be1f07c6733715b52cb3839acb17f244eb220e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 13:59:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3efe845345c4410b6dc3c034b7d2350 : SUCCESS in 10m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3d497bd27224fa2a310c143600991f0 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70bec73b1b6b4b2eabb746ef35398190 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30d5712a81024d709db7c38b818c16a7 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47aee11d8a2e4e75a4c8dcd16347987b : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2992dad685c4863a25286b0486b0dda : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d53f5afc614b54b37c293179fa1ba4 : SUCCESS in 6m 14s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/085b02cb7a5542c3929a09b4b6a32db0 : SUCCESS in 43m 37s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/43f2fbc340974b35a7f3c98b04432b9d : FAILURE in 14m 15s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/f43b34897d7d43e1b9473cdff6c82dbd : FAILURE in 11m 21s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/4b9e0982c7624d14b43e314684cfccd5 : FAILURE in 12m 43s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/040d9b97476b479ea66551c9b4970c4b : FAILURE in 12m 04s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/cfcee3c55ce34ea49e25cb1e33f9e9bc : FAILURE in 37m 15s\n- barbican-simple-crypto-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d6a4eb3b15c4fedbfd48998033614bf : SUCCESS in 35m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d8f41fb8ebd0455a98ecbff70e7bcde0 : SUCCESS in 50m 05s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/1090d1e229fc4169b2d3fde9acaa494d : FAILURE in 5m 48s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0dd52f1deedb47ed9eed7c276bcf2720 : FAILURE in 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4d28546c8ac1cceb0b8474dd6d78644c4df58e0","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-04-28 04:04:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1747b74dc105c04a5888912154eb49c5469ba2f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 05:18:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9404667783c542dfb745df1e13675940 : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3a2c8e92bd44991bc544bdbd6c54f3e : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dc78595d2a348f49882c98937f58ba7 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa7feba26bdc41ba9de7b558b911030d : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7c5b9ce3ead4be0ae5db6461d018fb2 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f9b3260a75540a3af7c8063c3753670 : FAILURE in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a18688b2394f77baba3895038629e1 : SUCCESS in 8m 35s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/d081e8db0ce54d02941432133dc9141e : SUCCESS in 41m 41s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/bd5852829262411682132616adaaf3f2 : FAILURE in 43m 57s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/9c57bd25b9bd40d1a056c8c21761310d : SUCCESS in 32m 56s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/408c6fbb447146c7b569995f78034694 : SUCCESS in 1h 12m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa57782d0d674f0282599fdc67fefdc5 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":2},{"id":"65fe1ef72aa33a4c2a853be0fee63b643a421e2d","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-04-28 17:32:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c018d2e7ae27b094d55299274cc274251ae590aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:33:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a46f0cc83cb761fc425287ce299d6358a1bdb10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 18:11:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65499ff8b73c450eb070a8d70020f5b9 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3527d2e57c1c4ad8a301e588b915c188 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca3fe664bbb440faa376ad9bfa971d1d : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c33c4cacfc204fc8b23ba75765fea835 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f957bff47fe4539a92f3c27d4919d79 : SUCCESS in 14m 06s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/4a5872302e2c4a8295e5613b5ead37e1 : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":2},{"id":"341c4a22c52e5d0edafc0999550c8fdbb0db4d85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 18:11:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cfaa1d3e2d6c762dac71d0c33103afed3a6c39e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 18:15:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b289feedb9b041cbb611d3e44c676e10 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e32856016c2714ea4497b4c911d28bc4f2110a3f","revisions":{"31991896ebbf278517c0f6ec3725e752ee7fb647":{"kind":"REWORK","_number":1,"created":"2020-02-04 14:01:27.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/86/705686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/86/705686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/86/705686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/86/705686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/86/705686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/86/705686/1"}}},"commit":{"parents":[{"commit":"7b14d983e0dce6dcffe9781b05c52335b8203fc7","subject":"Revert Secret Consumers API change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/7b14d983e0dce6dcffe9781b05c52335b8203fc7"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-02-04 14:00:51.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-02-04 14:00:51.000000000","tz":-360},"subject":"Add undercloud-containers to gate","message":"Add undercloud-containers to gate\n\nThis patch adds a TripleO job to our gate to make sure\nwe don\u0027t break TripleO with API changes again.\n\nChange-Id: Ic1088556c95ff122d422f06a9cfd0549303217eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/31991896ebbf278517c0f6ec3725e752ee7fb647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/31991896ebbf278517c0f6ec3725e752ee7fb647"}]},"branch":"refs/heads/master"},"e32856016c2714ea4497b4c911d28bc4f2110a3f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-28 04:04:07.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/86/705686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/86/705686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/86/705686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/86/705686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/86/705686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/86/705686/2"}}},"commit":{"parents":[{"commit":"b8266ef40261fe034817d711440f0337b040c38d","subject":"Use Zuulv3 devstack jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b8266ef40261fe034817d711440f0337b040c38d"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-02-04 14:00:51.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-04-28 04:04:07.000000000","tz":0},"subject":"Add undercloud-containers to gate","message":"Add undercloud-containers to gate\n\nThis patch adds a TripleO job to our gate to make sure\nwe don\u0027t break TripleO with API changes again.\n\nChange-Id: Ic1088556c95ff122d422f06a9cfd0549303217eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/e32856016c2714ea4497b4c911d28bc4f2110a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/e32856016c2714ea4497b4c911d28bc4f2110a3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
