)]}'
{"id":"openstack%2Fbarbican~758255","triplet_id":"openstack%2Fbarbican~master~I115cf1a7006a6c85f37c5e50ded13134a3dfd1a3","project":"openstack/barbican","branch":"master","hashtags":[],"change_id":"I115cf1a7006a6c85f37c5e50ded13134a3dfd1a3","subject":"Use serial number or label for PKCS#11 tokens","status":"MERGED","created":"2020-10-14 21:15:59.000000000","updated":"2020-11-02 17:15:41.000000000","submitted":"2020-10-21 14:56:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758255-1603292211835-072a3a1b","meta_rev_id":"737e8e83656cb8efdf5722acb5e3acacaf3af4a3","_number":758255,"virtual_id_number":758255,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-21 14:56:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2020-10-21 13:56:06.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":2,"date":"2020-10-19 21:26:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2020-10-21 13:56:06.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 00:47:29.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2020-10-15 00:47:43.000000000","updated_by":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"reviewer":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"state":"REVIEWER"},{"updated":"2020-10-19 21:26:38.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2020-10-21 13:56:06.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-10-21 14:56:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd1ff4918a1e39f16edf5b493d54cfe8ea880839","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-14 21:15:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c60c475cd44ca8d05300afd785378c5996955d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 22:34:03.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6bbabeaac4048c1abd09c1508a4457d : FAILURE in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0454613000204cc59afdb2de887217aa : FAILURE in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/898a6112d9114324a2954bd5f10be751 : FAILURE in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8339259782ab4abc97c69b51c7e67428 : FAILURE in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbbe01318cf34cfc9ccc595aa6bbff21 : SUCCESS in 7m 12s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/eb6dd6891e774fba9ef06c960cba029c : SUCCESS in 40m 21s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/7e186efea52f40ae8ead7c7b30cfaf44 : POST_FAILURE in 19m 10s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/4e72f3ab1f4f4e5db7a1beb1cbd5956c : SUCCESS in 32m 00s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/c34e3c29cfb64183aad1af998921e67a : SUCCESS in 53m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fd20b07b161544468613a505c710a9ac : SUCCESS in 45m 13s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/f5bd261901394c1eb043624ff9785347 : FAILURE in 38m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7242b63fdfa643ce82a40f2efc5a2015 : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d3fd7980e9429f33a85570c86a64fbb19f3b3b90","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-15 13:36:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb5bb2265d85e59151de297ee229500ff302257c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 15:00:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e4ca0a99e6b43b59408b49cde4d6ad6 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcec65793b4e4394a1afd15f9ba0a73b : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/338781907d80468c85bcf3f04fa17581 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77d793017323452291ccc515fde4220e : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a28cf4bdb5164d148fd59b7a1d0964fa : SUCCESS in 7m 14s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/331e401a823047868aa7c6daa3eb1457 : SUCCESS in 39m 50s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/a12f141d395241579bd01d82a1ce0f78 : POST_FAILURE in 17m 36s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/a58d7fc2f9f44b509d14f15e3d2e6914 : SUCCESS in 34m 13s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/164a828cbe4e4428b53f1b64664aea2a : SUCCESS in 1h 13m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5d125c6fad894fb6bb1857737382eed8 : SUCCESS in 41m 55s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/ee1ef93529f54f56b2a3f795b7582758 : SUCCESS in 54m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6a2b778d6e96409c99a0bc6504c7f028 : SUCCESS in 45m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d6207e26c18292c94b2d1b84114da84332e13c48","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-15 18:10:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"607fabadf8c0c038820c88a55c991632e7710ebc","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-15 18:14:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"deb16bfa4cf56e31c623c05756b3778ec5e0d537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 19:33:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/767342822ea94655a5ac3c45273235e2 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28af4320daaf46c395c9ba8e772c9bd2 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e70206cae2346028ba0f4f2e3aa3704 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7ff318a87654cfabf34eef01eb542f6 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26a4b522072343cca6fa96aa181417a0 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/666feb4d3c844f51955a2c0d73380e09 : SUCCESS in 6m 25s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/d0e00d66ef444c54945db33548c38383 : SUCCESS in 33m 26s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/7dbba68ac7ff462f9bbadeea27bb0265 : POST_FAILURE in 16m 06s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/279e27cf36664deda4f056d3cd43559b : SUCCESS in 42m 36s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/a8eef98b2b7f4eaca042453ceaec6f3d : SUCCESS in 1h 04m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fbf47bae2f2a4898b9c749cbda549d6d : SUCCESS in 43m 28s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/531d5c0e544a468faa87f5fdff0af20b : FAILURE in 34m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4509b06c9b8c43c7ade677052243a3de : SUCCESS in 38m 46s","accounts_in_message":[],"_revision_number":4},{"id":"8d344a19c1f49969c9bc1d9d4be17e9df9aff020","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-15 19:45:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e06c3983cfc0a67ace25cfd2fbf4c4f9902e9b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 20:59:24.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a50c8b5fe304ffc988697e8aecf456e : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1e3b7635a394c85a05cfcb6a7a7cbf1 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b5a6eed414346198fbc8770945a09ba : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f20e06574cd4fb39cc9561d42512869 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8db06b6a954220ada8d803ae6e1b89 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/664853c4d26a4f828f35507decfe7eb7 : SUCCESS in 5m 46s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/785ce5e7edf24790ab492baf80c8ad54 : SUCCESS in 58m 35s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/4ebcc171620f46379484004e81c23b0a : POST_FAILURE in 28m 25s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/2a97e0251b094c4fa18d23021f491a2a : SUCCESS in 41m 59s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/f513daec6e1a4805b878adef020399d3 : SUCCESS in 1h 04m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b51d6a98b1c84d3e9e4adcff32aa951e : FAILURE in 43m 30s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/5893522c9b51424c8d3f065722fc97ce : SUCCESS in 39m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7a12de3dcf254aa38dbb23f56f045008 : SUCCESS in 47m 33s","accounts_in_message":[],"_revision_number":4},{"id":"7411c876213c21fbc5b4a827730d28a669e5fab9","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-19 12:10:46.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2edc32b62e73613e3e7a9121e83cd8db4ed15915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 13:26:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f20556a994242be9d7ff5bdd2395b20 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be109b95e9247b48806301a02ce4ed1 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f959b7e20c84f958a494f6a759ae24e : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1287dc4d1c9443b9f2ab6fff6757c4d : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8343ed275d0e40d3b3ab43577bbf0f65 : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15aeb54f84a1406198ecd482c9df5655 : SUCCESS in 6m 49s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/d30e58b76a34451ebc1171956e98872c : SUCCESS in 36m 27s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/c89da6163d73490781dc7d744f65dee4 : POST_FAILURE in 22m 47s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/499b5f1188554ca2bdcddb911c52db92 : SUCCESS in 44m 13s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/522f1b83ce484438a943b2d07f75cf60 : SUCCESS in 1h 08m 11s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2e012532a52a40dfbdba4713023f0fd2 : SUCCESS in 49m 02s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d547c3a0b179459eb1311765011f8222 : SUCCESS in 42m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/51b1180b24ac484590a683423c8fda64 : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":4},{"id":"a6cb569a1c0e6856462143381f6bbe5641692b4f","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-19 15:45:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e7da2da0ca44075fcdae2ca477ef8707be74f2d7","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-10-19 19:17:09.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nWe should also document the new params not just in the release notes, eh?\nLike in the generated sample config file and in the generated docs?","accounts_in_message":[],"_revision_number":4},{"id":"a4cf9ab2b45a9acc9f156694b0d5912f3c8f9a4a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-19 19:51:49.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"12d93b585168869a61a8ccabb32ae3309ae60e2a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-19 20:39:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"98bd0045ad50b95b6e6116dce6ea36e1853063ee","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-10-19 20:57:49.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nIn general looks good, with nit on testing and notes on docs below:\n\nWe should probably be documenting this in doc/source/install/barbican-backend.rst, doc/source/_static/barbican.conf.sample and etc/barbican/barbican.conf","accounts_in_message":[],"_revision_number":5},{"id":"7d39405b7056f0fb6a87175b569482e45c1cc95a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-19 21:21:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a4d0d58de065c24bc1371bfb94f54d18df9abc2e","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-19 21:24:34.000000000","message":"Patch Set 5:\n\n(1 comment)\n\n@ade updated tests and barbican-backend.rst.  The conf files are not tracked in source, but they should get the updated information whenever a sample conf file is generated (e.g. via tox -e genconfig)","accounts_in_message":[],"_revision_number":5},{"id":"d66835d4b5cc43a0d5994a3fbe91450fedfa7feb","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-10-19 21:26:38.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"b5d1529ad72955236bad7f08135a631057ac15e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 22:34:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa77ad0400104ea4a2560f6c6a89c965 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12a5d39c079c407bbe54d5ce9248a3c3 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fe8931be2b143aebf8422bb3dc7ff36 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e76b25ec34924feeb4ba293dee8fe553 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6797e06d54241f3a565336255488251 : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb003fee8a08489da1c198d66aeed1cb : SUCCESS in 7m 12s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/9f1e16dabec04d86b8c8a46a98f7a70f : SUCCESS in 42m 01s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/416a20f914874e099b7c96b842549711 : POST_FAILURE in 22m 05s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/a4658d6220fd41c59b28d9a83accf720 : SUCCESS in 40m 20s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/8d3765fb4fc04307aa88e69bd2d36dd2 : SUCCESS in 1h 07m 07s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7eb5f20920cb4a84a97ae6fe09e6436f : SUCCESS in 46m 45s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d00e83353c954b49998f2129e448a68b : SUCCESS in 45m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/73d71c7e24ad44b8bac5663675fde48b : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":6},{"id":"fad2d4d47cfb5c35ee48598511478ecbfa6df47b","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-21 13:56:06.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6755dfb662f90e750fb7db169539c23e14698809","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 13:56:20.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f79b3fcd319de5c24d2f6693089ff5667ebb6419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 14:56:51.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63656f7599d54453a4456b78f422ff92 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c411b320c0f649d7ae5023462b2deae1 : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8ef738fcc60428580e04025c7357eac : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9fa26d17e74423b9cbfabd48300d349 : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebd8053bf60540098dd16df1d7f31df9 : SUCCESS in 7m 39s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/98c63a12993240f2a910de666f0d2566 : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":6},{"id":"8c7f6277c031055055f451e512bf0ce11813943d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 14:56:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"c6abfaac77de7b7010224faedea30f3eafdcb2dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 14:58:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f624c4c2719a4c08b9222c5b8d89251d : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/122b752c688d40ad8c6a0ad4dfe7cc6a : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":6},{"id":"b4ebd12526473dcb014e6e51596bcd556778952d","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-28 14:50:04.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 048d410f17d90f4db32d8005a63b7bd99d9c78f1","accounts_in_message":[],"_revision_number":6},{"id":"737e8e83656cb8efdf5722acb5e3acacaf3af4a3","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-11-02 17:15:41.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit e11a962bc78e6386b9fa10247bd3ffb504ec5964","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"69459a0ecf278410835e469347c5f32f5b43ffd5","revisions":{"89841c7e71b5fe86bb66bd77874e4ea7c1289be0":{"kind":"REWORK","_number":1,"created":"2020-10-14 21:15:59.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/55/758255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/55/758255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/55/758255/1"}}},"commit":{"parents":[{"commit":"49de1a9d34b793a2fa09407201d0f6757ef46699","subject":"Bump py37 to py38 in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/49de1a9d34b793a2fa09407201d0f6757ef46699"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-14 21:04:47.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-14 21:04:47.000000000","tz":-300},"subject":"WIP: Use serial number or label for PKCS#11 tokens","message":"WIP: Use serial number or label for PKCS#11 tokens\n\nThe PKCS#11 standard does not make any guarantees about\nslot numbering, so the slot ID alone should not be used\nto identify a token.  Instead, the token\u0027s Serial Number\nor Label should be used to ensure the correct token\nis being used.\n\nThis patch adds two new config options to the p11_crypto\nplugin: token_serial_number and token_label.  It also\ndeprecates the existing slot_id option.\n\nThese new options allow for more flexibility in configuring\nthe PKCS#11 module.  The config may include the serial number,\nthe label, or both.\n\nSerial numbers should be unique, so they take higher precedence.\n\nSome devices allow tokens to have the same label, so this patch\nensures that only one token with the specified label is present.\n\nIf both serial number and label are given, the new logic checks\nto make sure that both values match the token.\n\nslot_id is deprecated, but not removed.  If the conf contains\nonly the slot_id, it will be used.  If the serial number or\nlabel are also provided, the new logic will make sure that everything\nis consistent. e.g. given a serial number and slot it will ensure\nthat the token in the given slot has the given serial number.\n\nChange-Id: I115cf1a7006a6c85f37c5e50ded13134a3dfd1a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/89841c7e71b5fe86bb66bd77874e4ea7c1289be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/89841c7e71b5fe86bb66bd77874e4ea7c1289be0"}]},"branch":"refs/heads/master"},"5207de07bf1a887ba3ec6a97ab0fdfa3a01f3b6b":{"kind":"REWORK","_number":2,"created":"2020-10-15 13:36:53.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/55/758255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/55/758255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/55/758255/2"}}},"commit":{"parents":[{"commit":"49de1a9d34b793a2fa09407201d0f6757ef46699","subject":"Bump py37 to py38 in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/49de1a9d34b793a2fa09407201d0f6757ef46699"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-14 21:04:47.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-15 13:35:42.000000000","tz":-300},"subject":"WIP: Use serial number or label for PKCS#11 tokens","message":"WIP: Use serial number or label for PKCS#11 tokens\n\nThe PKCS#11 standard does not make any guarantees about\nslot numbering, so the slot ID alone should not be used\nto identify a token.  Instead, the token\u0027s Serial Number\nor Label should be used to ensure the correct token\nis being used.\n\nThis patch adds two new config options to the p11_crypto\nplugin: token_serial_number and token_label.  It also\ndeprecates the existing slot_id option.\n\nThese new options allow for more flexibility in configuring\nthe PKCS#11 module.  The config may include the serial number,\nthe label, or both.\n\nSerial numbers should be unique, so they take higher precedence.\n\nSome devices allow tokens to have the same label, so this patch\nensures that only one token with the specified label is present.\n\nIf both serial number and label are given, the new logic checks\nto make sure that both values match the token.\n\nslot_id is deprecated, but not removed.  If the conf contains\nonly the slot_id, it will be used.  If the serial number or\nlabel are also provided, the new logic will make sure that everything\nis consistent. e.g. given a serial number and slot it will ensure\nthat the token in the given slot has the given serial number.\n\nChange-Id: I115cf1a7006a6c85f37c5e50ded13134a3dfd1a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/5207de07bf1a887ba3ec6a97ab0fdfa3a01f3b6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/5207de07bf1a887ba3ec6a97ab0fdfa3a01f3b6b"}]},"branch":"refs/heads/master"},"d26fa94c53e887fb7daf320110042585b88ebca4":{"kind":"REWORK","_number":3,"created":"2020-10-15 18:10:23.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/55/758255/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/55/758255/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/55/758255/3"}}},"commit":{"parents":[{"commit":"49de1a9d34b793a2fa09407201d0f6757ef46699","subject":"Bump py37 to py38 in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/49de1a9d34b793a2fa09407201d0f6757ef46699"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-14 21:04:47.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-15 18:08:24.000000000","tz":-300},"subject":"Use serial number or label for PKCS#11 tokens","message":"Use serial number or label for PKCS#11 tokens\n\nThe PKCS#11 standard does not make any guarantees about\nslot numbering, so the slot ID alone should not be used\nto identify a token.  Instead, the token\u0027s Serial Number\nor Label should be used to ensure the correct token\nis being used.\n\nThis patch adds two new config options to the p11_crypto\nplugin: token_serial_number and token_label.\n\nThese new options allow for more flexibility in configuring\nthe PKCS#11 module.  The config may include the serial number,\nthe label, or both.\n\nSerial numbers should be unique, so they take higher precedence.\n\nSome devices allow tokens to have the same label, so this patch\nensures that only one token with the specified label is present.\n\nIf both serial number and label are given, the new logic checks\nto make sure that both values match the token.\n\nslot_id continues to work as expected.  If the conf contains\nonly the slot_id, it will be used.  If the serial number or\nlabel are also provided, the new logic will make sure that everything\nis consistent. e.g. given a serial number and slot_id it will ensure\nthat the token in the given slot has the given serial number.\n\nChange-Id: I115cf1a7006a6c85f37c5e50ded13134a3dfd1a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/d26fa94c53e887fb7daf320110042585b88ebca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/d26fa94c53e887fb7daf320110042585b88ebca4"}]},"branch":"refs/heads/master"},"17bdb280d33d5604e52982324593f9e7de725f2a":{"kind":"REWORK","_number":4,"created":"2020-10-15 18:14:08.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/55/758255/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/55/758255/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/55/758255/4"}}},"commit":{"parents":[{"commit":"49de1a9d34b793a2fa09407201d0f6757ef46699","subject":"Bump py37 to py38 in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/49de1a9d34b793a2fa09407201d0f6757ef46699"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-14 21:04:47.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-15 18:12:01.000000000","tz":-300},"subject":"Use serial number or label for PKCS#11 tokens","message":"Use serial number or label for PKCS#11 tokens\n\nThe PKCS#11 standard does not make any guarantees about\nslot numbering, so the slot ID alone should not be used\nto identify a token.  Instead, the token\u0027s Serial Number\nor Label should be used to ensure the correct token\nis being used.\n\nThis patch adds two new config options to the p11_crypto\nplugin: token_serial_number and token_label.\n\nThese new options allow for more flexibility in configuring\nthe PKCS#11 module.  The config may include the serial number,\nthe label, or both.\n\nSerial numbers should be unique, so they take higher precedence.\n\nSome devices allow tokens to have the same label, so this patch\nensures that only one token with the specified label is present.\n\nIf both serial number and label are given, the new logic checks\nto make sure that both values match the token.\n\nslot_id continues to work as expected, although its use is discouraged\nand may be deprecated in a future patch.  If the conf contains\nonly the slot_id, it will be used.  If the serial number or\nlabel are also provided, the new logic will make sure that everything\nis consistent. e.g. given a serial number and slot_id it will ensure\nthat the token in the given slot has the given serial number.\n\nChange-Id: I115cf1a7006a6c85f37c5e50ded13134a3dfd1a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/17bdb280d33d5604e52982324593f9e7de725f2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/17bdb280d33d5604e52982324593f9e7de725f2a"}]},"branch":"refs/heads/master"},"87b406f69b7456b81d6b951207059a3595ccd9ce":{"kind":"REWORK","_number":5,"created":"2020-10-19 20:39:51.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/55/758255/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/55/758255/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/55/758255/5"}}},"commit":{"parents":[{"commit":"49de1a9d34b793a2fa09407201d0f6757ef46699","subject":"Bump py37 to py38 in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/49de1a9d34b793a2fa09407201d0f6757ef46699"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-14 21:04:47.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-19 20:39:37.000000000","tz":-300},"subject":"Use serial number or label for PKCS#11 tokens","message":"Use serial number or label for PKCS#11 tokens\n\nThe PKCS#11 standard does not make any guarantees about\nslot numbering, so the slot ID alone should not be used\nto identify a token.  Instead, the token\u0027s Serial Number\nor Label should be used to ensure the correct token\nis being used.\n\nThis patch adds two new config options to the p11_crypto\nplugin: token_serial_number and token_label.\n\nThese new options allow for more flexibility in configuring\nthe PKCS#11 module.  The config may include either the token\u0027s\nserial number or its label.\n\nSerial numbers should be unique, so they take higher precedence.\n\nSome devices allow tokens to have the same label, so this patch\nensures that only one token with the specified label is present.\n\nIf both serial number and label are given, only the serial number\nwill be checked and an error will be raised if it is not found.\n\nslot_id continues to work as expected, although its use is discouraged\nand may be deprecated in a future patch.  If the conf contains\nonly the slot_id, it will be used.  If the serial number or\nlabel are also provided, the new logic will ignore the slot_id\nand search for the serial number or label instead.\n\nChange-Id: I115cf1a7006a6c85f37c5e50ded13134a3dfd1a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/87b406f69b7456b81d6b951207059a3595ccd9ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/87b406f69b7456b81d6b951207059a3595ccd9ce"}]},"branch":"refs/heads/master"},"69459a0ecf278410835e469347c5f32f5b43ffd5":{"kind":"REWORK","_number":6,"created":"2020-10-19 21:21:22.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/55/758255/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/55/758255/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/55/758255/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/55/758255/6"}}},"commit":{"parents":[{"commit":"49de1a9d34b793a2fa09407201d0f6757ef46699","subject":"Bump py37 to py38 in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/49de1a9d34b793a2fa09407201d0f6757ef46699"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-14 21:04:47.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-19 21:21:15.000000000","tz":-300},"subject":"Use serial number or label for PKCS#11 tokens","message":"Use serial number or label for PKCS#11 tokens\n\nThe PKCS#11 standard does not make any guarantees about\nslot numbering, so the slot ID alone should not be used\nto identify a token.  Instead, the token\u0027s Serial Number\nor Label should be used to ensure the correct token\nis being used.\n\nThis patch adds two new config options to the p11_crypto\nplugin: token_serial_number and token_label.\n\nThese new options allow for more flexibility in configuring\nthe PKCS#11 module.  The config may include either the token\u0027s\nserial number or its label.\n\nSerial numbers should be unique, so they take higher precedence.\n\nSome devices allow tokens to have the same label, so this patch\nensures that only one token with the specified label is present.\n\nIf both serial number and label are given, only the serial number\nwill be checked and an error will be raised if it is not found.\n\nslot_id continues to work as expected, although its use is discouraged\nand may be deprecated in a future patch.  If the conf contains\nonly the slot_id, it will be used.  If the serial number or\nlabel are also provided, the new logic will ignore the slot_id\nand search for the serial number or label instead.\n\nChange-Id: I115cf1a7006a6c85f37c5e50ded13134a3dfd1a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/69459a0ecf278410835e469347c5f32f5b43ffd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/69459a0ecf278410835e469347c5f32f5b43ffd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
