)]}'
{"id":"openstack%2Fbarbican~760139","triplet_id":"openstack%2Fbarbican~master~I391fb6d232062abd1f52c76af6effa20fce230f1","project":"openstack/barbican","branch":"master","topic":"stable/victoria","hashtags":[],"change_id":"I391fb6d232062abd1f52c76af6effa20fce230f1","subject":"DNM: Victoria health check","status":"ABANDONED","created":"2020-10-28 14:00:57.000000000","updated":"2020-10-28 14:44:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c1c781c9cc22db7974e17fef06dc73e6b62670c9","_number":760139,"virtual_id_number":760139,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-28 14:19:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 14:19:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"050a53ffbbb6390cc5576d2aaf0604c407ec275a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-28 14:00:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66ec5c8e3527a496081a0f6b4cbbea4473a27da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 14:19:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cf9307130d7742a9ab0025f0d8c60c28 : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c93acd4085264e3b9be0043d42c95f4d : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45bb809c924c4a51973bd10b81089c4a : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":1},{"id":"acfba054f99374639c98ce7d4c4307db201b78f4","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-28 14:44:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2de5c6ef0e89e0ca32df373917b5e1d363d1ab6","revisions":{"b2de5c6ef0e89e0ca32df373917b5e1d363d1ab6":{"kind":"REWORK","_number":1,"created":"2020-10-28 14:00:57.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/39/760139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/39/760139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/760139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/760139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/760139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/39/760139/1"}}},"commit":{"parents":[{"commit":"c168c8fad745721c42b8fbaef8d8cce615d09ee4","subject":"Merge \"Fix debug log string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/c168c8fad745721c42b8fbaef8d8cce615d09ee4"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-28 14:00:48.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-28 14:00:55.000000000","tz":-300},"subject":"DNM: Victoria health check","message":"DNM: Victoria health check\n\nChange-Id: I391fb6d232062abd1f52c76af6effa20fce230f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b2de5c6ef0e89e0ca32df373917b5e1d363d1ab6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b2de5c6ef0e89e0ca32df373917b5e1d363d1ab6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
