)]}'
{"id":"openstack%2Fbarbican~760141","triplet_id":"openstack%2Fbarbican~stable%2Ftrain~I42cc7b757609292f0dd2f2718c1b2b31487ffbbe","project":"openstack/barbican","branch":"stable/train","hashtags":[],"change_id":"I42cc7b757609292f0dd2f2718c1b2b31487ffbbe","subject":"DNM: Train health check","status":"ABANDONED","created":"2020-10-28 14:04:36.000000000","updated":"2020-10-28 20:25:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4738516cf75cbc54ad7d2fa2226c45c307bb6849","_number":760141,"virtual_id_number":760141,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-28 15:23:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 15:23:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59b7258139081dc41f8e4da746dddd6f8e214707","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-28 14:04:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59cb3cb2d5a74e555207f4bc528c0cf37287f92b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 15:23:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/542c2079f98040b9850758cf6ff3d1be : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe11af98952647f293385b66ba468bef : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642c6bb21d054bb8a8cc7b8686b75e7e : SUCCESS in 4m 28s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/45a34dccf73645888af1bd2557120665 : SUCCESS in 37m 27s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/6d2fe592a4104b84bf5de7c743ca469a : FAILURE in 31m 15s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/2e2243b01c3f450797be84376ce6b5f3 : FAILURE in 22m 03s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/dc733cc0cabe4df993f2e8cb5c8f129c : FAILURE in 9m 13s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/4ce4dc5fa82c43ee911f6a4f2a8668cc : FAILURE in 21m 40s\n- barbican-grenade-devstack https://zuul.opendev.org/t/openstack/build/967549e4b90f4141851521398c739fff : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4738516cf75cbc54ad7d2fa2226c45c307bb6849","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-10-28 20:25:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75a83f8db856dcc032fde773a857e1e500743275","revisions":{"75a83f8db856dcc032fde773a857e1e500743275":{"kind":"REWORK","_number":1,"created":"2020-10-28 14:04:36.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/41/760141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/41/760141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/41/760141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/41/760141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/41/760141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/41/760141/1"}}},"commit":{"parents":[{"commit":"07be1986b44191600de8b9bf919bef395c3a8613","subject":"Fix the barbicanclient installation not from source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/07be1986b44191600de8b9bf919bef395c3a8613"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-28 14:04:30.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-10-28 14:04:30.000000000","tz":-300},"subject":"DNM: Train health check","message":"DNM: Train health check\n\nChange-Id: I42cc7b757609292f0dd2f2718c1b2b31487ffbbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/75a83f8db856dcc032fde773a857e1e500743275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/75a83f8db856dcc032fde773a857e1e500743275"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
