)]}'
{"id":"openstack%2Fbarbican~769643","triplet_id":"openstack%2Fbarbican~master~I8084799db1706969892f84d4f34dea809c7fd3cc","project":"openstack/barbican","branch":"master","hashtags":[],"change_id":"I8084799db1706969892f84d4f34dea809c7fd3cc","subject":"Update doc8 version","status":"ABANDONED","created":"2021-01-07 02:35:41.000000000","updated":"2021-01-29 00:47:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc2bf880336eac17426a5b75728d82ceed13ed0f","_number":769643,"virtual_id_number":769643,"owner":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-26 14:23:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"all":[{"value":-1,"date":"2021-01-28 17:52:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":-1,"date":"2021-01-28 17:47:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}],"CC":[{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-07 03:27:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 13:14:22.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"CC"},{"updated":"2021-01-28 17:01:10.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2021-01-28 17:47:16.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"}],"messages":[{"id":"3355d8d32c88c7a29de936321a64071473a1ab16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"date":"2021-01-07 02:35:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d256fa70f61b168f0e59f994fa0d12db08570f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 03:27:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/baea00d04dc54e09b283b842f84823b0 : SUCCESS in 4m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/79d2fef4b09a474097856b7631a1733a : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1b8946022b749cfbaab0e52e379f8a5 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be50e53cc8e645f19f72ed6f247c326b : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3d745038e56431096f2815859f8f18a : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/888e669250884b20944b0e43828341ed : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1eeb32f9e9ef456881c76ab51bc5fb62 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3d6c4e03a64407b00ce9b3ff47bbe7 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0795b2976f4d4eca9980960453dfb072 : SUCCESS in 5m 19s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/279bc0910b2e47cf8a89b4856f36efba : SUCCESS in 39m 28s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/11fe4d7c31ba44848b30b1474b0e2fce : POST_FAILURE in 7m 45s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/795f0a55a472417f87faa6a2458fc4f4 : SUCCESS in 39m 32s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/6c200436a6fa4d0e8d33e11f38dc7290 : SUCCESS in 51m 28s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/bae547f63ed049fd8ab4317d6960a831 : FAILURE in 43m 59s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/467843dbbfdb45539de14a3de98d9ed8 : SUCCESS in 33m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/03de9a829ab547488af61e340f7ec54e : SUCCESS in 44m 17s","accounts_in_message":[],"_revision_number":1},{"id":"62df938ca7a1c0a34fbe4c4495d2a20c806f84a4","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-01-26 13:14:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4a4d3b770e1f4fc1ddb71b7404c6b05b544e3aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 14:23:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/26146e1e2e9f44ae8421feba66addf0f : SUCCESS in 6m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/62623be4dfb345adb3e7b4e1521c5560 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9355864ae6694afd83898b8ccdf728f9 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ca2aaab6ff4379b12e4531e083d4a4 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/816340cc30ef4a66bd0a356021cf32e5 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd40bb863006482fb2af3caf6a0f38d8 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e14a2df8d86419194ff808f250e530f : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba63b23256c941df9ffb6c7d506cae37 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e2540e1eb7d449f83f85d246cf32853 : SUCCESS in 5m 08s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/f0ec7d64aa454d51968cb36f7246eeae : SUCCESS in 34m 58s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/9643fe3f03084bfea5d500ec17992399 : POST_FAILURE in 16m 15s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/bf0351a7024e40f1b28969b908cd445b : SUCCESS in 39m 28s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/a9c00ec8f6bf47dda8ca3cd783e78c14 : SUCCESS in 57m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/32b6ab75d6d14c778182d4a9ce794c0b : SUCCESS in 42m 29s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/69f5e4defbd947bc889e9e0db3379bef : SUCCESS in 59m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b5f9458c26f4962a248bd920cd78a83 : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":1},{"id":"49fb72cf46a3f2d5ad4c6896d04111a8c57c06bc","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-01-28 17:01:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"24afa65c0599cbf2d13f3cee9026e21aeb3a1489","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-01-28 17:47:16.000000000","message":"Patch Set 1: Code-Review-1\n\nThis change has been already incorporated in Barbican.\nPlease refer https://review.opendev.org/c/openstack/barbican/+/767275","accounts_in_message":[],"_revision_number":1},{"id":"40cbf2bf7cd1ff06fce4e5e9df7196b938436518","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-01-28 17:51:41.000000000","message":"Removed Code-Review+1 by Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"afaf07a1f95d41dfc9e3cfa087f998a6db959e75","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-01-28 17:52:24.000000000","message":"Patch Set 1: Code-Review-1\n\nDuplicate of https://review.opendev.org/c/openstack/barbican/+/767275.\nPlease abandon it","accounts_in_message":[],"_revision_number":1},{"id":"fc2bf880336eac17426a5b75728d82ceed13ed0f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"date":"2021-01-29 00:47:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b848efb688adf676dcbfebc0e71a01a7ed953fb","revisions":{"3b848efb688adf676dcbfebc0e71a01a7ed953fb":{"kind":"REWORK","_number":1,"created":"2021-01-07 02:35:41.000000000","uploader":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"ref":"refs/changes/43/769643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/43/769643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/43/769643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/43/769643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/43/769643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/43/769643/1"}}},"commit":{"parents":[{"commit":"5734539d44672ab128c50e36def058b9f3e5729f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/5734539d44672ab128c50e36def058b9f3e5729f"}]}],"author":{"name":"YuehuiLei","email":"leiyuehui-s@inspur.com","date":"2021-01-07 02:34:18.000000000","tz":480},"committer":{"name":"YuehuiLei","email":"leiyuehui-s@inspur.com","date":"2021-01-07 02:35:03.000000000","tz":480},"subject":"Update doc8 version","message":"Update doc8 version\n\nThe doc8 lib supports Py36 starting from version 0.8.1\n\nChange-Id: I8084799db1706969892f84d4f34dea809c7fd3cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/3b848efb688adf676dcbfebc0e71a01a7ed953fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/3b848efb688adf676dcbfebc0e71a01a7ed953fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
