)]}'
{"id":"openstack%2Fbarbican~778239","triplet_id":"openstack%2Fbarbican~stable%2Fvictoria~Iaa2a82718cb75e291a0b81b896bd136c29fa6fa0","project":"openstack/barbican","branch":"stable/victoria","topic":"pkcs11-os-locking-stable/victoria","hashtags":[],"change_id":"Iaa2a82718cb75e291a0b81b896bd136c29fa6fa0","subject":"Use system locks in pkcs11 library","status":"MERGED","created":"2021-03-02 21:44:27.000000000","updated":"2021-03-05 20:36:10.000000000","submitted":"2021-03-05 20:33:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778239-pkcs11-os-locking-stable/victoria","cherry_pick_of_change":776996,"cherry_pick_of_patch_set":1,"meta_rev_id":"3e3302b1848b319db82bac4dfee860bf1de1e47d","_number":778239,"virtual_id_number":778239,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 20:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2021-03-05 19:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2021-03-05 19:22:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2021-03-05 19:30:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 22:41:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 19:22:03.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"}],"messages":[{"id":"9c57917d542962be4b932273834604fb832be63a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-02 21:44:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"16ed7c509ece5bc786e3fd9a18f1eccdd3ee775c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 22:41:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82832cad2c5747ffbf3554df4476801d : FAILURE in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a2aaf6fa83433792ca9fbad06beda4 : FAILURE in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7945634cd8754926a83b477409608dd8 : FAILURE in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40004690d23143e9a45f1021ae30bd66 : FAILURE in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c9610dffbdb451883900dc406c3c649 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37482c2134ec41aca0089ad558d9cf55 : SUCCESS in 5m 14s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/1583752508454fc7a9a96b394e1644b7 : FAILURE in 32m 03s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/46d515ce43c14939898ad10d81fde8a8 : POST_FAILURE in 18m 32s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/f575d2763eb84558a74a91b4a172e38f : FAILURE in 31m 37s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/5bc14b6f19fc459bbc98514b15ccab15 : SUCCESS in 43m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ccdadcd004bc4db394b84394cae5185e : FAILURE in 50m 16s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d16e0da7af2f410fbe74b2e6b1f96457 : SUCCESS in 39m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c958bcd4c5334744a4e08f055890e4ff : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1f5a507341c945476d02e78911066f0823f0894d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-04 19:57:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5730b4b08ca87a03569c32c99cf60dd02b12dba6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-04 20:43:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a999c1a808fc4a25b8de92da6ba1634a95930f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:39:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ad943130c4741aea1d83c7d4707a2c6 : FAILURE in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/286bfa3d9ddd448d9609d752cc657362 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86d78c0cfac541f0ab656c46313b9e6b : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29d9cc394f01483f8d1a10d8ade9d2db : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8af04f0cf95841eebaf9d6a4802642f5 : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/717d414140994db1837a82a8ae6c06e5 : SUCCESS in 6m 07s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/f1ad7178501d418f8945f2e33b36bf43 : SUCCESS in 33m 40s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/4de61acea95b401b839888a289655620 : POST_FAILURE in 18m 13s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/03a5e37040f145c0aa73c1ecb395d32c : SUCCESS in 47m 33s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/a49fe99fece84295a782cbe5d9f05a2d : SUCCESS in 53m 51s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e00dc4de0de84962885946a1da5a9397 : SUCCESS in 39m 30s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/31ddb2ca47154bb7852dfa1d759dc16d : SUCCESS in 36m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a4afe6d3824f4a6692c06158060a3f73 : SUCCESS in 42m 49s","accounts_in_message":[],"_revision_number":3},{"id":"1a42541b6211fd423a7f6ee30fe93b61509b499e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-04 23:06:15.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"974c60219f5f3d49ef00da2021bbf52f0a5ba78d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-04 23:28:05.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"71f923b81556a78ff9ded901072f4fd70acbf7b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 00:25:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/261e802d0a4a4ad19a7f554ae1a18c26 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78501a1fd0e48d78a8ae23b3dd75b10 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ababc9a91574ab792fc0d8e87117d27 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d769c47d1a674bcc9bc5e01eb30621cf : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3b6ae94fbd406eb26c3227dcb53f69 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/927564f8a054460faab69a642c8b011b : SUCCESS in 4m 28s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/d498f0eef53d4b6790b3ab3d248932dc : SUCCESS in 28m 00s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/7a2a64b43d464817b534af13768e3e1f : POST_FAILURE in 17m 49s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/26946f6f2d4e49c399a680c0e5ba1f3e : SUCCESS in 30m 24s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/6ec1aa5f96b545e7afd5c9c7f4fa272c : SUCCESS in 56m 11s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/102191b07ef243cbadbb6a19c40fa9b5 : SUCCESS in 39m 48s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/9b7f4738cb5245e8806f1d351d8533c0 : FAILURE in 35m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5ccc76376c84416ca343a9d9fcebe401 : SUCCESS in 41m 32s","accounts_in_message":[],"_revision_number":5},{"id":"2125ca0e39081f9b91dac351fa4b61348b4d9a21","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-05 13:10:33.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1ca298944cf8ba591ac14524dbff0560ee82dc7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:21:08.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc4ef687dc0c4e6d991aae42a8541b73 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b94215e531643359f8eda2523d5ad4d : SUCCESS in 11m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78b2ef8656ef4bd09f46670cc3edd2dd : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71969c2020ec405d9c0085dd487498a6 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaff7c822658491f8cd572639d2e29b6 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f8126f5adee41c289f232a5582afd83 : SUCCESS in 6m 54s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/1b0555b7c77645a4b278c957761940b3 : SUCCESS in 48m 38s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/8a67bcf52afb412e90d77b2288c3d946 : POST_FAILURE in 22m 00s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/4d771b1bb5f348b6b9c53ef615133d96 : SUCCESS in 41m 35s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/d432969f54404ea89a583852790b5e26 : SUCCESS in 1h 05m 06s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/cf49270c672841b8869d88d743064af2 : SUCCESS in 35m 19s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/8b11dddfec5b4bf2a47d928907ef204d : FAILURE in 49m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7f5039de1cea407eab9468b762e80eab : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":5},{"id":"4b34dbab55defebde33a8924029f7c18d5d4c868","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-05 14:36:11.000000000","message":"Patch Set 5:\n\nrecheck - timeout","accounts_in_message":[],"_revision_number":5},{"id":"62de00556b7a5993052472797ccba26be61c6573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 16:03:19.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f62e30908b949c180743daf69b3443c : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b284116b41e242b28f39c8feb9af0c7d : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96606e4155b34b84b62eeba37d485fb2 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e788911847440bb89cc6527fd1bbea8 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a0cab6ca0d4d1cad8b80475c38624e : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b91c4b5b2d3b47f782b0f659d59cea48 : SUCCESS in 6m 35s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/19d76acbe08f463087720dee35ab57f8 : SUCCESS in 39m 44s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/831cc684bf294c878d65ec689d37591c : POST_FAILURE in 27m 28s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/e969ae624c734b0887625ae29d5a8a5c : SUCCESS in 49m 14s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/739600483d9644708ab40ca390176040 : SUCCESS in 1h 21m 33s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8d69c9d27c484b309edbff6735963974 : FAILURE in 50m 03s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/1678f67906454674b6a1acefcce3d385 : SUCCESS in 47m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9429053de39d4542b1ed8c3a74dcf674 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":5},{"id":"3f718cdfe1898eb5d0243f4d9cf7fd3f56e6b3c5","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-05 16:57:09.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a04ab3a9920d6670c4d6c45da273b2902b950dc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 18:17:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee2964f9159648c5a2e4d680d826e627 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5baba5a1e7e64850a76f2ba772cb61cc : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00fc58d951a74140883cc1d9257fcb0f : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65a3a27ce06c4a6187cece1c68a28bcb : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e3ddc2633614a0ead1ebae76069f4ac : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/077a6e1e88c54608bf319dbec3d20d95 : SUCCESS in 5m 39s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/c550997884ff42558de7c31f6d6334dc : SUCCESS in 37m 06s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/53b7cc262cb54aa1928218d65f27f411 : POST_FAILURE in 19m 10s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/2cde5a4aea504c1797ea2268332a2806 : SUCCESS in 30m 27s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/d7ec61f77a554025bee3e4cb7e220021 : SUCCESS in 1h 12m 54s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a7ee1633fb6e40b5abd93bdcc527680f : SUCCESS in 43m 50s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/1702cf9c15ac44fa9d2e6676921aabc7 : SUCCESS in 59m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6a271995f38b4525bbc87b581ab893c0 : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":5},{"id":"2ddf29215aa8286dafb797d1c546fc6356f95887","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-03-05 19:22:03.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c645ddaf0c2d20af585f0c9257b09337fe9585ca","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-05 19:30:22.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c9b1b93bf1ca49fc09aec71adf1d520555a1bdae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 19:30:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"15de6c6985b80521b3d7b2a950d4272c804c80b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 20:33:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c028968805494b91a36f08369fcc8920 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e65bbd395dba4afe93a3f3d11f1a85bb : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd2712c298734d478edc32b5d9435605 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12fd1afa968544458fd0ec07992086b8 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29285923f48e42888052a8b6f03d247a : SUCCESS in 5m 05s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/ac5b65a47e484730a9ad6144758c6c65 : SUCCESS in 57m 05s","accounts_in_message":[],"_revision_number":5},{"id":"6ff3664e19e86a6a94fb4b4f74efa1dd072537d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 20:33:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3e3302b1848b319db82bac4dfee860bf1de1e47d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 20:36:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d88bce555234618b589d3256617a115 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07c74690f1fb4fd5a9dbfd2ebb9e6b71 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7a889cd200cc7a30ce1240cfd9718a60b1dbea21","revisions":{"84b848a12ba6ba9c223f0a7b1dd483ef7b83847e":{"kind":"REWORK","_number":1,"created":"2021-03-02 21:44:27.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/39/778239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/39/778239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/39/778239/1"}}},"commit":{"parents":[{"commit":"bbf024012e25b89622b58c7e8fc558d7e701e2d0","subject":"Merge \"Use barbican.conf in barbican-manage\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bbf024012e25b89622b58c7e8fc558d7e701e2d0"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-01-29 22:30:42.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-02 21:44:27.000000000","tz":0},"subject":"Use system locks in pkcs11 library","message":"Use system locks in pkcs11 library\n\nThis patch adds a new option to the PKCS#11 backend: `os_locking_ok`.\nWhen set to true, the flag CKF_OS_LOCKING_OK is passed to the\nC_Initialize function for the client library.\n\nChange-Id: Iaa2a82718cb75e291a0b81b896bd136c29fa6fa0\n(cherry picked from commit b5b350b4ced9aa5e7c7af424c439be7097052b21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/84b848a12ba6ba9c223f0a7b1dd483ef7b83847e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/84b848a12ba6ba9c223f0a7b1dd483ef7b83847e"}]},"branch":"refs/heads/stable/victoria"},"e879fb9702ecdf70f4620f4d999813c281a6ebd0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-04 19:57:28.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/39/778239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/39/778239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/39/778239/2"}}},"commit":{"parents":[{"commit":"03136fad09b23d46f7a30685e8d672c39afe8c0a","subject":"Use a newer version of hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/03136fad09b23d46f7a30685e8d672c39afe8c0a"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-01-29 22:30:42.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-04 19:57:28.000000000","tz":0},"subject":"Use system locks in pkcs11 library","message":"Use system locks in pkcs11 library\n\nThis patch adds a new option to the PKCS#11 backend: `os_locking_ok`.\nWhen set to true, the flag CKF_OS_LOCKING_OK is passed to the\nC_Initialize function for the client library.\n\nChange-Id: Iaa2a82718cb75e291a0b81b896bd136c29fa6fa0\n(cherry picked from commit b5b350b4ced9aa5e7c7af424c439be7097052b21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/e879fb9702ecdf70f4620f4d999813c281a6ebd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/e879fb9702ecdf70f4620f4d999813c281a6ebd0"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"b99f58a2c72889be9bb517f59cd23e022a03a2a5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-04 20:43:23.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/39/778239/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/39/778239/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/39/778239/3"}}},"commit":{"parents":[{"commit":"1aa486fa0b0821f47fe02da39015e0aed6964802","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1aa486fa0b0821f47fe02da39015e0aed6964802"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-01-29 22:30:42.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-04 20:43:23.000000000","tz":0},"subject":"Use system locks in pkcs11 library","message":"Use system locks in pkcs11 library\n\nThis patch adds a new option to the PKCS#11 backend: `os_locking_ok`.\nWhen set to true, the flag CKF_OS_LOCKING_OK is passed to the\nC_Initialize function for the client library.\n\nChange-Id: Iaa2a82718cb75e291a0b81b896bd136c29fa6fa0\n(cherry picked from commit b5b350b4ced9aa5e7c7af424c439be7097052b21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b99f58a2c72889be9bb517f59cd23e022a03a2a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b99f58a2c72889be9bb517f59cd23e022a03a2a5"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"6453c2c8483d951fd0c45090036ed5a0eecaa8a1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-04 23:06:15.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/39/778239/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/39/778239/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/39/778239/4"}}},"commit":{"parents":[{"commit":"53430861092bbf9e576958ca9002c575132e2fc2","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/53430861092bbf9e576958ca9002c575132e2fc2"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-01-29 22:30:42.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-04 23:06:15.000000000","tz":0},"subject":"Use system locks in pkcs11 library","message":"Use system locks in pkcs11 library\n\nThis patch adds a new option to the PKCS#11 backend: `os_locking_ok`.\nWhen set to true, the flag CKF_OS_LOCKING_OK is passed to the\nC_Initialize function for the client library.\n\nChange-Id: Iaa2a82718cb75e291a0b81b896bd136c29fa6fa0\n(cherry picked from commit b5b350b4ced9aa5e7c7af424c439be7097052b21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6453c2c8483d951fd0c45090036ed5a0eecaa8a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6453c2c8483d951fd0c45090036ed5a0eecaa8a1"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"7a889cd200cc7a30ce1240cfd9718a60b1dbea21":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-03-04 23:28:05.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/39/778239/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/39/778239/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/778239/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/39/778239/5"}}},"commit":{"parents":[{"commit":"df5fb602f0f7c62bca0b66b64cd1da903cc7dd9b","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/df5fb602f0f7c62bca0b66b64cd1da903cc7dd9b"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-01-29 22:30:42.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-04 23:28:05.000000000","tz":0},"subject":"Use system locks in pkcs11 library","message":"Use system locks in pkcs11 library\n\nThis patch adds a new option to the PKCS#11 backend: `os_locking_ok`.\nWhen set to true, the flag CKF_OS_LOCKING_OK is passed to the\nC_Initialize function for the client library.\n\nChange-Id: Iaa2a82718cb75e291a0b81b896bd136c29fa6fa0\n(cherry picked from commit b5b350b4ced9aa5e7c7af424c439be7097052b21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/7a889cd200cc7a30ce1240cfd9718a60b1dbea21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/7a889cd200cc7a30ce1240cfd9718a60b1dbea21"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
