)]}'
{"id":"openstack%2Fbarbican~778692","triplet_id":"openstack%2Fbarbican~stable%2Fvictoria~Iee4584c6fde08728c017468d9de1db73f2c79d8d","project":"openstack/barbican","branch":"stable/victoria","topic":"update-hacking-stable/victoria","hashtags":[],"change_id":"Iee4584c6fde08728c017468d9de1db73f2c79d8d","subject":"Update hacking for Python3","status":"MERGED","created":"2021-03-04 20:41:12.000000000","updated":"2021-03-05 14:02:48.000000000","submitted":"2021-03-05 14:00:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778692-update-hacking-stable/victoria","cherry_pick_of_change":716730,"cherry_pick_of_patch_set":16,"meta_rev_id":"0c639d881c1513c30c3184455bb3af4e3df72194","_number":778692,"virtual_id_number":778692,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 14:00:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2021-03-05 13:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2021-03-05 13:02:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 20:41:12.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-03-04 21:36:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c8b8055204f2d35d95364cc371cfd604a2cf775","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-04 20:41:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"161069d9e398ea1f0832b74274185d269ed8436e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:36:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/272fde7fe810455195200c528a59f84a : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/000b29a903394229a97274c5fe11e0e1 : FAILURE in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec34764e3e78440f98a4f4a5e33f6c0c : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/612bb6cf359d4498998c6b8e864b8849 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f816024221114daa9422a4a40c190123 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae31820af3546ed9e75d4f54512d28a : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66436899ad9d45e8ac37f0800a2a39de : SUCCESS in 5m 36s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/4e6ce0c5768d4f03bd2416822d5f571e : SUCCESS in 32m 28s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/415c1b727e60446da7a0890fa62588d2 : POST_FAILURE in 21m 52s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/51974d3b86e0437a93595f46a5d0ce50 : FAILURE in 31m 28s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/622f8ba6b3fb49028da580843bce4902 : SUCCESS in 53m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5d5bb3dd5e7f41e0aa3266ca3bf9510b : FAILURE in 36m 02s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/5ef498d96731447eb3e6a27e6b71ec4a : SUCCESS in 35m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b2ec8b5720ea481d975631a2891b057d : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"abee487a1eebd1f01b5a886d719a477310c946cf","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-04 21:39:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cf2224f63226b4392b9e369a6a06d794f46f2762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 22:36:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/df5b21ca25c94207b0ca615055174137 : SUCCESS in 4m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a91b7ba36f5a414f8b8e453724443f7a : FAILURE in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02296de201445f18b057448ff7dd410 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b2552ff04734b8594a0bd6919095545 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f774bfef9e2f4652afc44d7d4e59fcef : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca167f833024555a5e022b9bb8d553c : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bbe49ca21704ab1b585813283f3e3f8 : SUCCESS in 4m 53s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/41911e62646f415490964afb5fc66d66 : SUCCESS in 34m 12s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/00981a60bb24406bb344d7255409d889 : POST_FAILURE in 18m 26s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/217a9814e5264110aaa661df015347aa : SUCCESS in 39m 10s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/c2f310b426f64351ba54c253782dfb6a : SUCCESS in 53m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4fc7c5fced844597bbde15f46d65a0e9 : FAILURE in 43m 10s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/9a3dec53b37b4cddb4615e9ebcae328e : SUCCESS in 54m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ec85cb4efaa4d0f8eb5b2a1d96f3c12 : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ed78c23f20df3125dba6431dfbc32524c246e7f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-04 23:05:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3a8222663521f1069d83ef2bc597dfb8fb313e71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-04 23:27:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d706d726747cb759114a83c9450c893b08a54ac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 00:34:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/58800a0adb9a449b98d41a550beef119 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4db9299d386b4a99802d336a5dfa76dc : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1144227f02e840f5b3d563442e710135 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4940b4056774683bc846a20c0d14e02 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28c74430e0244fa69155048e567de942 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fed600b449c24551b1771298ef9c2ec9 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/204da0ff541c465a9d01a985e89e7a3d : SUCCESS in 4m 32s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/3cc326e75d484c91a8cf4090eb1acc8a : SUCCESS in 42m 14s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/c4d131576ed444e586ce9212e7cda8ad : POST_FAILURE in 23m 01s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/f50f07cd6b334a5f9a2df49c80d864cc : SUCCESS in 43m 52s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/13da76c33f434650b5914a6e16ee372e : SUCCESS in 1h 03m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/69bf35d2099c4b7d9cd339b5b310e3a3 : SUCCESS in 46m 26s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/f284b1b4e0ed43eca8a419df1dbf4475 : SUCCESS in 44m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d545b11677e44c6895c7d359960a1c4 : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":3},{"id":"8a1d2321839a3b513293165a244667adb8b36240","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-05 13:02:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"59a4cd66b62d69fbf8a01a1c07290d6795d982b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:08:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9eee7d3514a6f9903d9612c3ef750b9aab1d6d5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:00:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2474dc3e49d44ecc9b6004628e65a027 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e990552ca56432ca4193625be70e366 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/823763efe76e489db3dd19cdceb8f4e9 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/854b9ebe71794c33887f9e1bb9490cd3 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c48aa7b86e4a71a201c57e95b198dc : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/780793a644dd4309a6e879f1348a4c88 : SUCCESS in 5m 00s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/2c3d0d508fd94d29b21876ac88df3e02 : SUCCESS in 37m 01s","accounts_in_message":[],"_revision_number":3},{"id":"732ead183b7ad9f80d56b0f2ef33326880eab52f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:00:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0c639d881c1513c30c3184455bb3af4e3df72194","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:02:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ea78f9adb684ceb852bacb6107db032 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de93643a7dc448e38d12bba5203d86fa : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"df5fb602f0f7c62bca0b66b64cd1da903cc7dd9b","revisions":{"1aa486fa0b0821f47fe02da39015e0aed6964802":{"kind":"REWORK","_number":1,"created":"2021-03-04 20:41:12.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/92/778692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/92/778692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/92/778692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/92/778692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/92/778692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/92/778692/1"}}},"commit":{"parents":[{"commit":"bbf024012e25b89622b58c7e8fc558d7e701e2d0","subject":"Merge \"Use barbican.conf in barbican-manage\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bbf024012e25b89622b58c7e8fc558d7e701e2d0"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-01 18:56:29.000000000","tz":120},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-04 20:41:12.000000000","tz":0},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found by updated hacking version.\n\nUpdate local hacking checks for new flake8, remove\ntest B314 since that tests difference between Python 2 and 3,\nthere\u0027s no need to advise using six anymore.\nUse oslotest.base directly, this fixes the hacking tests.\n\nRemove ddt usage in testsuite, it does not work with current hacking\nversion anymore.\n\nChange-Id: Iee4584c6fde08728c017468d9de1db73f2c79d8d\n(cherry picked from commit 9dbeefb55eb55a9baa481ec6d6293527d46e04a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1aa486fa0b0821f47fe02da39015e0aed6964802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1aa486fa0b0821f47fe02da39015e0aed6964802"}]},"branch":"refs/heads/stable/victoria"},"53430861092bbf9e576958ca9002c575132e2fc2":{"kind":"REWORK","_number":2,"created":"2021-03-04 23:05:12.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/92/778692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/92/778692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/92/778692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/92/778692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/92/778692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/92/778692/2"}}},"commit":{"parents":[{"commit":"bbf024012e25b89622b58c7e8fc558d7e701e2d0","subject":"Merge \"Use barbican.conf in barbican-manage\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bbf024012e25b89622b58c7e8fc558d7e701e2d0"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-01 18:56:29.000000000","tz":120},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-04 23:04:33.000000000","tz":-360},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found by updated hacking version.\n\nUpdate local hacking checks for new flake8, remove\ntest B314 since that tests difference between Python 2 and 3,\nthere\u0027s no need to advise using six anymore.\nUse oslotest.base directly, this fixes the hacking tests.\n\nRemove ddt usage in testsuite, it does not work with current hacking\nversion anymore.\n\nChange-Id: Iee4584c6fde08728c017468d9de1db73f2c79d8d\n(cherry picked from commit 9dbeefb55eb55a9baa481ec6d6293527d46e04a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/53430861092bbf9e576958ca9002c575132e2fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/53430861092bbf9e576958ca9002c575132e2fc2"}]},"branch":"refs/heads/stable/victoria"},"df5fb602f0f7c62bca0b66b64cd1da903cc7dd9b":{"kind":"REWORK","_number":3,"created":"2021-03-04 23:27:44.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/92/778692/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/92/778692/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/92/778692/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/92/778692/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/92/778692/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/92/778692/3"}}},"commit":{"parents":[{"commit":"bbf024012e25b89622b58c7e8fc558d7e701e2d0","subject":"Merge \"Use barbican.conf in barbican-manage\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bbf024012e25b89622b58c7e8fc558d7e701e2d0"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-01 18:56:29.000000000","tz":120},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-04 23:27:30.000000000","tz":-360},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found by updated hacking version.\n\nUpdate local hacking checks for new flake8, remove\ntest B314 since that tests difference between Python 2 and 3,\nthere\u0027s no need to advise using six anymore.\nUse oslotest.base directly, this fixes the hacking tests.\n\nRemove ddt usage in testsuite, it does not work with current hacking\nversion anymore.\n\nChange-Id: Iee4584c6fde08728c017468d9de1db73f2c79d8d\n(cherry picked from commit 9dbeefb55eb55a9baa481ec6d6293527d46e04a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/df5fb602f0f7c62bca0b66b64cd1da903cc7dd9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/df5fb602f0f7c62bca0b66b64cd1da903cc7dd9b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
