)]}'
{"id":"openstack%2Fbarbican~779299","triplet_id":"openstack%2Fbarbican~master~Ibe9aac10edbf43c106eaa1769af5ee777a4be5aa","project":"openstack/barbican","branch":"master","topic":"new_rbac","hashtags":[],"change_id":"Ibe9aac10edbf43c106eaa1769af5ee777a4be5aa","subject":"Implement secure RBAC for secrets API","status":"MERGED","created":"2021-03-08 16:59:24.000000000","updated":"2021-03-10 13:22:05.000000000","submitted":"2021-03-10 13:20:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"779299-new_rbac","meta_rev_id":"45404320809039cc89db68c2bcd879706c8efacc","_number":779299,"virtual_id_number":779299,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 13:20:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2021-03-10 12:30:24.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2021-03-09 21:32:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2021-03-10 12:30:24.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 17:18:51.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2021-03-08 18:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 12:30:24.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"}],"messages":[{"id":"f7547ca10f67ed5865b28c27e1d96bb44728830f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-03-08 16:59:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1b2ef53d60947cc0163efc4b72126b1fb9666d9","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-08 17:18:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"679904bfed6c96146514d5c92af18381f8128371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 18:09:48.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe5c6e627de54f53a430463c74922576 : FAILURE in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92aa42c6412f467bb58710805168928b : FAILURE in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc60118216cc4354bff51099d197bb83 : FAILURE in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/227cffcbeaa44c8ea52e82dc4a3cd342 : FAILURE in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8c496227eba4eb1b6537c6371888a4a : FAILURE in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399a7d9bb01f4453b1b0a1b9ab8760dc : FAILURE in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53181d83608b4bbb9d8a38ea12a5d24c : SUCCESS in 5m 30s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/db259fef2ec14ea495b9e0314289c0fe : FAILURE in 27m 57s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/cfd29de23fdc45a9a7b9ca4b3078adf2 : POST_FAILURE in 29m 00s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/7504ef62ebf5425fb87f17ad8c2b6bbc : FAILURE in 42m 50s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/2eb729e59f294fb68e5aa6895b21f10b : SUCCESS in 1h 06m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/db1f33260396487abe6d3d8cc146eb53 : FAILURE in 39m 09s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/505d6a5d68fb4f83ab3a2041a0efbf80 : FAILURE in 37m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b80de294b1b34b9a8f1733f1dec19b70 : FAILURE in 35m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4e57d0184ec401bb9148ed05233cf51481beecf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-03-08 18:46:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"571f2c91c80e0df34c6a733fbc9c16fa670919f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 20:07:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62695d68f34048eca3cbd2509da9bc76 : FAILURE in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0c8685b5ac14ebd926f9581862da3c1 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd6644c2924441f9a6e2ebecea5023ae : FAILURE in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ed8520a93eb491980b5a33906fa5797 : FAILURE in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c5e76e7a5e84cfd9c82438279d969e9 : FAILURE in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203eb49b6dfe45b29816d6fe17a65ab5 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/437042ac3ce84482877edc89ad4504c0 : SUCCESS in 8m 17s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/a395b792e6da470a9a88549a7e459b08 : FAILURE in 31m 09s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/ea0b3029ddd843ebbc423bb58d04d7d8 : POST_FAILURE in 15m 43s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/ec274711f95641e4af55f6036a664dc3 : FAILURE in 38m 03s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/f57937185c2548469913389e92adc700 : SUCCESS in 1h 13m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/684fdaeb978d4d40ad629025b8acdc11 : FAILURE in 37m 54s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/b9a0834055554f1d9e9d0beaa892e62a : FAILURE in 36m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7eced18a6a045299d79f5b7b39eff8d : FAILURE in 38m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b0adda741decc03719e7f5adea9e5333d4cf4c00","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-09 15:13:33.000000000","message":"Topic new_rbac removed","accounts_in_message":[],"_revision_number":2},{"id":"bf4f5e6f65b58c315a85cf191aa696eec7b0f3a0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-09 15:13:40.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"97d58488d7ced7631b4461860dd6c9959aeb8987","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-03-09 19:18:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"90950f41302d31375eb2b8ece1917157b3d9ad09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 20:14:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ff0c1885c7041cbaf50c74d48ec6892 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b05b94559c4514b3b2678999a53c38 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad5179bc55fc4512bfd0fcf16aaae45a : SUCCESS in 8m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bb1de473956494292421cf8481d0630 : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53785e6b48cf464593291ef08a143732 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c5a42592374bbb8c40e06598c78e6f : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4754f5c8ecb04b978a692e736fbcc498 : SUCCESS in 6m 30s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/fe6f4de78fa04bb3ae796c887fb51587 : SUCCESS in 52m 55s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/2ad51adff9c64042baffd6f4a259f57d : POST_FAILURE in 27m 20s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/9bcf3fab4f94489d89c34351788e94d3 : SUCCESS in 29m 12s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/1e6bc398143a411a819eaa4fc5724091 : SUCCESS in 45m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/30347b296fc14923a7331293437bf1ab : SUCCESS in 43m 33s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/f82126f6b984418e9a811041f16bb43c : SUCCESS in 46m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9c2f2411cbf48c89e2dd3d8b7ff5fde : SUCCESS in 47m 29s","accounts_in_message":[],"_revision_number":3},{"id":"b341e4768ca79a60ee0f88eea338ec69f888e830","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-09 21:32:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0866e9c020ed73a95ba8ccb39d8ece6b269395f1","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-03-10 12:30:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e216712c9e647a5ac5eef877dbbda510d15f35d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:30:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7c3f2eeeb4f50c189a2cbbbf0d2f3cdb609e9782","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:20:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1696efe6caac435e924d8129ba05fc2d : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bc1d95b2c6043bc87767c47164a416a : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/516d30fb58b14dc9b4fe4b30a1b93689 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/240b5eafa3874a1bae5062a9bb495291 : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6af14f9f4eb940438a2ce4750f716897 : SUCCESS in 6m 21s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/7bc0b98df026457c87b989a2cf6a23ae : SUCCESS in 43m 01s","accounts_in_message":[],"_revision_number":3},{"id":"2db2e4510ebbd6985f518dc8855ccbe97f009bc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:20:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"45404320809039cc89db68c2bcd879706c8efacc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:22:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76a392d67f4543d09c095d59f44ba3ee : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c17b24da2844a53842d1f25283e3072 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eade0cfc719fedebced0fb2c6a8251f5f2a12682","revisions":{"a359c5ccab38d8b02f520054686da34fef69fe20":{"kind":"REWORK","_number":1,"created":"2021-03-08 16:59:24.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/99/779299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/99/779299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/779299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/779299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/779299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/99/779299/1"}}},"commit":{"parents":[{"commit":"1ca03610d7257c782b11d6bcf54074d66a79c545","subject":"Allow multiple token labels for PKCS#11 driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1ca03610d7257c782b11d6bcf54074d66a79c545"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-03-05 20:54:16.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-03-08 14:55:53.000000000","tz":-300},"subject":"Implement secure RBAC for secrets API","message":"Implement secure RBAC for secrets API\n\nAdd new project scope specific RBAC rules for the secrets API.  The old\nrules still apply, but eventually will be deprecated.  The new\nrules do include some changes to default policy, which are documented in\nthe release note.\n\nChange-Id: Ibe9aac10edbf43c106eaa1769af5ee777a4be5aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/a359c5ccab38d8b02f520054686da34fef69fe20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/a359c5ccab38d8b02f520054686da34fef69fe20"}]},"branch":"refs/heads/master"},"132f28a7e1c7fe2b8cb77dfe8d00ea87bafeddb6":{"kind":"REWORK","_number":2,"created":"2021-03-08 18:46:00.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/99/779299/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/99/779299/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/779299/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/779299/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/779299/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/99/779299/2"}}},"commit":{"parents":[{"commit":"1ca03610d7257c782b11d6bcf54074d66a79c545","subject":"Allow multiple token labels for PKCS#11 driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1ca03610d7257c782b11d6bcf54074d66a79c545"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-03-05 20:54:16.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-03-08 18:44:21.000000000","tz":-300},"subject":"Implement secure RBAC for secrets API","message":"Implement secure RBAC for secrets API\n\nAdd new project scope specific RBAC rules for the secrets API.  The old\nrules still apply, but eventually will be deprecated.  The new\nrules do include some changes to default policy, which are documented in\nthe release note.\n\nChange-Id: Ibe9aac10edbf43c106eaa1769af5ee777a4be5aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/132f28a7e1c7fe2b8cb77dfe8d00ea87bafeddb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/132f28a7e1c7fe2b8cb77dfe8d00ea87bafeddb6"}]},"branch":"refs/heads/master"},"eade0cfc719fedebced0fb2c6a8251f5f2a12682":{"kind":"REWORK","_number":3,"created":"2021-03-09 19:18:44.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/99/779299/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/99/779299/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/779299/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/779299/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/779299/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/99/779299/3"}}},"commit":{"parents":[{"commit":"1ca03610d7257c782b11d6bcf54074d66a79c545","subject":"Allow multiple token labels for PKCS#11 driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1ca03610d7257c782b11d6bcf54074d66a79c545"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-03-05 20:54:16.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-03-09 19:16:00.000000000","tz":-300},"subject":"Implement secure RBAC for secrets API","message":"Implement secure RBAC for secrets API\n\nAdd new project scope specific RBAC rules for the secrets API.  The old\nrules still apply, but eventually will be deprecated.  The new\nrules do include some changes to default policy, which are documented in\nthe release note.\n\nChange-Id: Ibe9aac10edbf43c106eaa1769af5ee777a4be5aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/eade0cfc719fedebced0fb2c6a8251f5f2a12682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/eade0cfc719fedebced0fb2c6a8251f5f2a12682"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
