)]}'
{"id":"openstack%2Fbarbican~782015","triplet_id":"openstack%2Fbarbican~master~I503641691f8414c4c4562cbedc31dc8047054f0c","project":"openstack/barbican","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I503641691f8414c4c4562cbedc31dc8047054f0c","subject":"Imported Translations from Zanata","status":"MERGED","created":"2021-03-22 06:22:03.000000000","updated":"2021-04-28 12:50:51.000000000","submitted":"2021-04-28 12:49:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782015-zanata/translations","meta_rev_id":"4d6873247eead38c8b50a42628716dd0cc41cde8","_number":782015,"virtual_id_number":782015,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 12:49:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2021-04-28 12:41:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2021-04-28 12:41:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 07:27:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-15 11:23:43.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"CC"},{"updated":"2021-04-28 12:41:04.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"5e1569d54d210dba6f30be7d3ea01a009befe6c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-22 06:22:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9545e8d2016ed00fb5f6b48f15a072779ccdd7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 07:27:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ef1d5f8a1fa4df4a969d1aef8757e8d : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a60f848f78f94ec4b8d9f2a60a622cf5 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df18cd6dc3724207bdbf23befdd9ea40 : SUCCESS in 8m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5a2490b9489483bbd2fa6b2419c5293 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0b4aa7150374df3b11e83d82fc370a4 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e68e324ef142fda7291ea62db456bd : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b23b0b9a0be0429cbf679b2cd12a8cfd : SUCCESS in 7m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/788ea7f691d9410d89ea9f109221492d : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c271f4f15c8a7b9b6bde65f5778ec51c978c47cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-30 06:23:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"01a16dc0a6ea724b13227a55211a5eee71bb2b9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 07:21:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf0eaa955dce4f54ba5bdc6d79a6eb51 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce55ebebc004e8a87d634b47396fd07 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b953cec1d744b4ab20e56350125b1f3 : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d30abbfdaee4187aa2939db91abf0b2 : SUCCESS in 9m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05f973d875d840ddbd960df2d1e4f2ea : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0794531c432747c4b1f36d735524bcaf : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15815b0264284a8997ba8ba6d1a61892 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e4fe3f9b9390436f97de4ad315e51913 : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":2},{"id":"63d41457a0829138a2dac64b839b25fffa6a49df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-31 06:13:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"26383f1b67a657dc9b1ba70ca68d6a6cf934e826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:28:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19872613993c4cb79f0c815b1b94b84d : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85dff2fd160848d49c3b2552d80e43c8 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a5f7d5e174b4927a426e5855dd6e8ca : SUCCESS in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1059b5ef53e44b50a59718fed5cab52d : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fba39eff8b65442697df72eb7787be5c : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b58b4c4efa3d402a946db8a07a02b23f : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f1f1b776a27438a82af21799ac106b6 : SUCCESS in 5m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3229180aa32b4822945d56c84b075c47 : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":3},{"id":"e46822b5e891c3029905fcc2d4bb830e9a4ab55b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-02 06:21:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6a4a3731a93150c11e9d9e6ac1d374ef3a5ce2d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 07:19:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42b20ab2b07a455890086119444d8f42 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c637f75a1254e498658c1d6357bb74b : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/968c81f62e054d55908526e171a7596c : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c503c64b43e04fd6a0f21a3c6dc0d42d : SUCCESS in 9m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d60f0cb3dad426ea7a63a40232c257f : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/690eb9edca8c4b5a8d787b69c7547608 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5ae2c8dece84475b0bc75774330c8c6 : SUCCESS in 6m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/99abd283f2704008bf1b0e8ebdfde3ee : SUCCESS in 53m 13s","accounts_in_message":[],"_revision_number":4},{"id":"948627d880067618419403ffc8ad3c45c81f272d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-14 06:26:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"76216eb470aefc4b3f9e6ea92f491d117f9fbc84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 07:19:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/caa0080aa0054111b733a0a16db732e0 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/937ea4698f7e4d9e925dd772786f0048 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a1c2f7f4ea0435dbc6458b54d682c93 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2653ff9e71a46768c39d97199387f5b : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b64e7fc5f1fe4495be3b088ead19db0d : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b39f4bee43d4479d9ff6ff29e769f86a : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59a59d79cc414092a4d15cd93f6aaecc : SUCCESS in 6m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bdafb719202d4c0cb7848490a3577b7e : FAILURE in 41m 39s","accounts_in_message":[],"_revision_number":5},{"id":"d405ba13b1b8ea95386c7dd6e2fde4f9ebb1bcca","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-04-15 11:23:43.000000000","message":"Patch Set 5:\n\nrecehck","accounts_in_message":[],"_revision_number":5},{"id":"f5d997ac27489ab034b145a70001d2041afd4298","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-27 06:11:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"76be3ee1809b95e202064b3c5ab06929fe0a5981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 07:30:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25ade056c9fb48688432019b1c049ad9 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35fff3b42b1642af93b65101b3e85287 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35ff7861db464927a0d43a7ca35cfbd5 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63332513a8e8490d97d3d39b3d8aab0f : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef9864d21a9a4bfa81cfc5578ce91dcd : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0b6762d0c84872aaf5a2cb1ddf7a53 : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f986f3b6031427b9fdbc6003753488e : SUCCESS in 7m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11e2948aebe04877be08cf3f3d209271 : FAILURE in 1h 08m 17s","accounts_in_message":[],"_revision_number":6},{"id":"b299b72a6cd118ebacd45e2e672080921547208c","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-04-27 12:40:53.000000000","message":"Patch Set 6:\n\nrecheck - octavia error","accounts_in_message":[],"_revision_number":6},{"id":"cd97b7f3e67d26703ca7ac2cbe46c18c167bd7b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 13:25:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/155e689f04184c459733e02469478907 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0eb6706b6bd452aa447b8d27400e101 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49acb431079846bdbdd9151972f7fe56 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf7608745c78464ca7f52f2e80c10bd2 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b73f8c47d7504d75876cfb88763d2ade : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73b41b5011c64dad8007388cf6778916 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d88a2c7d3b7446f1a33cddd2c7223cad : SUCCESS in 5m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e16e5454c6f44c1d983a217619205357 : SUCCESS in 42m 08s","accounts_in_message":[],"_revision_number":6},{"id":"b52479a8da9b8a184cef2cf12ed765d51874364d","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-04-28 12:41:04.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c7f6208bdd02448b001a9235ab988919ca6c4f6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 12:41:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"053496c3c4f16ae2a504cbba17f6d67a9da29b9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 12:49:23.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a7b32629f041079a87d4b82f6ea151 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2b09887d999422d857d468887756e3e : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48f295035753495184a18798eb30f2f9 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2891b8fb5a6429ea04b40f8a8b23a73 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbaf1406f3484c27b46fba380c63a845 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":6},{"id":"d3e0d65f00fef22a9d8beba30f82bbbbae234e29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 12:49:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"4d6873247eead38c8b50a42628716dd0cc41cde8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 12:50:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63bec993724c4bde9a15a8c876f82f71 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a147e74555c4421fa19839daa3cd8886 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6d0df1148ff41bce6ab8e7c5ae6c7807af826dde","revisions":{"1d5e5978b472a75d512ccadec8b0f8c8205bc0bb":{"kind":"REWORK","_number":1,"created":"2021-03-22 06:22:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/782015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/15/782015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/15/782015/1"}}},"commit":{"parents":[{"commit":"97ae3bc57add7d087f590afddf938cce2cddeac1","subject":"Merge \"Optimize conditional statements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/97ae3bc57add7d087f590afddf938cce2cddeac1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-22 06:22:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-22 06:22:02.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I503641691f8414c4c4562cbedc31dc8047054f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1d5e5978b472a75d512ccadec8b0f8c8205bc0bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1d5e5978b472a75d512ccadec8b0f8c8205bc0bb"}]},"branch":"refs/heads/master"},"25873087bbdb5357b246a340fccfd7d4f1fcf7ca":{"kind":"REWORK","_number":2,"created":"2021-03-30 06:23:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/782015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/15/782015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/15/782015/2"}}},"commit":{"parents":[{"commit":"7cefdaabe0566ec60ed08341a8b5d5338ab271a4","subject":"Merge \"Fix transport key policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/7cefdaabe0566ec60ed08341a8b5d5338ab271a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-30 06:23:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-30 06:23:32.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I503641691f8414c4c4562cbedc31dc8047054f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/25873087bbdb5357b246a340fccfd7d4f1fcf7ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/25873087bbdb5357b246a340fccfd7d4f1fcf7ca"}]},"branch":"refs/heads/master"},"01af0e7c84ebf76543b8ddbf4ee1025707a8d53d":{"kind":"REWORK","_number":3,"created":"2021-03-31 06:13:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/782015/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/15/782015/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/15/782015/3"}}},"commit":{"parents":[{"commit":"59aa7f71b9c34170648236b52e00fb7eaf3737bd","subject":"Merge \"Monkey patch original current_thread _active\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/59aa7f71b9c34170648236b52e00fb7eaf3737bd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-31 06:13:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-31 06:13:33.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I503641691f8414c4c4562cbedc31dc8047054f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/01af0e7c84ebf76543b8ddbf4ee1025707a8d53d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/01af0e7c84ebf76543b8ddbf4ee1025707a8d53d"}]},"branch":"refs/heads/master"},"250067e34079847ea1e5e82d0a39c7f022ac5feb":{"kind":"REWORK","_number":4,"created":"2021-04-02 06:21:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/782015/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/15/782015/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/15/782015/4"}}},"commit":{"parents":[{"commit":"922c68badaf5cc4b43862ab261119b68cdc68a37","subject":"Fix RBAC and ACL access for managing secret containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/922c68badaf5cc4b43862ab261119b68cdc68a37"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-02 06:21:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-02 06:21:53.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I503641691f8414c4c4562cbedc31dc8047054f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/250067e34079847ea1e5e82d0a39c7f022ac5feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/250067e34079847ea1e5e82d0a39c7f022ac5feb"}]},"branch":"refs/heads/master"},"37899b45f20cf904c14e87f73bb21b72cc114c1b":{"kind":"REWORK","_number":5,"created":"2021-04-14 06:26:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/782015/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/15/782015/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/15/782015/5"}}},"commit":{"parents":[{"commit":"5257006dd1eadf3649e33d0170de8f331d81c6a6","subject":"Merge \"Add Python3 xena unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/5257006dd1eadf3649e33d0170de8f331d81c6a6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-14 06:26:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-14 06:26:32.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I503641691f8414c4c4562cbedc31dc8047054f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37899b45f20cf904c14e87f73bb21b72cc114c1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37899b45f20cf904c14e87f73bb21b72cc114c1b"}]},"branch":"refs/heads/master"},"6d0df1148ff41bce6ab8e7c5ae6c7807af826dde":{"kind":"REWORK","_number":6,"created":"2021-04-27 06:11:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/782015/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/15/782015/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/782015/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/15/782015/6"}}},"commit":{"parents":[{"commit":"21661bebd93e3e78755becd542ed8e0fffbde64d","subject":"Fix Vault functional test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/21661bebd93e3e78755becd542ed8e0fffbde64d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-27 06:11:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-27 06:11:21.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I503641691f8414c4c4562cbedc31dc8047054f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6d0df1148ff41bce6ab8e7c5ae6c7807af826dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6d0df1148ff41bce6ab8e7c5ae6c7807af826dde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
