)]}'
{"id":"openstack%2Fbarbican~783239","triplet_id":"openstack%2Fbarbican~stable%2Fwallaby~I4a806ca1bcc8a7dd506b7121fde9cfe6fb34dcd8","project":"openstack/barbican","branch":"stable/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"I4a806ca1bcc8a7dd506b7121fde9cfe6fb34dcd8","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","status":"MERGED","created":"2021-03-26 10:15:55.000000000","updated":"2021-03-30 20:41:47.000000000","submitted":"2021-03-30 20:40:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783239-create-wallaby","meta_rev_id":"39a1cf1b49e648a0e91f1b72a2b420a85899766b","_number":783239,"virtual_id_number":783239,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-30 20:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2021-03-30 20:25:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2021-03-30 20:25:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 10:41:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 20:25:29.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"118a3c7d705c0b901202c7ffc283f001ae9d6d91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-26 10:15:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8514a9f273145798db91d27f6cd62b51deb465b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 10:41:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/897f42f19d9a4625a6796eff62ad6047 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a57a452c85435ca0796ba87c0d1522 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dd1b09483114963b32c69a69a29806b : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97743dc98c3242bcb273297de370d3dd : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45d98a89fa7948d899f66d21d2474714 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a73dda52c94f0f82937c7448ded19c : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15634d4d6c3e422da3ff871aca98b2ba : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6828ed5af29663b6d10041d71606cd4cafd64936","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-30 20:25:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a2133b3fe0ef9da9191717184603de87b99af3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 20:25:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e377ee7660369d344b608c44dd18e580c394b44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 20:40:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29d96b31aa4241a7bcec326a6d537a66 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56bd658cebc141e792b2620590fe7e8f : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d179b11c87b4a218a12e2e0743e3a43 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e5fd94f005b491e9b7284fbbf0b0a09 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2655382db9e41a4b5ef24b372b1785f : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"7abfa13a425f97773d1aa3f3938345b889ca7311","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 20:40:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"39a1cf1b49e648a0e91f1b72a2b420a85899766b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 20:41:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc57533c54f44203adac731d9c746539 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91e05f9a97c04c1d8afd6c01744a54d6 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c31e86828c7ee0a324de46d22ee67a3f0fcee50","revisions":{"9c31e86828c7ee0a324de46d22ee67a3f0fcee50":{"kind":"REWORK","_number":1,"created":"2021-03-26 10:15:55.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/39/783239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/39/783239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/783239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/783239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/783239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/39/783239/1"}}},"commit":{"parents":[{"commit":"0e11b9c4ecc89b2af4656b67908be21abc62691e","subject":"Update .gitreview for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/0e11b9c4ecc89b2af4656b67908be21abc62691e"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-26 10:15:54.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-26 10:15:54.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","message":"Update TOX_CONSTRAINTS_FILE for stable/wallaby\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/wallaby branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I4a806ca1bcc8a7dd506b7121fde9cfe6fb34dcd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/9c31e86828c7ee0a324de46d22ee67a3f0fcee50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/9c31e86828c7ee0a324de46d22ee67a3f0fcee50"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
