)]}'
{"id":"openstack%2Fbarbican~798115","triplet_id":"openstack%2Fbarbican~stable%2Fqueens~If1b493707bd012d4a779dc6deba1a70ac3293ec6","project":"openstack/barbican","branch":"stable/queens","topic":"octavia-v1","hashtags":[],"change_id":"If1b493707bd012d4a779dc6deba1a70ac3293ec6","subject":"[Rocky and before] Drop octavia v1 job","status":"ABANDONED","created":"2021-06-28 13:09:34.000000000","updated":"2022-12-05 18:41:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":798295,"cherry_pick_of_patch_set":2,"meta_rev_id":"b1c5e9c4694371ca3e54d323d4f1fb879bfc06ed","_number":798115,"virtual_id_number":798115,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-28 14:22:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-28 14:22:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa55c59045cbb937877f5882a0978469bea13352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 13:09:34.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"7e097fe3910116a1da7c0440ea8ffe846995e4c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 13:11:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"95103f1de923e418817d640f7ea3f5eec7d97a6b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 13:11:51.000000000","message":"Topic octavia-v1-stable/queens removed","accounts_in_message":[],"_revision_number":2},{"id":"4f41b0c002f23a5bc4c0244d35bd5d2d31b93889","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 13:11:55.000000000","message":"Topic set to octavia-v1","accounts_in_message":[],"_revision_number":2},{"id":"b2b9744b2d1b570740c4ac0508dd5297a55a5b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:22:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eb9f610f26541a59c30343868d55ee7 : SUCCESS in 7m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/814eaa9f0a174c40a21b7cda7ef3e587 : FAILURE in 6m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9067e4f6bea94001bb03f1b924403181 : SUCCESS in 8m 50s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/30fcdeeec6b64749bcea6a3e2a9579bb : RETRY_LIMIT in 5m 39s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/e196066fd5fb4a4bafa3f6360aa2f4e0 : FAILURE in 34m 51s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/df336dd6f8034bc4aa090675958612a5 : FAILURE in 22m 05s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/b1725746487a466dbbfa044d4256eeb0 : FAILURE in 28m 37s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/f4554e61974a4929966dac0f15e8be9f : FAILURE in 23m 17s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/f8155ab4f2cb4d83b217bd6eec7ff8a2 : FAILURE in 14m 31s\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/c027a874ba8d4711b867fae8d83c9dd7 : FAILURE in 22m 12s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/08b0df4c57a54582952906724bb6b111 : FAILURE in 32m 27s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ad3a536cf34544048b8b306b46382bb5 : FAILURE in 1h 09m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b14ec3e53df43b0adf725b967603908 : FAILURE in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b1c5e9c4694371ca3e54d323d4f1fb879bfc06ed","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-12-05 18:41:50.000000000","message":"Abandoned\n\nThis branch of the project has been moved to End of Life. In order to be able to delete the branch open patches need to be abandoned.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02b2460c7145a7554f482bcd6a97cf3421a66353","revisions":{"72f3e43860c3e7224ed6a6cb7ceb1f8b0a5e45b1":{"kind":"REWORK","_number":1,"created":"2021-06-28 13:09:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/798115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/15/798115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/798115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/798115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/798115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/15/798115/1"}}},"commit":{"parents":[{"commit":"9436e5a2e2cb8e9e858eee2c9ecf9d5299b5faf2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/9436e5a2e2cb8e9e858eee2c9ecf9d5299b5faf2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:07:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:09:34.000000000","tz":0},"subject":"[Rocky and before] Drop octavia v1 job","message":"[Rocky and before] Drop octavia v1 job\n\nSince Octavia has EOLed its stable/stein and older branches, there is\nno longer octavia v1 job defined. Because the job has been non-voting\nlet\u0027s remove the job to resolve job definition error.\n\nChange-Id: If1b493707bd012d4a779dc6deba1a70ac3293ec6\n(cherry picked from commit 8e1e2327df473a41b639fcb28ebe67c7c66f94f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/72f3e43860c3e7224ed6a6cb7ceb1f8b0a5e45b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/72f3e43860c3e7224ed6a6cb7ceb1f8b0a5e45b1"}]},"branch":"refs/heads/stable/queens"},"02b2460c7145a7554f482bcd6a97cf3421a66353":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-28 13:11:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/798115/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/15/798115/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/798115/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/798115/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/15/798115/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/15/798115/2"}}},"commit":{"parents":[{"commit":"9436e5a2e2cb8e9e858eee2c9ecf9d5299b5faf2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/9436e5a2e2cb8e9e858eee2c9ecf9d5299b5faf2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:07:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:11:30.000000000","tz":540},"subject":"[Rocky and before] Drop octavia v1 job","message":"[Rocky and before] Drop octavia v1 job\n\nSince Octavia EOLed its stable/stein and older branches, there is no\nlonger octavia v1 job defined. Because the job has been non-voting\nlet\u0027s remove the job to resolve job definition error.\n\nChange-Id: If1b493707bd012d4a779dc6deba1a70ac3293ec6\n(cherry picked from commit 863c714bdcb07534ad7cc5991a6d9a1afc655e86)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/02b2460c7145a7554f482bcd6a97cf3421a66353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/02b2460c7145a7554f482bcd6a97cf3421a66353"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
