)]}'
{"id":"openstack%2Fbarbican~798295","triplet_id":"openstack%2Fbarbican~stable%2Frocky~If1b493707bd012d4a779dc6deba1a70ac3293ec6","project":"openstack/barbican","branch":"stable/rocky","topic":"octavia-v1","hashtags":[],"change_id":"If1b493707bd012d4a779dc6deba1a70ac3293ec6","subject":"[Rocky and before] Drop octavia v1 job","status":"ABANDONED","created":"2021-06-28 13:08:38.000000000","updated":"2022-12-05 18:48:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ad2fa2381b03b6aab06d3621d6a071a4ab9b45cc","_number":798295,"virtual_id_number":798295,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-26 02:18:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-28 13:48:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c290bb04aee63fd45538e6447552771f8c6638a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 13:08:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3823b24cf1f4d956ca521d3a1f5bbd39698423fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 13:09:04.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1b9696c75ab645f381dcbbd3296350d91289b641","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 13:09:17.000000000","message":"Topic set to octavia-v1","accounts_in_message":[],"_revision_number":2},{"id":"07f481ebf568a1b04ac8a5c78bc24e0cf9e978c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 13:11:11.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"44de95d5acb5ef6b4f8ad550b6c9e2709a42ea2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 13:48:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dff900e037154b9e9886d8762b07b54f : SUCCESS in 7m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51d3f349a103481d9e6e6b72bc73b92a : FAILURE in 6m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3223aef4da44450d8806f914bcad62f4 : SUCCESS in 8m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b5e7383c87444f52849476a29aa836dd : SUCCESS in 7m 42s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/dca55e15138f448393848e364368e3b5 : FAILURE in 35m 13s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/8578807dcf66489dbeae95cc910f5b97 : FAILURE in 22m 49s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/705d47cee1584fd1bc598f14fa9bd637 : FAILURE in 26m 49s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/f6902e222316494fafb31319d211d8f3 : FAILURE in 23m 15s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/46759f95c06444499a040843e0fa2d97 : FAILURE in 14m 50s (non-voting)\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/196b9e406dfb4ee59f541bb3acbe434c : FAILURE in 18m 35s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/edac1f9e9ac940b98d11f1afd9c4a5f4 : FAILURE in 23m 06s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/80e72f4d0f804ef3baa4be727f15e401 : FAILURE in 33m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5013a2ffcad64d8f97945626870f9903 : FAILURE in 5m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04fb52cb4efa4870a28573e1dc764a13 : FAILURE in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a84219878f4dff8b0eeac0cba077a7c8d2b1f62c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-26 02:05:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8d8a4d6776500ff9e96a18bd856dca5b8fbd2837","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-26 02:06:54.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"3f1f00ce9a9c5da4c45ac9f73e095b5d64b1d8f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 02:18:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc4d0b962534cec9cc0ceb55c0c4abe : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/495db1014c174b79aecc2c6959a11806 : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a55252c5446f4949ae3f536bfc5622df : SUCCESS in 6m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7fd5b733420a49e1bb5f77727660dd91 : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98d0e4db3e244304b8cf5a031d60c4c3 : SUCCESS in 4m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a610c5bed8d04a75b44a6a72ef58c04d : FAILURE in 2m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfed9cf9e54846ef81201458c5f1f30b : FAILURE in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8380c856ea945d1a877c127e58f5f0c : FAILURE in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ad2fa2381b03b6aab06d3621d6a071a4ab9b45cc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-12-05 18:48:50.000000000","message":"Abandoned\n\nThis branch of the project has been moved to End of Life. In order to be able to delete the branch open patches need to be abandoned.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"827eb5c86c4811108f57daa193f987ccfceed4a5","revisions":{"b1ae3285080ea9c402ea4d73ba3fa02fc8668f01":{"kind":"REWORK","_number":1,"created":"2021-06-28 13:08:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/798295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/95/798295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/95/798295/1"}}},"commit":{"parents":[{"commit":"37bc0d00df8a3944cc78824343a1882ed4028dc1","subject":"Fix the barbicanclient installation not from source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37bc0d00df8a3944cc78824343a1882ed4028dc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:07:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:07:19.000000000","tz":540},"subject":"[Rocky and before] Drop octavia v1 job","message":"[Rocky and before] Drop octavia v1 job\n\nSince Octavia has EOLed its stable/stein and older branches, there is\nno longer octavia v1 job defined.\n\nChange-Id: If1b493707bd012d4a779dc6deba1a70ac3293ec6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b1ae3285080ea9c402ea4d73ba3fa02fc8668f01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b1ae3285080ea9c402ea4d73ba3fa02fc8668f01"}]},"branch":"refs/heads/stable/rocky"},"8e1e2327df473a41b639fcb28ebe67c7c66f94f8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-28 13:09:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/798295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/95/798295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/95/798295/2"}}},"commit":{"parents":[{"commit":"37bc0d00df8a3944cc78824343a1882ed4028dc1","subject":"Fix the barbicanclient installation not from source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37bc0d00df8a3944cc78824343a1882ed4028dc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:07:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:08:42.000000000","tz":540},"subject":"[Rocky and before] Drop octavia v1 job","message":"[Rocky and before] Drop octavia v1 job\n\nSince Octavia has EOLed its stable/stein and older branches, there is\nno longer octavia v1 job defined. Because the job has been non-voting\nlet\u0027s remove the job to resolve job definition error.\n\nChange-Id: If1b493707bd012d4a779dc6deba1a70ac3293ec6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/8e1e2327df473a41b639fcb28ebe67c7c66f94f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/8e1e2327df473a41b639fcb28ebe67c7c66f94f8"}]},"branch":"refs/heads/stable/rocky"},"863c714bdcb07534ad7cc5991a6d9a1afc655e86":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-06-28 13:11:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/798295/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/95/798295/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/95/798295/3"}}},"commit":{"parents":[{"commit":"37bc0d00df8a3944cc78824343a1882ed4028dc1","subject":"Fix the barbicanclient installation not from source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37bc0d00df8a3944cc78824343a1882ed4028dc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:07:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:10:46.000000000","tz":540},"subject":"[Rocky and before] Drop octavia v1 job","message":"[Rocky and before] Drop octavia v1 job\n\nSince Octavia EOLed its stable/stein and older branches, there is no\nlonger octavia v1 job defined. Because the job has been non-voting\nlet\u0027s remove the job to resolve job definition error.\n\nChange-Id: If1b493707bd012d4a779dc6deba1a70ac3293ec6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/863c714bdcb07534ad7cc5991a6d9a1afc655e86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/863c714bdcb07534ad7cc5991a6d9a1afc655e86"}]},"branch":"refs/heads/stable/rocky"},"737611edd73f8697bd67164fcfce8da67c165baa":{"kind":"REWORK","_number":4,"created":"2021-07-26 02:05:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/798295/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/95/798295/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/95/798295/4"}}},"commit":{"parents":[{"commit":"37bc0d00df8a3944cc78824343a1882ed4028dc1","subject":"Fix the barbicanclient installation not from source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37bc0d00df8a3944cc78824343a1882ed4028dc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:07:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-26 02:04:07.000000000","tz":540},"subject":"[Rocky and before] Drop octavia v1 job","message":"[Rocky and before] Drop octavia v1 job\n\nSince Octavia EOLed its stable/stein and older branches, there is no\nlonger octavia v1 job defined. Because the job has been non-voting\nlet\u0027s remove the job to resolve job definition error.\n\nThis change also cap bandit to v1.6.2 because bandit 1.6.3[1] release\nhas dropped support for Python 2[2].\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n\nChange-Id: If1b493707bd012d4a779dc6deba1a70ac3293ec6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/737611edd73f8697bd67164fcfce8da67c165baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/737611edd73f8697bd67164fcfce8da67c165baa"}]},"branch":"refs/heads/stable/rocky"},"827eb5c86c4811108f57daa193f987ccfceed4a5":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-07-26 02:06:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/798295/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/95/798295/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/95/798295/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/95/798295/5"}}},"commit":{"parents":[{"commit":"37bc0d00df8a3944cc78824343a1882ed4028dc1","subject":"Fix the barbicanclient installation not from source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37bc0d00df8a3944cc78824343a1882ed4028dc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:07:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-26 02:06:42.000000000","tz":540},"subject":"[Rocky and before] Drop octavia v1 job","message":"[Rocky and before] Drop octavia v1 job\n\nSince Octavia EOLed its stable/stein and older branches, there is no\nlonger octavia v1 job defined. Because the job has been non-voting\nlet\u0027s remove the job to resolve job definition error.\n\nThis change also caps bandit to v1.6.2 because bandit 1.6.3[1] release\nhas dropped support for Python 2[2].\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n\nChange-Id: If1b493707bd012d4a779dc6deba1a70ac3293ec6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/827eb5c86c4811108f57daa193f987ccfceed4a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/827eb5c86c4811108f57daa193f987ccfceed4a5"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
