)]}'
{"id":"openstack%2Fbarbican~798299","triplet_id":"openstack%2Fbarbican~stable%2Focata~If1b493707bd012d4a779dc6deba1a70ac3293ec6","project":"openstack/barbican","branch":"stable/ocata","topic":"octavia-v1","hashtags":[],"change_id":"If1b493707bd012d4a779dc6deba1a70ac3293ec6","subject":"[Rocky and before] Drop octavia v1 job","status":"ABANDONED","created":"2021-06-28 13:12:50.000000000","updated":"2021-07-08 10:29:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84e8196f040347e8939a8aff9d1c0e12fb2d8a92","_number":798299,"virtual_id_number":798299,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-28 13:25:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-28 13:25:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2cab67abf40b98f295456b75db6d5d0242787f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 13:12:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8edfa8ea1c9fcb7c91078f0a2c72a0e6efaf5135","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 13:13:09.000000000","message":"Topic set to octavia-v1","accounts_in_message":[],"_revision_number":1},{"id":"5d35d4b3414eacef98dd66137b764964a9b59b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 13:25:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e04149d1dab435ea59acdbd45e84133 : FAILURE in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fe48f2ca79c4dfb869b9012e791e459 : SUCCESS in 7m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d27f6fa78dc84e1d8f9297ee5e66ad2c : SUCCESS in 7m 50s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9ad05a8e415543279b3e44f64d9c34d9 : SUCCESS in 7m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4ee01c7b42846989ff11fea0190a725 : FAILURE in 8m 24s (non-voting)\n\nWarning:\n  Comments left for invalid file functionaltests/api/v1/functional/test_secrets.py","accounts_in_message":[],"_revision_number":1},{"id":"7b59c9c4e0cc29bdb4beb48fc63e6a009c40d804","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-07-08 10:14:03.000000000","message":"Abandoned\n\n@Takashi: as the branch will be deleted we don\u0027t need this to land.\n\nstable/ocata of barbican is tagged as End-of-Life. To be able to delete stable/ocata we need to abandon every open patch.","accounts_in_message":[],"_revision_number":1},{"id":"84e8196f040347e8939a8aff9d1c0e12fb2d8a92","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-08 10:29:17.000000000","message":"Patch Set 1:\n\nAh, sorry I was not aware that EOL was going on[1]\n\n[1] https://review.opendev.org/c/openstack/releases/+/795647\n\nMy intention was to fix the issue in all open branches so I\u0027m ok with just abandoning this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11b0e419a1df88fe508c73b5ee4cd0318aeaa5dd","revisions":{"11b0e419a1df88fe508c73b5ee4cd0318aeaa5dd":{"kind":"REWORK","_number":1,"created":"2021-06-28 13:12:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/798299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/99/798299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/798299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/798299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/99/798299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/99/798299/1"}}},"commit":{"parents":[{"commit":"f5101f461a2e551d221ba4b678a67c5297549904","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f5101f461a2e551d221ba4b678a67c5297549904"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:07:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 13:12:43.000000000","tz":540},"subject":"[Rocky and before] Drop octavia v1 job","message":"[Rocky and before] Drop octavia v1 job\n\nSince Octavia EOLed its stable/stein and older branches, there is no\nlonger octavia v1 job defined. Because the job has been non-voting\nlet\u0027s remove the job to resolve job definition error.\n\nChange-Id: If1b493707bd012d4a779dc6deba1a70ac3293ec6\n(cherry picked from commit 863c714bdcb07534ad7cc5991a6d9a1afc655e86)\n(cherry picked from commit 02b2460c7145a7554f482bcd6a97cf3421a66353)\n(cherry picked from commit f3ee6102d1cb5cb13975033a020a00a3ef4e9acc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/11b0e419a1df88fe508c73b5ee4cd0318aeaa5dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/11b0e419a1df88fe508c73b5ee4cd0318aeaa5dd"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
