)]}'
{"id":"openstack%2Fbarbican~808656","triplet_id":"openstack%2Fbarbican~stable%2Frocky~I283dc51a18aa7b0521cef8cc15a9a356995a119e","project":"openstack/barbican","branch":"stable/rocky","topic":"cleanup_featuresets_old_branch-stable/rocky","hashtags":[],"change_id":"I283dc51a18aa7b0521cef8cc15a9a356995a119e","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-13 09:19:38.000000000","updated":"2022-12-05 18:48:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":808692,"cherry_pick_of_patch_set":1,"meta_rev_id":"dba748ba40e64e94122f9d00871353a2841057e2","_number":808656,"virtual_id_number":808656,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-13 11:17:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 09:21:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d26a4a12a928549d17f92c8fa5a44bb1238c869","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-13 09:19:38.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"c0123957c49804e7a1c6667314f24d1d60a26b3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 09:21:33.000000000","message":"Patch Set 1: Verified-1\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":1},{"id":"6ce7daf7a273b05d939cfad6aa81163a353b8373","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-13 09:21:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"27376a06c1e72c16bf2f2bb9c7289df8079a7743","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-13 09:21:56.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"130d203085e993827e90e5208496c649217f6ee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 09:22:27.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/barbican on branch stable/rocky.  The error was:\n\n  Job octavia-v1-dsvm-scenario not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs\n        - openstack-python35-jobs\n        - publish-openstack-sphinx-docs\n        - release-notes-jobs\n        - check-requirements\n      check:\n        jobs:\n          - barbican-simple-crypto-devstack-tempest\n          - barbican-simple-crypto-devstack-functional\n          - barbican-simple-crypto-devstack-functional-py35\n          - barbican-simple-crypto-devstack-functional-postgres\n          - barbican-dogtag-devstack-functional-fedora-latest\n          - barbican-kmip-devstack-functional:\n              voting: false\n          - barbican-vault-devstack-functional\n          - grenade-devstack-barbican:\n              voting: false\n          - openstack-tox-lower-constraints\n          - openstack-tox-cover:\n              voting: false\n          - octavia-v1-dsvm-scenario:\n              voting: false\n      gate:\n        queue: barbican\n        jobs:\n          - barbican-simple-crypto-devstack-tempest\n          - barbican-simple-crypto-devstack-functional\n          - barbican-simple-crypto-devstack-functional-py35\n          - barbican-kmip-devstack-functional:\n              voting: false\n          # TODO(redrobot): add this back once job is voting again\n          # - barbican-dogtag-devstack-functional-fedora-latest\n          - barbican-vault-devstack-functional\n          - grenade-devstack-barbican:\n              voting: false\n          - openstack-tox-lower-constraints\n      experimental:\n        jobs:\n          - barbican-simple-crypto-devstack-tempest-py35\n\n  in \"openstack/barbican/.zuul.yaml@stable/rocky\", line 163, column 3","accounts_in_message":[],"_revision_number":2},{"id":"4ef98ad4c153420deb1eadaae5f65754dd5af98d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-13 10:31:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"17b0f19761eb40e008ab8e2373b0267a19293c62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 11:17:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fdb4533df3e4a96a4ad82ef06eb4bf8 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c4c150125284f2bb92ead752f96d03a : FAILURE in 3m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/43009ffa56014136b572b03dddeb8849 : SUCCESS in 6m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1dbdf657c19e491a852b8818b578bb37 : SUCCESS in 5m 22s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/96339729d3644cc591b288176ff0b850 : FAILURE in 40m 41s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/5698e4e0185943bc903b3ca5349d6fc8 : FAILURE in 22m 56s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/f2334ac3d8e34f7582011b7309d532bc : FAILURE in 28m 11s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/ba3a804e69a04b8db17752aeab2d403d : FAILURE in 22m 10s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/40716f4c5ce04932aab6475606580ec3 : FAILURE in 16m 19s (non-voting)\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/d6eacae796d54ec5ae8d61d564d67705 : FAILURE in 23m 14s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/cf1a63c18b2747bbb0c798a2037406e8 : FAILURE in 24m 03s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/4a177cde09d94f30961f34cc38dd8ee0 : FAILURE in 34m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/854da2515636415cbefb3d7d83dade5e : FAILURE in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd6264a3166242bd98e1e53505eae598 : FAILURE in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"27ab8dae130432a14df336e6e248a1d857b96f1e","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-20 06:16:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"71d1958aa98c76047101e077cf0e72452e066920","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 07:53:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d03c5289d8a4defba9b864a7c734c9c : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/83d68ecf52bb4712ab918ebfe01d8d45 : FAILURE in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2375617af1b440ce89a68673f8f35e74 : SUCCESS in 6m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/861f48039a384004b7762a628252a558 : SUCCESS in 6m 10s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/b36adc8ff28e4320920d92d5f8175a21 : FAILURE in 35m 02s\n- barbican-simple-crypto-devstack-functional https://zuul.opendev.org/t/openstack/build/c58803cf36804d24b71b3d5b629d162a : FAILURE in 24m 55s\n- barbican-simple-crypto-devstack-functional-py35 https://zuul.opendev.org/t/openstack/build/76e7552872dd4b908a29874685b2c04d : FAILURE in 25m 25s\n- barbican-simple-crypto-devstack-functional-postgres https://zuul.opendev.org/t/openstack/build/06bdb69126fd41d9967f0589655aa7bf : FAILURE in 21m 05s (non-voting)\n- barbican-dogtag-devstack-functional-fedora-latest https://zuul.opendev.org/t/openstack/build/4ec4c934dad34c068435d7ac0d3add64 : FAILURE in 15m 57s (non-voting)\n- barbican-kmip-devstack-functional https://zuul.opendev.org/t/openstack/build/18ea1ff8a94444c7a8e7ede4fcf16251 : FAILURE in 16m 35s (non-voting)\n- barbican-vault-devstack-functional https://zuul.opendev.org/t/openstack/build/25a95d30fc9e48f0b7d51982e92e259a : FAILURE in 24m 30s (non-voting)\n- grenade-devstack-barbican https://zuul.opendev.org/t/openstack/build/4a691f0976464265a0fbda207bd6593f : FAILURE in 39m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/587cfbaf5069406a8c8237f2232aadce : FAILURE in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f456e5d37b6b4beb8ac6e0175725f44b : FAILURE in 8m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dba748ba40e64e94122f9d00871353a2841057e2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-12-05 18:48:45.000000000","message":"Abandoned\n\nThis branch of the project has been moved to End of Life. In order to be able to delete the branch open patches need to be abandoned.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bc4fae90fd1016d552c74917c8ed0818712fcb93","revisions":{"b88c14bfd0a5d62a0261c8b14c04b13805684c9a":{"kind":"REWORK","_number":1,"created":"2021-09-13 09:19:38.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/56/808656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/56/808656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/56/808656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/56/808656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/56/808656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/56/808656/1"}}},"commit":{"parents":[{"commit":"37bc0d00df8a3944cc78824343a1882ed4028dc1","subject":"Fix the barbicanclient installation not from source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37bc0d00df8a3944cc78824343a1882ed4028dc1"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 09:17:53.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 09:19:38.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I283dc51a18aa7b0521cef8cc15a9a356995a119e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b88c14bfd0a5d62a0261c8b14c04b13805684c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b88c14bfd0a5d62a0261c8b14c04b13805684c9a"}]},"branch":"refs/heads/stable/rocky"},"7e4127a9d5018cd9d8c50e3289d9b62eb68a8e27":{"kind":"REWORK","_number":2,"created":"2021-09-13 09:21:47.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/56/808656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/56/808656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/56/808656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/56/808656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/56/808656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/56/808656/2"}}},"commit":{"parents":[{"commit":"37bc0d00df8a3944cc78824343a1882ed4028dc1","subject":"Fix the barbicanclient installation not from source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37bc0d00df8a3944cc78824343a1882ed4028dc1"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 09:17:53.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 09:21:34.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I283dc51a18aa7b0521cef8cc15a9a356995a119e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/7e4127a9d5018cd9d8c50e3289d9b62eb68a8e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/7e4127a9d5018cd9d8c50e3289d9b62eb68a8e27"}]},"branch":"refs/heads/stable/rocky"},"bc4fae90fd1016d552c74917c8ed0818712fcb93":{"kind":"REWORK","_number":3,"created":"2021-09-13 10:31:11.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/56/808656/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/56/808656/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/56/808656/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/56/808656/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/56/808656/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/56/808656/3"}}},"commit":{"parents":[{"commit":"37bc0d00df8a3944cc78824343a1882ed4028dc1","subject":"Fix the barbicanclient installation not from source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/37bc0d00df8a3944cc78824343a1882ed4028dc1"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 09:17:53.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 10:30:59.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I283dc51a18aa7b0521cef8cc15a9a356995a119e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bc4fae90fd1016d552c74917c8ed0818712fcb93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bc4fae90fd1016d552c74917c8ed0818712fcb93"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
