)]}'
{"id":"openstack%2Fbarbican~828829","triplet_id":"openstack%2Fbarbican~master~I5651a69f93288ac1dfdc1c8b1ad0f904e370c127","project":"openstack/barbican","branch":"master","topic":"secret-consumers","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-05-05 22:16:07.000000000","reason":"Douglas Mendizábal replied on the change"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-05-06 17:44:07.000000000","reason":"Change was submitted"},"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-05-06 17:44:07.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-05-06 17:44:07.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-02-21 18:30:08.000000000","reason":"Ade Lee replied on the change"}},"hashtags":[],"change_id":"I5651a69f93288ac1dfdc1c8b1ad0f904e370c127","subject":"Implement microversions, bring back secret consumers API","status":"MERGED","created":"2022-02-11 10:01:24.000000000","updated":"2022-05-06 17:45:18.000000000","submitted":"2022-05-06 17:44:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"828829-secret-consumers","meta_rev_id":"b08cac0f591544d7af6159f1551d2b188df67492","_number":828829,"virtual_id_number":828829,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-06 17:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2022-05-06 17:12:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2022-05-05 17:07:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2022-05-06 17:12:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-11 11:55:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-11 14:28:13.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-02-16 14:19:44.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2022-05-05 22:16:07.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"state":"REVIEWER"},{"updated":"2022-05-05 22:16:07.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"7300ccf436594fd68fa229bd112983173945c94d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-11 10:01:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afafef43215f257a712c5f5263467c8c65da8387","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-11 10:01:47.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"04cbbbafdf49dee8fd2539ec8b9a9b8d34a83185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 11:55:08.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/368b2dd41f70456a865bc80eabccd659 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/648d14b794254b0791194ded6765bde5 : FAILURE in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dc18abea66a4cb7a74f7530fc726a48 : FAILURE in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e4e8fff4d4045908a8e05373bf96f2d : FAILURE in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a11ff06fb0d54ceb9bf171b8572e50c6 : FAILURE in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20731ab8fb1d4e97897834fd0efea2bc : SUCCESS in 7m 15s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/f4dea7baa12846a088053fcf128c1623 : SUCCESS in 35m 50s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/14111c61b5564639aa7e73f582eef45e : SUCCESS in 43m 12s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/d5217387e9d3464891ec8dd74300e965 : SUCCESS in 53m 16s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/de144eb1da5c46429742fb46f92fbb72 : SUCCESS in 34m 48s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/672e93b375754c7291c355ab25d438fb : SUCCESS in 35m 39s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/a8446ed2d1d045b99f08deb8117e2c2a : SUCCESS in 35m 14s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/a2d9d7a20c0a426d8f713b5f7af043d4 : SUCCESS in 46m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4ca3535dd34049009516511be6f5cbae : SUCCESS in 35m 20s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/16e71eec218046f296900b7f679dd4e9 : FAILURE in 1h 50m 27s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b0124bf8729d4856807ca1150d019c7f : SUCCESS in 1h 00m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9610c54f8f8f775e8cb2c6f73e59a65f33019c7a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-11 14:12:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"25d7045466d820f24c93da4edceb9c0c83a96978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 15:55:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ece7ad1a07f349fd821f63d82e53adbf : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/358471167e7f42fd8718caf67bd8a5a6 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e87b224ecf14497b7218f0c3e38d98b : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b945845999c445bfb828be074d8e5cd2 : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe04d6ee2f2a4f3c80eeeb31ea62f979 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba1f959d48ab4bb19bda75b99244053e : SUCCESS in 6m 37s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/d5ae4c49e7f8479693ccbbdd329f82da : SUCCESS in 35m 11s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/f77bfcf3ffcb4997a74780c8475a978a : SUCCESS in 28m 25s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/ddd6099c2baf4c1f9881aa856d76ff11 : SUCCESS in 52m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/636de1e003534f3e92e07e2af52523b5 : SUCCESS in 31m 02s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/8f1caa5d10ed4de7a36ced1904fe4d0f : SUCCESS in 32m 01s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/7e7c88bc1f4b4c7e870d09ebca8475f5 : SUCCESS in 38m 37s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/a0527aa517d842bcb6ae833f424c9bc2 : SUCCESS in 37m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/822db381ae6c4f43a4406ddb86c1a6c8 : SUCCESS in 35m 59s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f8f3acf0e35a42ec915e711d508c2bbf : FAILURE in 1h 40m 03s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/329c48f3f1da423ba9f5097ad404c11c : SUCCESS in 58m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ce9cb5e911daae4b4ae2185fbd8fc74efb37468","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-16 14:09:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ffe26ce25d98dba79e9c6d259e9b0d8632d77ab7","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-16 14:19:44.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"e2e740f1d14084f06acecd140611457a30eff7ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:11:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/41c34264a10c459e84ecc9489025bf9b : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9a04ed974f549d396ef310e3338db1d : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf378969a0a4dd8828a32b4231d407a : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/664597fd141f416eaa30d269ed705a88 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/292e9f02c094478b92f604db738ec90d : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2956241267e549f5921b667cf80849f6 : SUCCESS in 8m 48s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/35dbe71d023845bab9394bb550907897 : SUCCESS in 36m 39s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/8d09ec8c500643a0adb65ce0558606f3 : SUCCESS in 41m 36s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/853c5aecc6124bf384e17c8ee4bda1a1 : SUCCESS in 1h 04m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/cc7cb0bdfd3d43d5b97bb93781eddf39 : SUCCESS in 41m 59s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/90d14fbc84ff4c328d840177f58fe4c9 : SUCCESS in 40m 43s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/5b3b83c252544e00bfa5731e7158fd60 : SUCCESS in 41m 36s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/9aab142aa3a4431d9ac2ba9eb36275cd : SUCCESS in 50m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1b9760147be467f9c4a37e1c4603399 : SUCCESS in 1h 00m 56s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/00dedb05e0c946de85e7da9ce082fb8b : FAILURE in 1h 56m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c1528844f6bbce081b828f2c968dfc67dd7b8af8","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-21 18:30:08.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c8b978b9c0bfa6d8de6bec42ebc7449e0ae656ef","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-02-24 20:57:55.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7ff07608df0540ae1f11cc49b33fdc030230ba05","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-24 21:40:56.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b8f0c4ce7ddb1274a04af016287e813e9f824956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-03-04 09:58:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"33f24e7a60f71efe0df33827c9d77dae907ec699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 11:47:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1a685149bfed41b7aef00b6b83da1b76 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80cd6b194d9f44829a9ac48fb32262aa : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad154399923040d4a4abdb2a37190dc4 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/335115b983b24663af89ce9694694f3b : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/feef82de0e6d407592ead16633d73fae : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e065712197e4cb896ff7ff55dfc3f6d : SUCCESS in 8m 22s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/1708473058774645b7b9c1c931e12578 : SUCCESS in 53m 58s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/99f0fea7ec314d4d8f55839885c34af0 : SUCCESS in 30m 16s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/014f1098693644cca66c9bd79c4d417d : SUCCESS in 46m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/cefbe2333ee5401aa07a018534112f07 : SUCCESS in 33m 23s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/c6f5ad6d5dab4ab1983921be6464c417 : SUCCESS in 29m 41s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/84a6ea52896e47478314e08a835cd63c : SUCCESS in 31m 10s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/04096412da2740dea040dd6a849168b4 : FAILURE in 38m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/81d2b9edbee548af965556087fcd17fb : SUCCESS in 34m 54s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/82d74cef3b604314837ea42663a8375d : FAILURE in 1h 46m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d863c3a927486c5dad095890ae6261a80216a952","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-03-04 20:56:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d99e135e38a62722b12705631e54c72d52527e12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 22:46:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/784f29ab9e6e4589805a84d7d6eb3131 : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1107e9405bfc4e558f6b0a6ec867d67d : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d0da7fcef4465a8734abf049a5612c : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2a905def84b4c4c89c5bda291c616d0 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2cfba74b1c9432f9b43ef7b8a311495 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc4a7c04f8c4289a76da5fae8cd3a02 : SUCCESS in 8m 27s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/4010dcdd60c945c3b057829b2eed819e : SUCCESS in 41m 02s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/315ae910393242da827bd409412af375 : SUCCESS in 50m 04s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/2433de4821de4a1da2ad511a93a59cd4 : SUCCESS in 52m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1409e83f4e42476cb32f48b1a098bb61 : SUCCESS in 47m 35s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/e9d4c9d7a2094a329354a851e5994bef : SUCCESS in 39m 46s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/10a607ec1bc845548d51f321ec19f982 : SUCCESS in 48m 28s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/920357288fe34fcca44caaf6cc67796a : SUCCESS in 57m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cec476478d6a4dd28e4e699c18583e32 : SUCCESS in 39m 46s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5e1c6d3b253741fb9ed5bcf251be1294 : FAILURE in 1h 44m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d294613fc04928dfa8c9edd5303b2d1ab25e0b83","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-04-25 17:24:09.000000000","message":"Topic set to secret-consumers","accounts_in_message":[],"_revision_number":4},{"id":"c75faa4525aa12d38e44705711ee9962e8bc2945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-04-25 18:51:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"87f09537717c99fa1aee0710e4e399c1cc4c6b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 18:51:56.000000000","message":"Patch Set 5: Verified-1\n\nUnable to freeze job graph: Job tripleo-ci-centos-8-scenario002-standalone does not specify a run playbook","accounts_in_message":[],"_revision_number":5},{"id":"5598be1fd895f4c91bcb589aa412b288a0c114da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-05-03 17:07:25.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"6b7f429f4c9e04e7d91e90ded7e560653c1d177c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:57:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d668214c102a42ecb5ae60c93777b6c5 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d6322d4cf2a4a1287dce3a5413faa85 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75f0fd659b944635b1fa4904d1363b65 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d59499d71d943129800d09bd9c6dd09 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f73165819b654d709eff3a416c4d1d57 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d9a0d0131c94f4eb929a98e61a3d91d : FAILURE in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f613d549e342778c3bf0b9a3577027 : SUCCESS in 6m 07s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/72338da981df427ea0249e01279530dc : SUCCESS in 36m 07s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/f5651bd690274122b5ce83635f33fc20 : SUCCESS in 37m 24s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/2b5fc9d756e347349fdde178ccfceb57 : SUCCESS in 41m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2ac170138fa14357a3d270279cc0ffa9 : SUCCESS in 40m 03s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/94712d721d0f4024a33f3e743eb8e8c6 : SUCCESS in 37m 00s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/9303e3003994403fae6b0d7150b3d4ce : SUCCESS in 36m 13s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/8bb891640e884436b700f49870187dc5 : POST_FAILURE in 13m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c7fc0827e10b4c128f8e39806517ae19 : SUCCESS in 33m 28s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/7a6768b3edb44e91a0f2b1167eb5b138 : SUCCESS in 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"68258cc4d5b21a8ef125874c4d6641ca13b832e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-05-04 15:02:32.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"aefd354233e0172c8feb3654846395020d90d55f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 16:16:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/914fe8585ab04d478fe8cad38f00620b : SUCCESS in 2m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0cd4e091c2642e8a2d7b75232c6068e : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05532557078744c8ba7feb3307111433 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/febd0053dab54bd682504bcf23026e77 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10df5088d9394a10bc358126bb39a08a : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c74a504ceda34b3abd7cf4d915d67298 : FAILURE in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5b1332d7c69435f9774d2ec39de3347 : SUCCESS in 6m 43s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/b95408d0853f4916aeff3da44b5dc80e : SUCCESS in 41m 42s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/2b9f222b9d1347358c84d3b414d76ae4 : SUCCESS in 34m 10s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/42dde665532046a68ee602fe19c8d8b9 : SUCCESS in 1h 05m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ad149c922e3c4d878b0b2affad054f27 : SUCCESS in 46m 12s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/408adb5f1c7a4768b0aba44792992c85 : SUCCESS in 41m 15s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/7b11e41dfe3645fa8eb2e7be42224bd6 : SUCCESS in 44m 56s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/d7be3c2a43f246ae8a567f43adf19db2 : SUCCESS in 37m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2d555616fa494475a31ccf99f36bb56b : SUCCESS in 41m 19s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/898a44cdefa04b61a6dd8e98a927ed33 : SUCCESS in 44m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7156028eef25cfdacdde81b4d9e487db92db2bba","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-05-05 17:07:34.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"610ed66d0b5f6e4c9f0fea6562e6a9044bcb831a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-05-06 17:12:27.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"745c46b6888423950729be51ce25c92ed66d58c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 17:13:00.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5062e9a1f509d6626c3ea121ee307c8762f3814b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 17:44:07.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/12ba7f0426f34746a4ea21cbe4f2ab97 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/723d922ecc114ec78615993e66ff7137 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3155c70b06944b1bd05079329fd552c : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b259bcd5a6634285ad0a0ba22f667be5 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d809576f0910423b8c8290dca931462e : SUCCESS in 6m 08s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/1344b35c581e492bb67becb3bf39817a : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":7},{"id":"95abf0a32316888bf8100a6f930ba564f3a69fac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 17:44:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"b08cac0f591544d7af6159f1551d2b188df67492","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 17:45:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa724fd63b5342578446a32288f83666 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0f74877c7731669745d81e9f8c504704f1987a57","revisions":{"032d66c2501c4cde08adf785fe1d4907b2ce2171":{"kind":"REWORK","_number":1,"created":"2022-02-11 10:01:24.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/29/828829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/29/828829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/29/828829/1"}}},"commit":{"parents":[{"commit":"a9d2b133c8dc3307974f119f9a2b23a4ba82e8ce","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/a9d2b133c8dc3307974f119f9a2b23a4ba82e8ce"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-10 19:38:16.000000000","tz":0},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-11 10:01:02.000000000","tz":0},"subject":"Implement microversions, bring back secret consumers API","message":"Implement microversions, bring back secret consumers API\n\nThis reverts commit 7b14d983e0dce6dcffe9781b05c52335b8203fc7.\n\nChange-Id: I5651a69f93288ac1dfdc1c8b1ad0f904e370c127\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/032d66c2501c4cde08adf785fe1d4907b2ce2171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/032d66c2501c4cde08adf785fe1d4907b2ce2171"}]},"branch":"refs/heads/master"},"b458f2477c2ea3feb7a4833a34e431f3605b5d50":{"kind":"REWORK","_number":2,"created":"2022-02-11 14:12:50.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/29/828829/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/29/828829/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/29/828829/2"}}},"commit":{"parents":[{"commit":"a9d2b133c8dc3307974f119f9a2b23a4ba82e8ce","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/a9d2b133c8dc3307974f119f9a2b23a4ba82e8ce"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-10 19:38:16.000000000","tz":0},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-11 14:11:58.000000000","tz":0},"subject":"Implement microversions, bring back secret consumers API","message":"Implement microversions, bring back secret consumers API\n\nThis reverts commit 7b14d983e0dce6dcffe9781b05c52335b8203fc7.\n\nChange-Id: I5651a69f93288ac1dfdc1c8b1ad0f904e370c127\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b458f2477c2ea3feb7a4833a34e431f3605b5d50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b458f2477c2ea3feb7a4833a34e431f3605b5d50"}]},"branch":"refs/heads/master"},"86d2ba0e9001354c68b88571ed793edfeb7e1306":{"kind":"REWORK","_number":3,"created":"2022-02-16 14:09:12.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/29/828829/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/29/828829/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/29/828829/3"}}},"commit":{"parents":[{"commit":"a9d2b133c8dc3307974f119f9a2b23a4ba82e8ce","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/a9d2b133c8dc3307974f119f9a2b23a4ba82e8ce"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-10 19:38:16.000000000","tz":0},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-16 14:04:01.000000000","tz":0},"subject":"Implement microversions, bring back secret consumers API","message":"Implement microversions, bring back secret consumers API\n\nThe implementation follows nova and implements an is_supported\nfunction, that can be used in controllers, to check the requested\nversion and take different code paths depending on the result.\n\nThis reverts commit 7b14d983e0dce6dcffe9781b05c52335b8203fc7.\n\nChange-Id: I5651a69f93288ac1dfdc1c8b1ad0f904e370c127\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/86d2ba0e9001354c68b88571ed793edfeb7e1306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/86d2ba0e9001354c68b88571ed793edfeb7e1306"}]},"branch":"refs/heads/master"},"0532aa67a353f8e075ab721c90a47f9c7c8f6217":{"kind":"REWORK","_number":4,"created":"2022-03-04 09:58:35.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/29/828829/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/29/828829/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/29/828829/4"}}},"commit":{"parents":[{"commit":"a9d2b133c8dc3307974f119f9a2b23a4ba82e8ce","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/a9d2b133c8dc3307974f119f9a2b23a4ba82e8ce"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-10 19:38:16.000000000","tz":0},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-03-04 09:56:36.000000000","tz":0},"subject":"Implement microversions, bring back secret consumers API","message":"Implement microversions, bring back secret consumers API\n\nThe implementation follows nova and implements an is_supported\nfunction, that can be used in controllers, to check the requested\nversion and take different code paths depending on the result.\n\nThis reverts commit 7b14d983e0dce6dcffe9781b05c52335b8203fc7.\n\nChange-Id: I5651a69f93288ac1dfdc1c8b1ad0f904e370c127\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/0532aa67a353f8e075ab721c90a47f9c7c8f6217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/0532aa67a353f8e075ab721c90a47f9c7c8f6217"}]},"branch":"refs/heads/master"},"f6133b7203becf42f7a2a31da14bf2b20205f70f":{"kind":"REWORK","_number":5,"created":"2022-04-25 18:51:02.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/29/828829/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/29/828829/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/29/828829/5"}}},"commit":{"parents":[{"commit":"a9d2b133c8dc3307974f119f9a2b23a4ba82e8ce","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/a9d2b133c8dc3307974f119f9a2b23a4ba82e8ce"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-10 19:38:16.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-04-25 18:50:54.000000000","tz":-300},"subject":"Implement microversions, bring back secret consumers API","message":"Implement microversions, bring back secret consumers API\n\nThe implementation follows nova and implements an is_supported\nfunction, that can be used in controllers, to check the requested\nversion and take different code paths depending on the result.\n\nThis reverts commit 7b14d983e0dce6dcffe9781b05c52335b8203fc7.\n\nChange-Id: I5651a69f93288ac1dfdc1c8b1ad0f904e370c127\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f6133b7203becf42f7a2a31da14bf2b20205f70f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f6133b7203becf42f7a2a31da14bf2b20205f70f"}]},"branch":"refs/heads/master"},"3d7c73494472e8bdf94f285ca2bfd188c71e9436":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-05-03 17:07:25.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/29/828829/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/29/828829/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/29/828829/6"}}},"commit":{"parents":[{"commit":"b7d95f9738ec48f9a990357bacab3bb73f0f54a3","subject":"Replace TripleO CentOS 8 jobs with CentOS 9 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b7d95f9738ec48f9a990357bacab3bb73f0f54a3"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-10 19:38:16.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-05-03 17:07:25.000000000","tz":0},"subject":"Implement microversions, bring back secret consumers API","message":"Implement microversions, bring back secret consumers API\n\nThe implementation follows nova and implements an is_supported\nfunction, that can be used in controllers, to check the requested\nversion and take different code paths depending on the result.\n\nThis reverts commit 7b14d983e0dce6dcffe9781b05c52335b8203fc7.\n\nChange-Id: I5651a69f93288ac1dfdc1c8b1ad0f904e370c127\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/3d7c73494472e8bdf94f285ca2bfd188c71e9436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/3d7c73494472e8bdf94f285ca2bfd188c71e9436"}]},"branch":"refs/heads/master","description":"Rebase"},"0f74877c7731669745d81e9f8c504704f1987a57":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-05-04 15:02:32.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/29/828829/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/29/828829/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/29/828829/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/29/828829/7"}}},"commit":{"parents":[{"commit":"935c7158b072e0a52634be338d050c1eb3120ac1","subject":"Replace TripleO CentOS 8 jobs with CentOS 9 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/935c7158b072e0a52634be338d050c1eb3120ac1"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-10 19:38:16.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-05-04 15:02:32.000000000","tz":0},"subject":"Implement microversions, bring back secret consumers API","message":"Implement microversions, bring back secret consumers API\n\nThe implementation follows nova and implements an is_supported\nfunction, that can be used in controllers, to check the requested\nversion and take different code paths depending on the result.\n\nThis reverts commit 7b14d983e0dce6dcffe9781b05c52335b8203fc7.\n\nChange-Id: I5651a69f93288ac1dfdc1c8b1ad0f904e370c127\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/0f74877c7731669745d81e9f8c504704f1987a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/0f74877c7731669745d81e9f8c504704f1987a57"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
