)]}'
{"id":"openstack%2Fbarbican~854539","triplet_id":"openstack%2Fbarbican~stable%2Fxena~I6bdac945a6aef822d395dda7f460501ff20619a7","project":"openstack/barbican","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-08-26 08:44:32.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-08-26 08:03:52.000000000","reason":"\u003cGERRIT_ACCOUNT_9914\u003e replied on the change","reason_account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}},"hashtags":[],"change_id":"I6bdac945a6aef822d395dda7f460501ff20619a7","subject":"Set versioned jobs to set microversion correctly","status":"MERGED","created":"2022-08-25 10:24:48.000000000","updated":"2022-08-26 08:46:49.000000000","submitted":"2022-08-26 08:44:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854539","cherry_pick_of_change":854606,"cherry_pick_of_patch_set":1,"meta_rev_id":"83494e8433696bf61111bf0c9f4809f8fa57fd1c","_number":854539,"virtual_id_number":854539,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 08:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":2,"date":"2022-08-26 08:03:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":1,"date":"2022-08-26 08:03:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-25 11:30:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6d2786150478650475450837aa39be647d6b643","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-08-25 10:24:48.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"59ab6415b1acc717771e093803d2f154c9f3771c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 11:30:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4826f8a25c5a4b2f8686ee9e39866191\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54ac1c1c602643919651da035f544a76 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08dbf865b16a45a79f700b2bb48665da : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3cbc3cb53db431a9fe1d47ebff4804c : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a301c72c5d84755b3a8603a6b68e303 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03c98b97adbd4d9cbfe03dce9894e090 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ee5c1c61ea4951b2a177ef7ecae78c : SUCCESS in 6m 46s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/ff15b21222d54a068c30087512859512 : SUCCESS in 28m 50s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/aaa46a87f2504b4f89fd8292462cc805 : POST_FAILURE in 9m 25s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/b8a060b79b124eee90bf0087cdf87d64 : SUCCESS in 29m 17s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/005260cb8d1346e6ba8df9e847e42aaa : SUCCESS in 51m 24s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1914b0962c624460b537d7a52eba794c : FAILURE in 45m 22s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/b747777ecbd441b6b1b0fdeca570f9c9 : FAILURE in 45m 45s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/fd00ea4e4d2d46cb9c3c4e59bcd4d0d5 : RETRY_LIMIT in 20m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f1c360e53e36466ab5292871b48343ed : SUCCESS in 47m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d440929c6b585280a848e9dd5b91973556de6acf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-08-25 12:13:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"51e47b17226068d8c2c326390afc6743eb60acd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 13:28:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f77a4010d934421280ef992f78aa7e50\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9388173394ff41639514961a6a9670bb : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85cd5aecb0444c9d9b50a61b81f9319d : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84e24acf59364663b5a97ab22c02af3a : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87a437631b064df2a41e6d0d2d8add45 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a1a4898ea33488a9337376522017e31 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da26118fc295481eb5a3f6a55ea6b149 : SUCCESS in 7m 19s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/3f6dadbeeb644703b66075ff82d43304 : SUCCESS in 42m 31s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/8991e47cf4b141b8ab74ecde043e16c4 : POST_FAILURE in 20m 56s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/79ae402a67744af7813c84d45f07aeae : SUCCESS in 43m 19s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/68089aad6209466cadcbbefed4feabae : SUCCESS in 1h 01m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/ef5d91a53a0049c1b25040b29c46e949 : SUCCESS in 33m 50s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-xena https://zuul.opendev.org/t/openstack/build/be55c4f16b7245979d75441ec626d159 : SUCCESS in 32m 06s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/1fc5cefcba02462080655b5b7688dadb : RETRY_LIMIT in 11m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b3c8e3ebcbc4b51baa127984f0a0b6f : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":2},{"id":"9f6cfa6d64324fe8111c1755af0adbe7f16ffe00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-08-25 22:06:55.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0a4e8b174d6fd2b317b5459ff44a01d0e5cc3185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 23:20:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8916d00fd7fc4cd78b66bfafec97a6ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e4649e36c694cb4a5c3443fafb0b8de : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a7424958aa4d5089ca0c3305f9a9c0 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a9d0e4dbea044869e58668167ec767d : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/034aff2242104ae28b7a345041919132 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/578c01e91d18483d9b3a305e79aaac9e : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee4fa9c1f594d219240c37cfd84759b : SUCCESS in 6m 13s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/ba855215e792478f866eb93dd8705002 : SUCCESS in 36m 59s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/7e656a50fc9b4c0eb3b32d06c6a48205 : POST_FAILURE in 17m 40s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/4ce9890eccae4b0e987038e452db00d7 : SUCCESS in 38m 42s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/588c82a39b3447b08937317e33d94807 : SUCCESS in 1h 05m 51s (non-voting)\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/61903fcf2ca44adda545c796b101140c : SUCCESS in 39m 45s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-xena https://zuul.opendev.org/t/openstack/build/07b9c660c801469b89ba167f9f75eea6 : SUCCESS in 41m 26s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/b4a585a60f764ace9a6d31e9a9fea647 : SUCCESS in 57m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7ca0c55a688345d9b3b435f9050fbf45 : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":3},{"id":"64212c2b8ff496dc1fac3f2201f26c34530440e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-26 02:47:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1b5e01f07b0662a4bd0c16b2a0d4de64772dd9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:51:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c9c05265fd148eb8e6f1dfd362a2e71\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7178312b11134181a3ed3339f7b556b8 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb212be53eaf4f509ff0138b00cd282d : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de5f303376af47cdb6bb863045c62328 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8b30add078b470d8de6253e080ddcab : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ceebfdbf82a411da2fc3945657a2dc0 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/219a7063ea884713bacc39b2e8ff10ee : SUCCESS in 8m 26s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/66a3b9814d11420ab659669d8c31b5f3 : SUCCESS in 35m 53s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/a79b58ac861f417e8a091a491b6687be : POST_FAILURE in 18m 50s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/3bddde47aba344f1b07b01982485ddc8 : SUCCESS in 34m 28s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/1a48705b838649b4be0c38dfd4a546bf : SUCCESS in 1h 01m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/5c107c20dd1745c9971603889be7e6c2 : SUCCESS in 40m 41s\n- barbican-tempest-plugin-simple-crypto-secure-rbac-xena https://zuul.opendev.org/t/openstack/build/fd531684b81f4cf7875b81f4aadd0c85 : SUCCESS in 42m 14s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-xena https://zuul.opendev.org/t/openstack/build/f1efbea4d64148ddbdb2b214bbfd8fae : SUCCESS in 41m 43s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/3b6ebafc32784578830da8e29fa0c611 : SUCCESS in 47m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/240d7278895142a78a45b01447d84a4c : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":4},{"id":"8cbff14160f96990e1ffc0463dca0e3300d461f1","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-08-26 08:03:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"785647bd1477680578c24e2605bba4f896ff437e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:04:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1e679dcc46732f0c93e588984b3f27ed6422d313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:44:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bcc9b8aa4eb4353a5f0073c42ba2193\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d65954ab134476966d1c1e8f1a171e : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd98a4d4bebd4483a360b8102ed9803e : SUCCESS in 10m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c988249102b14a2a9d910a2b91912eca : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa28f6c9c4474261b3cf8b8b8487d945 : SUCCESS in 8m 14s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/d2dab2457bdd4c4f90ded4b08ae1ad00 : SUCCESS in 39m 40s","accounts_in_message":[],"_revision_number":4},{"id":"951e6e2545a199616794cc9b02c398b8fcebdea6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:44:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"83494e8433696bf61111bf0c9f4809f8fa57fd1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:46:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/304f9797b2424f299c25df405e0bd123\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a71640dc6c4787bb85a8c256ae3dea : SUCCESS in 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6328e38ad9f569b7609cfe6ee8fb54a1698af632","revisions":{"7f49b7b325e735dfcf584a7babcdfa4b7f8ffa50":{"kind":"REWORK","_number":1,"created":"2022-08-25 10:24:48.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/39/854539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/39/854539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/39/854539/1"}}},"commit":{"parents":[{"commit":"bb2779473f786f3bb7b84bc4f445089da8cbfe88","subject":"Allow users with \"creator\" role to edit ACLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bb2779473f786f3bb7b84bc4f445089da8cbfe88"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-08-25 10:07:39.000000000","tz":120},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-08-25 10:24:48.000000000","tz":0},"subject":"Set microversions to prevent consumer tests being run","message":"Set microversions to prevent consumer tests being run\n\nConsumers were only added in zed, including a new microversion\nof 1.1.\n\nChange-Id: I6bdac945a6aef822d395dda7f460501ff20619a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/7f49b7b325e735dfcf584a7babcdfa4b7f8ffa50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/7f49b7b325e735dfcf584a7babcdfa4b7f8ffa50"}]},"branch":"refs/heads/stable/xena"},"c9a5da154be0ebebbf7a08e830eae9c4be0ce565":{"kind":"REWORK","_number":2,"created":"2022-08-25 12:13:26.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/39/854539/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/39/854539/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/39/854539/2"}}},"commit":{"parents":[{"commit":"bb2779473f786f3bb7b84bc4f445089da8cbfe88","subject":"Allow users with \"creator\" role to edit ACLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bb2779473f786f3bb7b84bc4f445089da8cbfe88"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-08-25 10:07:39.000000000","tz":120},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-08-25 12:13:15.000000000","tz":0},"subject":"Set versioned jobs to set microversion correctly","message":"Set versioned jobs to set microversion correctly\n\nPrevent 1.1 (secret consumer) tests being run.\n\nDepends-On: https://review.opendev.org/c/openstack/barbican-tempest-plugin/+/854614\n\nChange-Id: I6bdac945a6aef822d395dda7f460501ff20619a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/c9a5da154be0ebebbf7a08e830eae9c4be0ce565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/c9a5da154be0ebebbf7a08e830eae9c4be0ce565"}]},"branch":"refs/heads/stable/xena"},"853f7568e670c8b3d0572e84d56a6b92e7f0b75b":{"kind":"REWORK","_number":3,"created":"2022-08-25 22:06:55.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/39/854539/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/39/854539/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/39/854539/3"}}},"commit":{"parents":[{"commit":"bb2779473f786f3bb7b84bc4f445089da8cbfe88","subject":"Allow users with \"creator\" role to edit ACLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bb2779473f786f3bb7b84bc4f445089da8cbfe88"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-08-25 10:07:39.000000000","tz":120},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-08-25 22:06:51.000000000","tz":0},"subject":"Set versioned jobs to set microversion correctly","message":"Set versioned jobs to set microversion correctly\n\nPrevent 1.1 (secret consumer) tests being run.\n\nAlso fixed FIPS jobs by adding nslookup_target and moving to centos 9.\n\nDepends-On: https://review.opendev.org/c/openstack/barbican-tempest-plugin/+/854614\nChange-Id: I6bdac945a6aef822d395dda7f460501ff20619a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/853f7568e670c8b3d0572e84d56a6b92e7f0b75b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/853f7568e670c8b3d0572e84d56a6b92e7f0b75b"}]},"branch":"refs/heads/stable/xena"},"6328e38ad9f569b7609cfe6ee8fb54a1698af632":{"kind":"REWORK","_number":4,"created":"2022-08-26 02:47:24.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/39/854539/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/39/854539/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/39/854539/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/39/854539/4"}}},"commit":{"parents":[{"commit":"bb2779473f786f3bb7b84bc4f445089da8cbfe88","subject":"Allow users with \"creator\" role to edit ACLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bb2779473f786f3bb7b84bc4f445089da8cbfe88"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-08-25 10:07:39.000000000","tz":120},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-26 02:44:36.000000000","tz":-240},"subject":"Set versioned jobs to set microversion correctly","message":"Set versioned jobs to set microversion correctly\n\nPrevent 1.1 (secret consumer) tests being run.\n\nAlso fixed FIPS jobs by adding nslookup_target and moving to centos 9.\n\nDepends-On: https://review.opendev.org/c/openstack/barbican-tempest-plugin/+/854614\nChange-Id: I6bdac945a6aef822d395dda7f460501ff20619a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6328e38ad9f569b7609cfe6ee8fb54a1698af632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6328e38ad9f569b7609cfe6ee8fb54a1698af632"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[]}
