)]}'
{"id":"openstack%2Fbarbican~854677","triplet_id":"openstack%2Fbarbican~master~I4e91ca55e3d4030dfdc55d78b92ec3ad3f94bd64","project":"openstack/barbican","branch":"master","topic":"story/2010235","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-08-25 22:27:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e91ca55e3d4030dfdc55d78b92ec3ad3f94bd64","subject":"Fix Secure RBAC policies for Containers API","status":"MERGED","created":"2022-08-25 20:46:08.000000000","updated":"2022-08-25 22:29:30.000000000","submitted":"2022-08-25 22:27:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854677-story/2010235","meta_rev_id":"9a4995f2296a9dd004449473eac82c0f134e1679","_number":854677,"virtual_id_number":854677,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-25 22:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":2,"date":"2022-08-25 21:05:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":1,"date":"2022-08-25 21:57:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-25 21:05:27.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2022-08-25 21:45:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ef7f2c82e78022f37d44540296ab958d4e4f5b6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-25 20:46:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"416a37eb99506692c726b60b354b6cce17240e15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-25 20:54:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7fa8c37ab358b60d3ae069b12e8d47650dd741db","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-08-25 21:05:27.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9b1bd774fcfc6c05e142a545e5d0e458eea9023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 21:45:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d51c43c7c7b476aa4cfaabf9098ca10\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acb0d54a9b634e1fa5aa717f9837de9c : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5330ee74e4844de09c5a0fee2a5ffe7a : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f23725e774b4260939e21c92ba35096 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/294988d724ef454b8c3da0897bcdf191 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71b05dc18d4c4cb5a09814b8deb79930 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c306a35add484386598945e0aefa58 : SUCCESS in 7m 28s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/11c4e00e23414de98234fa0ed3d2d3ba : SUCCESS in 31m 59s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/ab62f224415944c8b2c104d5d2640314 : SUCCESS in 33m 39s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/0a32f42b2c92465daf2c66c14d5907b2 : SUCCESS in 49m 35s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/40cd5a7026e94577b3cb2201406ccd81 : SUCCESS in 37m 30s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/ce5b683238d44956b648ef7dfca0cef4 : SUCCESS in 36m 50s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/8b81df7896fd4d518999ac03113ab71c : SUCCESS in 32m 57s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/19fdc055db20483e8aa8a3763879dab8 : SUCCESS in 41m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/deadd8d2136841da86d269fc6b552f7a : SUCCESS in 46m 20s","accounts_in_message":[],"_revision_number":2},{"id":"96286bf2678780bfce2acb30b9e39ba2d3b89740","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-08-25 21:57:28.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4efe0ba570dd6995fe60d4deef98b208cb703c03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 21:58:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4928428a59933ba800685b653dbd8906ea6d9041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 22:27:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1ca55fc53da4530948a8f1932f017e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeeda041d8bb4d4d8c0e00eb57277b07 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f71337027ecf4dfeba376a11cf1bdd1f : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e28f13fd2fc44016ab278b386bf6cd80 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd86f65556214deeb2247898f17de119 : SUCCESS in 6m 28s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/7ead71aad1304547a7030464bfad3f65 : SUCCESS in 28m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b2c32cb6d493fe4791e1d63583d1c1a89819115a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 22:27:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9a4995f2296a9dd004449473eac82c0f134e1679","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 22:29:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a355ae2b54b4851b7123046cb08cb84\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99d681922fd4cdea2ac6ca5a1c6bbd2 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e53363f52dcbc32c9b0494379623bbdc55fbba9","revisions":{"795417a7e37cb7cf51228fc480b3616b6000c913":{"kind":"REWORK","_number":1,"created":"2022-08-25 20:46:08.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/77/854677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/77/854677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/77/854677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/77/854677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/77/854677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/77/854677/1"}}},"commit":{"parents":[{"commit":"b6002aa3d04c0dc3bf49c05686b76f1780de358f","subject":"Fix Secure RBAC policies for Consumers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b6002aa3d04c0dc3bf49c05686b76f1780de358f"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-25 20:45:23.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-25 20:45:23.000000000","tz":-300},"subject":"Fix Secure RBAC policies for Containers API","message":"Fix Secure RBAC policies for Containers API\n\nThis patch fixes the Consumers API policies to ensure\nthat they are only evaluated when enforce_new_defaults \u003d True\n\nChange-Id: I4e91ca55e3d4030dfdc55d78b92ec3ad3f94bd64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/795417a7e37cb7cf51228fc480b3616b6000c913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/795417a7e37cb7cf51228fc480b3616b6000c913"}]},"branch":"refs/heads/master"},"6e53363f52dcbc32c9b0494379623bbdc55fbba9":{"kind":"REWORK","_number":2,"created":"2022-08-25 20:54:42.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/77/854677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/77/854677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/77/854677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/77/854677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/77/854677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/77/854677/2"}}},"commit":{"parents":[{"commit":"b6002aa3d04c0dc3bf49c05686b76f1780de358f","subject":"Fix Secure RBAC policies for Consumers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b6002aa3d04c0dc3bf49c05686b76f1780de358f"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-25 20:45:23.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-25 20:54:33.000000000","tz":-300},"subject":"Fix Secure RBAC policies for Containers API","message":"Fix Secure RBAC policies for Containers API\n\nThis patch fixes the Consumers API policies to ensure\nthat they are only evaluated when enforce_new_defaults \u003d True\n\nChange-Id: I4e91ca55e3d4030dfdc55d78b92ec3ad3f94bd64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6e53363f52dcbc32c9b0494379623bbdc55fbba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6e53363f52dcbc32c9b0494379623bbdc55fbba9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[]}
