)]}'
{"id":"openstack%2Fbarbican~854713","triplet_id":"openstack%2Fbarbican~stable%2Fxena~I191f41372a5c0b334ff858743a9303325db40cb6","project":"openstack/barbican","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-08-26 08:48:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I191f41372a5c0b334ff858743a9303325db40cb6","subject":"Fix Secure RBAC policies for Consumers","status":"MERGED","created":"2022-08-26 03:07:08.000000000","updated":"2022-08-26 08:50:05.000000000","submitted":"2022-08-26 08:48:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854713","meta_rev_id":"dfffe22705e1c9fd7b46e360469fde1025bd2dfe","_number":854713,"virtual_id_number":854713,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 08:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":2,"date":"2022-08-26 08:04:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":1,"date":"2022-08-26 08:04:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 04:27:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 08:04:27.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"}],"messages":[{"id":"85fd146a1cf43638fc5272286831e733a226f6a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-26 03:07:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f49301990609d1d53d826c5570f3f07cd8b0fb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 04:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b40f99ae6fd4ab8ac6a376762cea1d4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9527f107b804970ad0b5fa3aa5a5faf : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc8a3e8aebc74381982ccde14d37e1e1 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9934ffc5f8844e3a2e6cbe1b34aad90 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f8b3531627442938e90e2cea6ac2dcd : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d320391b65af48799793bfcdfaee61ad : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a91e4ed0d148618e8bc6a5a66b8e65 : SUCCESS in 7m 47s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/2801744debf74ff38b961231a97bc577 : SUCCESS in 30m 54s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/198af30f101a46bb8f5f60033e4944c8 : POST_FAILURE in 11m 14s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/85a097390d744bc49cfaee5be9551b85 : SUCCESS in 37m 24s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/d289c31a61ef45d39f3bde91b6de217d : SUCCESS in 54m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/d02ef92b432247128251f28c2d639362 : SUCCESS in 33m 15s\n- barbican-tempest-plugin-simple-crypto-secure-rbac-xena https://zuul.opendev.org/t/openstack/build/b93f62bb5ced4507841870264939c0be : SUCCESS in 45m 52s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-xena https://zuul.opendev.org/t/openstack/build/697ea6a611fe45a2ad905f16ef05c4b4 : SUCCESS in 34m 32s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/e6759f6f5deb4d1da806e6fc5960a677 : SUCCESS in 58m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b53c8376415b48babeff94a866ef1ac7 : SUCCESS in 50m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5709969333d4b024bce139d69ee83be4e7c955b4","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-08-26 08:04:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fcdef9acbc749efc79fae278342e78f7e535854e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:05:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06bf496c2f2f02cfce1c07cb6171d9a6d9bb135a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:48:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b972e3c8724349c7831bcc9998a5b1ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7e4a8ced6d42a9bce8dc70ea8fb2b4 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9ce79667cf64bc188310c1241807cfd : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90d52fdd7bd54a9ca4439428d918deed : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf2f3023b88d44938ac58fdbbc8ad7b8 : SUCCESS in 8m 10s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/c6b52e761d8d425dac8bc8a3c7e12dda : SUCCESS in 34m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1046e812bb77cb5b1f427c1a2751b6bdafbd9893","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:48:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dfffe22705e1c9fd7b46e360469fde1025bd2dfe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:50:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0838d341b7f41e5878870117691a179\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8448abf2b4544caea76878726e7d7694 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b1c6b3d8005a9d64383a9928ada010d5a9db455","revisions":{"3b1c6b3d8005a9d64383a9928ada010d5a9db455":{"kind":"REWORK","_number":1,"created":"2022-08-26 03:07:08.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/13/854713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/13/854713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/854713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/854713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/854713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/13/854713/1"}}},"commit":{"parents":[{"commit":"8c44a2f955c25f364b7e98599840ebc949e60db4","subject":"Fix Secure RBAC policies for secret_metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/8c44a2f955c25f364b7e98599840ebc949e60db4"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-25 16:28:27.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-26 03:06:31.000000000","tz":-240},"subject":"Fix Secure RBAC policies for Consumers","message":"Fix Secure RBAC policies for Consumers\n\nThis patch fixes the Consumers API policies to ensure\nthat they are only evaluated when enforce_new_defaults \u003d True\n\nStory: 2010235\nChange-Id: I191f41372a5c0b334ff858743a9303325db40cb6\n(cherry picked from commit b6002aa3d04c0dc3bf49c05686b76f1780de358f)\n(cherry picked from commit 8b5c1d135f380e7db9875693633b6e6e0459bd8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/3b1c6b3d8005a9d64383a9928ada010d5a9db455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/3b1c6b3d8005a9d64383a9928ada010d5a9db455"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[]}
