)]}'
{"id":"openstack%2Fbarbican~854849","triplet_id":"openstack%2Fbarbican~stable%2Fxena~Ief495c50bb120e2aa671dbcc80734ccb5a839b74","project":"openstack/barbican","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-08-29 15:28:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief495c50bb120e2aa671dbcc80734ccb5a839b74","subject":"Fix remaining Secure RBAC policies","status":"MERGED","created":"2022-08-26 22:25:19.000000000","updated":"2022-08-29 15:29:50.000000000","submitted":"2022-08-29 15:28:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854849","meta_rev_id":"36a523faf76cc1e576ea7f99ca2cfe2ad21db214","_number":854849,"virtual_id_number":854849,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-29 15:28:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":2,"date":"2022-08-29 14:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":1,"date":"2022-08-29 14:41:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 23:43:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 14:41:24.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"}],"messages":[{"id":"563e47c882ca58c4f2c17d723a9dedc2d1f3c8ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-26 22:25:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38c1de978ce1bb80d7f0c0e6eb2d51cb6ee414d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-26 22:27:36.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ae1ccc810018eca359ed494f6319a9ff2c159732","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-26 22:29:00.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"edd0d5fea2d1ebf98c81583f43a6bd39a093327a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 23:43:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e925b3173b2d45129fdf742256c2d10c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a2aac319001416190d70d8d0d0a84a5 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ba2027d65848379241bdcdf3415f4f : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c81923031ba5469e9e700632a4dd57d2 : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb519b3e9de9411cb84de86f2204d185 : SUCCESS in 8m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c15af1df9a249f18b3a910a93290e9a : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27bb0cdd41ec4f2f85e1f15623b586cd : SUCCESS in 10m 07s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/0ebfd53a20724400b43aefa55d5fc21b : FAILURE in 44m 12s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/16b9bb5e34834d5aa80768892a25d261 : POST_FAILURE in 18m 12s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/718dad674a3840e1906d461f2318383b : FAILURE in 49m 01s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/f2f987a7fe4842268df72e95ed4519c0 : SUCCESS in 1h 10m 30s (non-voting)\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/dd1ef60fde0841b2abf35f14f21a03ad : SUCCESS in 49m 51s\n- barbican-tempest-plugin-simple-crypto-secure-rbac-xena https://zuul.opendev.org/t/openstack/build/7f8572e72f6347efa9502cdcbaf7b18d : FAILURE in 50m 13s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-xena https://zuul.opendev.org/t/openstack/build/453807837bf146b28a0bd63c69b0c234 : SUCCESS in 47m 42s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/843e9b735ebc4e36bb7ae5a7940fb792 : SUCCESS in 59m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39bfbd0efdad43b19f928e3149f236bd : SUCCESS in 58m 41s","accounts_in_message":[],"_revision_number":3},{"id":"3cf6543d36335229e5d2d4a8b231b8007d3a9a4b","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-27 00:21:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b905c6bba78700a0738f5403ecaae1d734ec420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-27 01:11:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dba2ec737f2346e891f0571530985a5b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03ebe4926a0641bc819bcf6c1b02b228 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d076b62fc8d64c169e3fc5db5a00ad1c : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29127043afbf4d93a881c9bfaad30cc6 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aacdcad1e3a84f98abeb7493a92f11ef : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abee63eb68084e75b4da774a842b5e13 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31bcf769faed4a8aa4c26c535e790c53 : SUCCESS in 6m 08s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/9215e8cec2554bf99aa9009e71318967 : SUCCESS in 25m 32s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/75e6b87e74f140c9a6fb4c0d3c762766 : POST_FAILURE in 9m 18s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/cac29387314a44338591046440445a4f : SUCCESS in 27m 00s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/f03c7a21bcf341559f5de6bc25d8ddc5 : SUCCESS in 48m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto-xena https://zuul.opendev.org/t/openstack/build/452b8d5b4be14e8daf5db5344f041065 : SUCCESS in 29m 13s\n- barbican-tempest-plugin-simple-crypto-secure-rbac-xena https://zuul.opendev.org/t/openstack/build/8dfc242cf4784063a13961380b2833c3 : SUCCESS in 30m 08s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-xena https://zuul.opendev.org/t/openstack/build/2082f87deb6643fc8de2ebeb60e6412c : SUCCESS in 28m 46s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/b0fa03ad8ce944e4a177be598714f447 : SUCCESS in 35m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/78d56d10d9fc4985868e5cec587231ec : SUCCESS in 48m 48s","accounts_in_message":[],"_revision_number":3},{"id":"3d22d59a56e76c814f78aff04c4cd54a6bb7daca","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-08-29 14:41:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ac719fed935426917133cf87123921ca396be3e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 14:42:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e49da1f8f37596cdde7c80ac8ab91ae2d5ee92a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:28:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ebae4e8ad434e2e9e63219aa5096dc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7c65dc09ff49ff915297baee8dabe7 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5d74fa40ca7416fa730ead111c0559e : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8ff3a2625824ba69cef413e9c7107c6 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0fadc3355864c848b1eca4e3458258f : SUCCESS in 8m 13s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/788e7eeb13df478abb958c211b5eede7 : SUCCESS in 35m 56s","accounts_in_message":[],"_revision_number":3},{"id":"b5cad3637285905384c94a84ba29f89c396128a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:28:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"36a523faf76cc1e576ea7f99ca2cfe2ad21db214","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:29:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8873d45b32144cd5a1c00436a7548d26\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc4bec6eadd6407a803def7fcbde8be7 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"74bab1d4254ed49d29f57eec2a66368ce966e287","revisions":{"553409e337769bfa47762ccef7089d430a260940":{"kind":"REWORK","_number":1,"created":"2022-08-26 22:25:19.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/49/854849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/49/854849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/854849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/854849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/854849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/49/854849/1"}}},"commit":{"parents":[{"commit":"1ebdd8f53d36eda6afb88c5fe8143cdc68fc9f40","subject":"Fix Secure RBAC policies for Containers API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1ebdd8f53d36eda6afb88c5fe8143cdc68fc9f40"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-26 14:23:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-26 22:25:13.000000000","tz":-240},"subject":"Fix remaining Secure RBAC policies","message":"Fix remaining Secure RBAC policies\n\nThis patch fixes the remaining default policies to ensure they are only\nevaluated when enforce_new_defaults\u003dTrue.\n\nThe \u0027reader\u0027 role is removed from the policy tests in thise codebase\nbecause they now correctly fail.\n\nThe \u0027reader\u0027 role in was introduced into the policy tests in\n060ca2ee369af1fff3bf833d466f33d1889fb72e and the dynamic tests that were\ngenerated passed because of the issue being resolved in this patch.\n\nNow that the new default policies are not being evaluated when\nenforce_new_defaults\u003dFalse the requests using the \u0027reader\u0027 role are\ncorrectly denied.\n\nStory: 2010235\nTask: 46036\nChange-Id: Ief495c50bb120e2aa671dbcc80734ccb5a839b74\n(cherry picked from commit 950420f3c91053564bd3bdb2ec2f021ee0f53446)\n(cherry picked from commit 9cb5de7bac16e7867cbcf38469e4bf42fca47eb2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/553409e337769bfa47762ccef7089d430a260940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/553409e337769bfa47762ccef7089d430a260940"}]},"branch":"refs/heads/stable/xena"},"bd17cd16f349854931f0cf229cef7cb56524017b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-08-26 22:27:36.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/49/854849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/49/854849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/854849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/854849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/854849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/49/854849/2"}}},"commit":{"parents":[{"commit":"1ebdd8f53d36eda6afb88c5fe8143cdc68fc9f40","subject":"Fix Secure RBAC policies for Containers API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1ebdd8f53d36eda6afb88c5fe8143cdc68fc9f40"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-26 14:23:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-26 22:27:22.000000000","tz":-240},"subject":"Fix remaining Secure RBAC policies","message":"Fix remaining Secure RBAC policies\n\nThis patch fixes the remaining default policies to ensure they are only\nevaluated when enforce_new_defaults\u003dTrue.\n\nThe \u0027reader\u0027 role is removed from the policy tests in thise codebase\nbecause they now correctly fail.\n\nThe \u0027reader\u0027 role in was introduced into the policy tests in\n060ca2ee369af1fff3bf833d466f33d1889fb72e and the dynamic tests that were\ngenerated passed because of the issue being resolved in this patch.\n\nNow that the new default policies are not being evaluated when\nenforce_new_defaults\u003dFalse the requests using the \u0027reader\u0027 role are\ncorrectly denied.\n\nStory: 2010235\nTask: 46121\nChange-Id: Ief495c50bb120e2aa671dbcc80734ccb5a839b74\n(cherry picked from commit 950420f3c91053564bd3bdb2ec2f021ee0f53446)\n(cherry picked from commit 9cb5de7bac16e7867cbcf38469e4bf42fca47eb2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bd17cd16f349854931f0cf229cef7cb56524017b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bd17cd16f349854931f0cf229cef7cb56524017b"}]},"branch":"refs/heads/stable/xena"},"74bab1d4254ed49d29f57eec2a66368ce966e287":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-08-26 22:29:00.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/49/854849/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/49/854849/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/854849/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/854849/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/854849/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/49/854849/3"}}},"commit":{"parents":[{"commit":"1ebdd8f53d36eda6afb88c5fe8143cdc68fc9f40","subject":"Fix Secure RBAC policies for Containers API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1ebdd8f53d36eda6afb88c5fe8143cdc68fc9f40"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-26 14:23:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-08-26 22:29:00.000000000","tz":0},"subject":"Fix remaining Secure RBAC policies","message":"Fix remaining Secure RBAC policies\n\nThis patch fixes the remaining default policies to ensure they are only\nevaluated when enforce_new_defaults\u003dTrue.\n\nThe \u0027reader\u0027 role is removed from the policy tests in thise codebase\nbecause they now correctly fail.\n\nThe \u0027reader\u0027 role in was introduced into the policy tests in\n060ca2ee369af1fff3bf833d466f33d1889fb72e and the dynamic tests that were\ngenerated passed because of the issue being resolved in this patch.\n\nNow that the new default policies are not being evaluated when\nenforce_new_defaults\u003dFalse the requests using the \u0027reader\u0027 role are\ncorrectly denied.\n\nStory: 2010235\nTask: 46121\nChange-Id: Ief495c50bb120e2aa671dbcc80734ccb5a839b74\n(cherry picked from commit 950420f3c91053564bd3bdb2ec2f021ee0f53446)\n(cherry picked from commit 69307b024b98c990baa2ca3bf2044b773c1869e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/74bab1d4254ed49d29f57eec2a66368ce966e287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/74bab1d4254ed49d29f57eec2a66368ce966e287"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[]}
