)]}'
{"id":"openstack%2Fbarbican~869801","triplet_id":"openstack%2Fbarbican~master~I7d0cd4d00244a3dc791b35f112e3689c2b6dd790","project":"openstack/barbican","branch":"master","topic":"secret-consumers","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-01-12 14:20:10.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2023-01-12 13:36:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9914\u003e replied on the change","reason_account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}},"hashtags":[],"change_id":"I7d0cd4d00244a3dc791b35f112e3689c2b6dd790","subject":"Add content_types to the response dict","status":"MERGED","created":"2023-01-11 09:44:18.000000000","updated":"2023-01-12 14:21:28.000000000","submitted":"2023-01-12 14:20:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869801-secret-consumers","meta_rev_id":"10b22da140116173d8114f84e048bf4465a2d13f","_number":869801,"virtual_id_number":869801,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 14:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2023-01-12 13:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2023-01-12 13:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2023-01-12 13:40:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-01-12 13:38:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-11 10:47:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-12 13:10:21.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2023-01-12 13:40:40.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"574b63896dad923b8d4e95444572fee1390630da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-01-11 09:44:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ec083417852d32acf11c83249e4af6c9cce321b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 10:47:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/730955d37d614a9a871f24b753c76ab8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aa00d8add414491a37c148aa08e0064 : FAILURE in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61fd1f750c454e83a53edd8f58aa9baf : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ddf08b9b38b45d182332fcb98768aad : FAILURE in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a52a77efa08445ceb6c175cbfc7a4118 : FAILURE in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb9ef891e1084b9184d699c0c6d1e840 : SUCCESS in 7m 10s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/f077baa0a15c47d3a2f9158db00b0149 : SUCCESS in 44m 54s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/a62c3740f66941f6ba6712fd3dd87813 : SUCCESS in 39m 20s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/3751f4d250754f528ecc4c52114376f3 : FAILURE in 49m 55s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/079b90c567d544b58dc333dc0dd2b9df : SUCCESS in 47m 18s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/48a87a10b9304bd5a2e94848c3fe0ae3 : SUCCESS in 52m 28s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/24797f70102647f696c98dc405ef9df7 : SUCCESS in 42m 13s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/500fd679f0754d379dc4079c194b221f : SUCCESS in 49m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b8daf49d7f045ec9750e300381161e7 : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"46d649fcfc9d39138c2781e403ab12d9fda0f1ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-01-12 09:16:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b3429c23a5b180dd20403910442fbf710260b3a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:16:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18ac32f918ee427681b2520b674c08fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afee6f5e12f541a4bc80f5581ab42eaa : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c70a4ea89e44c8a87f9cbef91e15a9 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19a68cfdb988476496ea1dbc389bcaf1 : SUCCESS in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d17b7e749c8043d3b21f54735c6be56a : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed6d50d9b764033970eecd30044a0e3 : SUCCESS in 7m 00s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/5f991ab597ed48ebbc6a723602630d58 : SUCCESS in 41m 37s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/43b6521524cf49d5867e081e49c2ec7c : SUCCESS in 39m 57s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/4b9cd7b57e234d13afc08ed3023ebc1c : FAILURE in 52m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/967d0664fbaf4d1b856496ab0bc865a9 : SUCCESS in 47m 56s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/5b14fd84fffc421e82dec767750d4830 : SUCCESS in 46m 12s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/c437be3bf6bf4d6080860189e1c3c3d3 : SUCCESS in 42m 59s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/98b77c3dea4748a29779ae29e601e72c : SUCCESS in 52m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/25bf48ace1dd46058a56abf746efece1 : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d68ad85ade206a97937bb674fa4c670a9e23b478","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-01-12 11:05:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2df02d23b85c12ac0c5738064cc2c936053bdc61","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-01-12 12:10:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb8f003698b4dc994d2c43fa4613408b42396b0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:20:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5aea5f48ce564ab597b8bda217206461\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd89433c08da46e996eb56a8212ac056 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df196eeb9ff44d16bbefb73a9d9b2c3c : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3433ce61b5fd4f8d959e916ed41cc12e : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/793d4cb14bb94432886f6a3865114ebd : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6498e4ebb8d340198bf58924dcf67733 : SUCCESS in 6m 52s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/c1dcbb9def084ef7b83f4104de6b2f3f : SUCCESS in 40m 31s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/8ed280d76f7d4f3486794b5a376a207a : FAILURE in 24m 04s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/18d31c6b1fa145cfaf3a261c2191884b : FAILURE in 52m 46s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5b1e37c66434439ea9e05b79d88172b5 : FAILURE in 23m 45s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/56d9e353e8544d9f8c69ddbecf387112 : SUCCESS in 57m 10s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/5792a6cb33314fc686f8dcda08ea8902 : FAILURE in 22m 15s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/f7b32f66ae894bb883b4f9487b932194 : FAILURE in 1h 08m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f47f0acfe9214ed08e5b96ca8e5950ad : SUCCESS in 33m 00s","accounts_in_message":[],"_revision_number":3},{"id":"17d69bcdef988b43de11e44df4321489e52e04fc","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-01-12 12:20:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6d1a0090d4aca3080d65f622be39150094ac2d82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:18:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94e214ccf187442489be83a8afcb0cc6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdf4a0cd6d4b45bda634eedebcad1a2d : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b95c876a6b4a9da1f50a2a6b709989 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0eff20d931f34567911fc0ec5b2cf585 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b7656ec783d4d23b859de9c592409f3 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163f6d8e6a79427ca73d80a347e3a2ce : SUCCESS in 6m 19s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/8cb2d66b25c14ea289d1df22ed5fa15b : SUCCESS in 26m 34s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/92711696645946e4af39846f002d0429 : SUCCESS in 27m 20s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/62e469127bda4c56a2554462bdacf3a6 : FAILURE in 52m 07s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b2fd219be88b4b709900cd9320cec132 : SUCCESS in 31m 14s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/04925bb27be9472d9173313dbcbe82ef : SUCCESS in 47m 56s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/03133ef2b1e8430a9e941d2f21fa77b3 : SUCCESS in 42m 44s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/61cc8f6615d947098f8ab940b6b2c278 : SUCCESS in 48m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/331740f76a2e4987abceea80f6f68a74 : SUCCESS in 38m 11s","accounts_in_message":[],"_revision_number":3},{"id":"737efcbc0c14f9a7a3dc9473f30f60101a84b666","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-01-12 13:36:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"af9e408d5b659fc749d675d2ca23f06514ac4ecd","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-01-12 13:38:36.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"201592e0fb2570ddb9d483528d772e8fed9861aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:39:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"427d5c6752bf6b04127f9088f893cee33c370db0","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-01-12 13:40:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6bf73c70edd942f249d19e9b1594c784e6a47754","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:20:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c12c7b44bdd94612b0ecaf392c7a9c19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/981d9572156f43b394029ea6df3f8c40 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe7ae29bd269419b805eebdd0965ca4b : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92c401e6059a4b789119de35907e901e : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa8571c8df9740bb8399b58baf7d3935 : SUCCESS in 5m 55s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/f4b39ed1db8f4d249734d63997db6a8c : SUCCESS in 36m 21s","accounts_in_message":[],"_revision_number":3},{"id":"c18a13226c71bd707ea2c1fec4aa9f98e4f22883","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:20:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"10b22da140116173d8114f84e048bf4465a2d13f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:21:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7a17147fc0544d1b369ca0cfe62d791\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f65255b33c334afca731958f120735ab : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f848fa9b4a28c92a61dcc6fe07795dee1f3b9c35","revisions":{"f71eac02c6482192cf5f9f52269e6e07d850e1f6":{"kind":"REWORK","_number":1,"created":"2023-01-11 09:44:18.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/01/869801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/01/869801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/01/869801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/01/869801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/01/869801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/01/869801/1"}}},"commit":{"parents":[{"commit":"24fe0b1f33b53f82725a34ee7ec1451fa7961810","subject":"Fix typo and docs gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/24fe0b1f33b53f82725a34ee7ec1451fa7961810"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-01-11 09:42:21.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-01-11 09:42:21.000000000","tz":60},"subject":"Add content_types to the response dict","message":"Add content_types to the response dict\n\nThis change adds content_types to the consumers responses\nreturning a secret.\n\nChange-Id: I7d0cd4d00244a3dc791b35f112e3689c2b6dd790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f71eac02c6482192cf5f9f52269e6e07d850e1f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f71eac02c6482192cf5f9f52269e6e07d850e1f6"}]},"branch":"refs/heads/master"},"be0f26e573c1ce3890d9f634619d65d97cabd13a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-12 09:16:23.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/01/869801/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/01/869801/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/01/869801/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/01/869801/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/01/869801/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/01/869801/2"}}},"commit":{"parents":[{"commit":"659ea4c5c7f70675a302e34db0afd36fcac37a98","subject":"Merge \"Secret consumers documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/659ea4c5c7f70675a302e34db0afd36fcac37a98"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-01-11 09:42:21.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-01-12 09:16:23.000000000","tz":0},"subject":"Add content_types to the response dict","message":"Add content_types to the response dict\n\nThis change adds content_types to the consumers responses\nreturning a secret.\n\nChange-Id: I7d0cd4d00244a3dc791b35f112e3689c2b6dd790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/be0f26e573c1ce3890d9f634619d65d97cabd13a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/be0f26e573c1ce3890d9f634619d65d97cabd13a"}]},"branch":"refs/heads/master","description":"Rebase"},"f848fa9b4a28c92a61dcc6fe07795dee1f3b9c35":{"kind":"REWORK","_number":3,"created":"2023-01-12 11:05:01.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/01/869801/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/01/869801/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/01/869801/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/01/869801/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/01/869801/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/01/869801/3"}}},"commit":{"parents":[{"commit":"659ea4c5c7f70675a302e34db0afd36fcac37a98","subject":"Merge \"Secret consumers documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/659ea4c5c7f70675a302e34db0afd36fcac37a98"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-01-11 09:42:21.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-01-12 11:04:49.000000000","tz":60},"subject":"Add content_types to the response dict","message":"Add content_types to the response dict\n\nThis change adds content_types to the consumers responses\nreturning a secret.\n\nChange-Id: I7d0cd4d00244a3dc791b35f112e3689c2b6dd790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f848fa9b4a28c92a61dcc6fe07795dee1f3b9c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f848fa9b4a28c92a61dcc6fe07795dee1f3b9c35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[]}
