)]}'
{"id":"openstack%2Fbarbican~880813","triplet_id":"openstack%2Fbarbican~master~I317072eab37ffe284c820a4c41cb6df82f1a05c5","project":"openstack/barbican","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-07-21 14:54:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I317072eab37ffe284c820a4c41cb6df82f1a05c5","subject":"tox: Remove basepython","status":"MERGED","created":"2023-04-19 11:41:48.000000000","updated":"2023-07-21 14:55:11.000000000","submitted":"2023-07-21 14:54:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880813-sqlalchemy-20","meta_rev_id":"a0c8c66b1094bae61ecaa15f875ac229cc3913da","_number":880813,"virtual_id_number":880813,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-21 14:54:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2023-07-21 14:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2023-07-18 12:06:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2023-07-21 14:08:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 12:58:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-18 12:06:11.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2023-07-21 14:08:34.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"2de5d6e34ce7befcd35e2ad98ba94d3616e506a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-19 11:41:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2084fb8cc6bc885c731902fb862594538219e2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 12:58:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af0384234f4348e582fd118da71eede9\n\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/afb8a61ade6e4ce9a9bec6a8a3952aae : SUCCESS in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e51001eb7d0445bfaf8d381c24fe0e16 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b88f201164646d3a320baccef2ad14f : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/046a89b09c724d62a4b309b61b2593d9 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abdbc435f77548ba918d5b78d69c3442 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40afc66491b04858b4bc63196b19811e : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f879c2dd76a4b0b97c400fd4ccade0a : SUCCESS in 4m 26s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/cd1f6eed703344cabca4092ddc50fbbe : FAILURE in 31m 19s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/9ed2002c343547b4a4cf493612440e4c : FAILURE in 37m 09s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/ea4ff2e8ecec4337a6165d4462eac86e : FAILURE in 35m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/90d23ad043e443838c693ac30b6c2b80 : SUCCESS in 39m 26s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/0c127f15e1ed462bb2a68365f25bc453 : SUCCESS in 47m 33s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/9a6ba46efdae4613b73f8e654fabb90d : SUCCESS in 29m 23s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/41103163a32b4c798a463fc663025806 : FAILURE in 37m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b2e8bf0baf2f41fd993c16a7d6245f7d : SUCCESS in 1h 11m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d08c2ad969f26928fdb5924d06e95946addc0a71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-27 10:29:15.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e865f78840d80ae375aacf6e70525975f0ea13bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 11:44:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aadbc541922044839e16b7efeb522341\n\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/a2bc583db85d48ca8de0a3b806dfd9d8 : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1f5500c92d44792849dbd850301c1b6 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b33b313b1f1e4a1e9ae99cc5b4d73630 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df9dc248809a4b108023defd65a3d6d8 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7508a2f041648d6b4944aa3e0c37d4c : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd2e2f21406b49549f0e4b37c7696792 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e90c70c8c334fdab8996532ff923cd2 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c442c0517d604849b0ee46e8398b2bbc : SUCCESS in 4m 47s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/f505a19bb78a4ec1a3e685a6723d1c78 : SUCCESS in 52m 36s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/2311e0bf8436414eac50b86d5c4862d6 : SUCCESS in 47m 13s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/81d8496e4405411c9d735e811a05e8a5 : FAILURE in 58m 33s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a9f2d946b9e54a9fa10244269c85827c : SUCCESS in 37m 12s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/068e13540ea94cb3b3f168922ac9aa35 : FAILURE in 36m 32s (non-voting)\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d885a1ba605849ddb74e0d48a8640989 : SUCCESS in 1h 02m 41s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/45c679de7a5544ada4b3e6bd9e379d57 : SUCCESS in 1h 00m 55s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a84c1694b36b4816b0f8bcd51e71c5eb : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3863e45835a6557e384d47fb38ff91686712ec0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-12 17:29:18.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7a45c55711d9e6bd0b1368af578266e2919edbd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 18:30:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd0fd8dde5e544d4a33b49203a0da3c7\n\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/12ea8bc873d0452ab030ddf13b0c8047 : SUCCESS in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/597c171ef82242b8bbf90c149277eaea : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d00bb173b2744c5aa1ba778b46fe400 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/894f23679ecb4e5ab4f43360ebd7dc13 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7bc3e32925444ad895731f41a74bef3 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e521e0c13884af4a3e129ae8a7b5e11 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d7aebde93a7415fae2f8c96ffa23fa0 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1aa8f3aaf754295a0050374d409e48f : SUCCESS in 4m 46s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/b8315435e816434cbff54c813dccce7a : SUCCESS in 31m 38s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/723641ccf76b409a9e7530b565b7f456 : SUCCESS in 45m 55s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/1a5d01ffa48c4352863f75fb015658c1 : FAILURE in 33m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1725d68cc7514ef58eeed7a1338b780b : SUCCESS in 27m 55s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/73467a8d5ae448c2b58d97f7197597f1 : SUCCESS in 38m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/86a63477fea54607a0c79fe3a9a7cb6b : SUCCESS in 33m 50s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/0c4f611ca3b04dd2a07fac5574956981 : SUCCESS in 40m 42s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/03091d0ffd9247b9b08e7a20c86f1793 : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":3},{"id":"ed3f416534c7e0f5fd3dfe8dfec80341dcd32f0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-13 09:04:54.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"48fbcac188646e866ec827477b58bedd694ec0f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 09:45:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e522a8e8220141e5bb5a874966476115\n\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/c6f17e4ea4ff466599a0f4e0e11a6862 : SUCCESS in 4m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bc199d7ff68454982b8f4afaa615fd2 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352c6eaf915f4e7ab9c075ac8421150c : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53f634ed15554464839015d102bef3c2 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef40b486506d4c9f8e1acc0f0ea8e1ba : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f06cab07e4424931a27b452e504b005b : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12c0868c903046ddbe3399492cc72713 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f873317e603c4e9e9d040d9080e4b7e4 : SUCCESS in 4m 46s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/be7e2c8aedba4c8390f97e70e1c8e914 : SUCCESS in 25m 41s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/b80f26f4be1c43de8401f8084e49c6b6 : SUCCESS in 25m 46s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/0843b98a863d4f9daefb76c50a77c1ca : FAILURE in 32m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/59df1f0277ba4af3bfb98d5c29ee7e6e : SUCCESS in 29m 48s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/1560328332a04779882528358d293025 : SUCCESS in 34m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/abbc6cf75be34dd09668f11c8873c69c : SUCCESS in 28m 10s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/5d9333f21af4466ea24891b3571e2fa7 : FAILURE in 35m 38s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/540b941ef4b84269abd1e64fb394af9b : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":4},{"id":"ca77ab5a89434bc5e3fdf7b00e462c099e641542","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-07-18 12:06:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"30cf53dae7a263b161db543b134229a85c12a669","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-07-21 14:08:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fa159e69bd77e93a973d7a6f9f316f74069207ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 14:09:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a468d4e0eacf450c20e1df925a5f8ece82c3d936","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 14:54:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3323bc25461a42e2a1a728d96ab46789\n\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/0b88485b671f4730a97bd965cf8f2e25 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/958b3fe531bc4af9ac70de37ab913017 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acd2f2f132ec42eaa7746aca0837af12 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9b8dd9b90f14aeaa0d3c2d3f1b1ab62 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f491c99364684ba78c53e272c5f766fe : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d931e741af4e48048a89ea905f95e31d : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1564c2b0baf4566a73cc945f58287b9 : SUCCESS in 5m 04s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/89d37f91882443b28217310362fd2a31 : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":4},{"id":"2b6742e8507b13fe0810a572bbdb99b7431550dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 14:54:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a0c8c66b1094bae61ecaa15f875ac229cc3913da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 14:55:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6d08380172748919eb46fb83a91a07d\n\n- promote-api-guide https://zuul.opendev.org/t/openstack/build/0264801cd3b04434bc7ed69ee529924a : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab72bbff3f741c694de98e9a9897fb4 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63f7867869ae49e8becebc9fd231c0ef : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"34979553e79fb6a6e887abe811c13762dd0916c0","revisions":{"59befcc47ac564f9286350f36c338320ed940a6f":{"kind":"REWORK","_number":1,"created":"2023-04-19 11:41:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/880813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/13/880813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/13/880813/1"}}},"commit":{"parents":[{"commit":"9ba66a67b99782c8cea986b51744c0d3d5d25f63","subject":"Merge \"Remove TripleO job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/9ba66a67b99782c8cea986b51744c0d3d5d25f63"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-19 10:54:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-19 11:21:45.000000000","tz":60},"subject":"tox: Remove basepython","message":"tox: Remove basepython\n\nPython 3 is EOL. No environment should be defaulting to it. Our CI\nenvironments certainly aren\u0027t.\n\nChange-Id: I317072eab37ffe284c820a4c41cb6df82f1a05c5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/59befcc47ac564f9286350f36c338320ed940a6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/59befcc47ac564f9286350f36c338320ed940a6f"}]},"branch":"refs/heads/master"},"1c6b16154aa21e003875dec039de3dac7fbe27cd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-27 10:29:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/880813/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/13/880813/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/13/880813/2"}}},"commit":{"parents":[{"commit":"5a1b80f38a894044235ba1209256352daed7bbfe","subject":"Merge \"Remove System scope from policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/5a1b80f38a894044235ba1209256352daed7bbfe"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-19 10:54:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-27 10:27:08.000000000","tz":60},"subject":"tox: Remove basepython","message":"tox: Remove basepython\n\nPython 3 is EOL. No environment should be defaulting to it. Our CI\nenvironments certainly aren\u0027t.\n\nChange-Id: I317072eab37ffe284c820a4c41cb6df82f1a05c5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1c6b16154aa21e003875dec039de3dac7fbe27cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/1c6b16154aa21e003875dec039de3dac7fbe27cd"}]},"branch":"refs/heads/master"},"963f464ebfe68fd19e8280cdff87e5528e6203ba":{"kind":"NO_CHANGE","_number":3,"created":"2023-07-12 17:29:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/880813/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/13/880813/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/13/880813/3"}}},"commit":{"parents":[{"commit":"5a1b80f38a894044235ba1209256352daed7bbfe","subject":"Merge \"Remove System scope from policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/5a1b80f38a894044235ba1209256352daed7bbfe"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-19 10:54:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-12 16:20:18.000000000","tz":60},"subject":"tox: Remove basepython","message":"tox: Remove basepython\n\nPython 3 is EOL. No environment should be defaulting to it. Our CI\nenvironments certainly aren\u0027t.\n\nChange-Id: I317072eab37ffe284c820a4c41cb6df82f1a05c5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/963f464ebfe68fd19e8280cdff87e5528e6203ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/963f464ebfe68fd19e8280cdff87e5528e6203ba"}]},"branch":"refs/heads/master"},"34979553e79fb6a6e887abe811c13762dd0916c0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-07-13 09:04:54.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/880813/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/13/880813/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/880813/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/13/880813/4"}}},"commit":{"parents":[{"commit":"d929cd84e40e2093a2647fd5d9515c4269fd47ee","subject":"Merge \"Add support for Vault Namespaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/d929cd84e40e2093a2647fd5d9515c4269fd47ee"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-19 10:54:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-13 08:55:49.000000000","tz":60},"subject":"tox: Remove basepython","message":"tox: Remove basepython\n\nPython 3 is EOL. No environment should be defaulting to it. Our CI\nenvironments certainly aren\u0027t.\n\nChange-Id: I317072eab37ffe284c820a4c41cb6df82f1a05c5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/34979553e79fb6a6e887abe811c13762dd0916c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/34979553e79fb6a6e887abe811c13762dd0916c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
