)]}'
{"id":"openstack%2Fbarbican~880849","triplet_id":"openstack%2Fbarbican~stable%2Fyoga~Icba7968657ac476861a4fe4cbffd05a728ca54b0","project":"openstack/barbican","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-02-01 15:30:20.000000000","reason":"Change was abandoned"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-02-01 15:30:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icba7968657ac476861a4fe4cbffd05a728ca54b0","subject":"Fix functional tests","status":"ABANDONED","created":"2023-04-20 14:47:31.000000000","updated":"2024-02-01 15:30:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":880919,"cherry_pick_of_patch_set":1,"meta_rev_id":"f7b2807ec393275af66b97c2f290d1527fd46bfc","_number":880849,"virtual_id_number":880849,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-25 15:06:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 15:52:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 15:30:20.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"ce887e13885ed3cd9ef47647a5fe29d0b68574fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-04-20 14:47:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e8c8720ffba53bbf151d263ca6ffcf54d9c8b7d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 15:52:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3047eeadcde4fcd82783319276ef6d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/857a5cb704ea42718060e4f3853bf5aa : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c21eef061334b56afb347f09b4d426e : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43d47806bf6f417ea3590fcd4a56ec08 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d667bcb0fed4692afd9eaca27a2b4c5 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e816e5a0004305bb111759b80558b0 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e32796be96ff48c8afb000f97c0b9d74 : SUCCESS in 3m 57s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/dc260f1b1e9d430b8700d86ec7299044 : FAILURE in 35m 45s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/018a49c9d846443fbaa3e2d8b5944038 : FAILURE in 37m 27s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/ffd9f25116914e759ceafcfa9ad9fafa : SUCCESS in 58m 13s (non-voting)\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/ecdce353a4974c4aa8a63ef9ddbd86d3 : SUCCESS in 32m 27s\n- barbican-tempest-plugin-simple-crypto-secure-rbac-yoga https://zuul.opendev.org/t/openstack/build/49382ef46dc54807b177f518bd8c4bf4 : SUCCESS in 33m 04s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-yoga https://zuul.opendev.org/t/openstack/build/429cacb429ee4145bdb0f29142fd8f3c : SUCCESS in 46m 08s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/81b327df006f4d5ba96647af2078c517 : SUCCESS in 35m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0cae0f6c8640447bb4d0781d4651e58c : SUCCESS in 34m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e48c52d5db421698697e13538de05959fc266fff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-01-25 13:55:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"36c8e3cf7e8169e90d5862947c8f4a808a804f2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 15:06:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54c710e6085a44feb8c1ea8a5c7d33a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c914107366a41cbb9e6a5052e3badf7 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/856d0706c86d40169d8234423895a572 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/397b2fd23c7d44debc41b3ce71418f2f : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92332764687444eb869551980395c8ed : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d8e1cce23c4236984ba9e9ece13134 : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8245ad20079f48cb8294eaacc04d6022 : SUCCESS in 4m 28s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/4bae5204e5134544989bec14dafce1fa : SUCCESS in 40m 14s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/72cec52b5a1941738194519627f61f0a : SUCCESS in 39m 11s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/91c766075be94279b2756027841ba7c3 : SUCCESS in 1h 01m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto-yoga https://zuul.opendev.org/t/openstack/build/d77437db71244d09a7979a4307e20122 : SUCCESS in 42m 36s\n- barbican-tempest-plugin-simple-crypto-secure-rbac-yoga https://zuul.opendev.org/t/openstack/build/fd399f044ae240de9367be3823a24f7c : FAILURE in 48m 58s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-yoga https://zuul.opendev.org/t/openstack/build/2e9a023cb1ac49ef822be6e1ca64b82f : SUCCESS in 32m 21s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/29ff281b5e934a50b1af8b8f74aebe83 : SUCCESS in 44m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/177d6d04f855443fa72c11c89b69276e : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f7b2807ec393275af66b97c2f290d1527fd46bfc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-01 15:30:20.000000000","message":"Abandoned\n\nstable/yoga branch of openstack/barbican is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/yoga if you want to further work on this patch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee99f9bfc2d99cbe109a584bea4727ed5e67dd4a","revisions":{"ea575b7a3048de02a09f1b91411e5552fe5ff034":{"kind":"REWORK","_number":1,"created":"2023-04-20 14:47:31.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/49/880849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/49/880849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/880849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/880849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/880849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/49/880849/1"}}},"commit":{"parents":[{"commit":"f5147debb9af3eb9b5a9ce61d8c5177470bdfc9a","subject":"Fix typo in check_str for system_admin policy rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f5147debb9af3eb9b5a9ce61d8c5177470bdfc9a"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-04-20 08:22:44.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-04-20 14:47:31.000000000","tz":0},"subject":"Fix functional tests","message":"Fix functional tests\n\nThis adds find to the allow list in tox.\n\nChange-Id: Icba7968657ac476861a4fe4cbffd05a728ca54b0\n(cherry picked from commit 64eac2407155b86b68978479504dba653b7d51c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ea575b7a3048de02a09f1b91411e5552fe5ff034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ea575b7a3048de02a09f1b91411e5552fe5ff034"}]},"branch":"refs/heads/stable/yoga"},"ee99f9bfc2d99cbe109a584bea4727ed5e67dd4a":{"kind":"REWORK","_number":2,"created":"2024-01-25 13:55:37.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/49/880849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/49/880849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/880849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/880849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/880849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/49/880849/2"}}},"commit":{"parents":[{"commit":"f5147debb9af3eb9b5a9ce61d8c5177470bdfc9a","subject":"Fix typo in check_str for system_admin policy rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f5147debb9af3eb9b5a9ce61d8c5177470bdfc9a"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-04-20 08:22:44.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-01-25 13:55:30.000000000","tz":60},"subject":"Fix functional tests","message":"Fix functional tests\n\nThis adds find to the allow list in tox.\n\nDepends-On: https://review.opendev.org/c/openstack/barbican-tempest-plugin/+/906476\nChange-Id: Icba7968657ac476861a4fe4cbffd05a728ca54b0\n(cherry picked from commit 64eac2407155b86b68978479504dba653b7d51c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ee99f9bfc2d99cbe109a584bea4727ed5e67dd4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ee99f9bfc2d99cbe109a584bea4727ed5e67dd4a"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
