)]}'
{"id":"openstack%2Fbarbican~890141","triplet_id":"openstack%2Fbarbican~stable%2Fvictoria~Id7811a15fb00041544dc47c2e63bd33c50a771b1","project":"openstack/barbican","branch":"stable/victoria","topic":"pin-barbican-tempest-plugin","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2024-03-26 18:20:04.000000000","reason":"Change was abandoned"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-03-26 18:20:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id7811a15fb00041544dc47c2e63bd33c50a771b1","subject":"[DNM] Testing newer version of tempest - victoria","status":"ABANDONED","created":"2023-07-31 14:44:33.000000000","updated":"2024-03-26 18:20:04.000000000","total_comment_count":12,"unresolved_comment_count":2,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"3495a6014365a008d8420ae788faba543d9af92b","_number":890141,"virtual_id_number":890141,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-08 11:47:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"CC":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"CC":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"reviewer_updates":[{"updated":"2023-07-31 16:08:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-02 12:50:17.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2023-08-08 11:47:08.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-06 19:51:22.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"CC"},{"updated":"2024-03-26 18:20:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"77acbc80fc71578175c103d13a486ae9694fc127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-07-31 14:44:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21228b6dfe4ab36d78faf2834de3bf3e73379f87","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-07-31 14:44:52.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"2b347ddc9d3d683fa094048831b392066db63c71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 16:08:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f98e9a20b77c4f5a91dcf017d0349868\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0730ed69ca748d781f8f038a0cd4372 : FAILURE in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a8c1920d1f748b98380aecb9243a06a : FAILURE in 2m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99355c9acaca4f18b8abf05f201debbe : FAILURE in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67ebece32651457da1f2628378917fe0 : FAILURE in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/998a2e54d4b94d018a951655fefe0eda : SUCCESS in 7m 16s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/8d9edd6aae304c23a98a5bada2a454ea : FAILURE in 44m 47s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/fab7f3baa24947908d55ed86f61c0612 : RETRY_LIMIT in 2m 41s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/dc4edfe0a06c463b951700388f9e9857 : FAILURE in 29m 21s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/8d12cda4becc4e42a207c93c2902c098 : SUCCESS in 1h 17m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/15c2a8451d65498fa196340fad4a36b1 : FAILURE in 43m 04s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-victoria https://zuul.opendev.org/t/openstack/build/27eb9afa92ed43958f69584e0f89d5e8 : FAILURE in 45m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/be8546e886784781bc82f0e24df20c63 : FAILURE in 31m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1baedfa4bb95ed684131fc848e91c9672aa7ac07","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-08-02 12:50:17.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11523b43113125f96dce7b8196395535cdb22afc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-08 11:46:50.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f2994a2ec1389fc22a8ddd9aed6d28371108c4f4","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-08 11:47:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f265042c34f917affc8fef882935ba41e1b1238","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-08 11:47:08.000000000","message":"Patch Set 2:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"aed114d86cdd2809c504df2eccb030a8de0c9ec5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-08 11:47:23.000000000","message":"Patch Set 2:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"59102f713c0f34e94892543f4216ca977f7a7534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 11:47:27.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"43871a67c0142e2ffacf853d922618f963d3bd84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 11:47:52.000000000","message":"Patch Set 2:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"0ff5b05422649c0f40f510791ebc71ddaf1fe6b1","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-08 11:48:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9909c955a697a5e5e867b31e57ccec384243349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 12:47:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39e0816d091743388ec16fce65f283f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac5b807f9be8431ab0bc63c6d1f33492 : FAILURE in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba296c933b164305ab58bde4fa2ec85d : FAILURE in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d68ae4809055406c92c07e10f50e1cc6 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4b6161088b543c293d0721375050ddf : FAILURE in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c17a769684442d4959949eeb21b6004 : SUCCESS in 5m 44s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/1ce437d67b844e7b8c3b070e3dc13370 : FAILURE in 45m 38s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/29fcd100ac704fec98988bb52956998b : RETRY_LIMIT in 3m 08s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/3fa4857165e5425fb9e666bffd894231 : FAILURE in 30m 27s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/c57ba0cb8319418ab9427638522e7be3 : SUCCESS in 46m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/1919f3dd93024fc387b35cda5ba45c88 : FAILURE in 38m 51s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-victoria https://zuul.opendev.org/t/openstack/build/f919d28d646140228f18b6c1b3b36e99 : SUCCESS in 36m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7df29b3940845db8f0d37c0f09c3431 : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c67bc853f6dc45536cfa40361256d8b407425952","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-11 13:01:50.000000000","message":"Topic set to pin-barbican-tempest-plugin","accounts_in_message":[],"_revision_number":2},{"id":"1afa3d5fbecbd538273dd9c935ed4da9b5702f59","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-08-11 14:31:00.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"995054483ac6cf448d95dd0edc18802b433dbc98","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-14 08:09:36.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3ccdfe3c9cd345ade49001432e1847af07a23680","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-08-15 08:46:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c1c7426e3efe413ac4e9b46628d9d60e73563100","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 09:40:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d44cff093914f47bc94116d516b2fb7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62f30e3ac3dc4666b13336dd202d3afe : FAILURE in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3a5dff987ea433ba00d038113197460 : FAILURE in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9186390d9440450eb6197211d8850a42 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/769f94a08c354abea514a4964d4cd626 : FAILURE in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/664275f235724090a2c361b6156aee79 : SUCCESS in 7m 12s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/6f4edc94a041443aab54a5df7b3ff35f : FAILURE in 39m 46s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/23c1c1a64f7f44e28e6f1491042d61b6 : RETRY_LIMIT in 3m 48s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/11b96fd5ef1445d5a7ee4798498484d7 : FAILURE in 31m 29s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/a3c1e51f569644868122f0b0546388cd : SUCCESS in 45m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/a45dd53cd5ed4b43a2376c9a2aff8f47 : FAILURE in 31m 16s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-victoria https://zuul.opendev.org/t/openstack/build/e274a0f4724c4297a70dc09a20a99b34 : SUCCESS in 28m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/edffa2c125dc421e9f0d771064c3b2d3 : SUCCESS in 45m 48s","accounts_in_message":[],"_revision_number":2},{"id":"72a4df8ef2779f37b19e1f31b8618a99e2c8980e","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-15 09:48:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da59c603342272face5b829b8a96c38e23de6723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 11:06:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a5dc1631c9d4e7984e27d3a767bc59f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2e1b970bb32414389e2b2f6839016c4 : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d77d22ccb24c484dba24e79f25ff5499 : FAILURE in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f76702eaa1f44753b279425dd71febd6 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fda5b21f22ff4cd9aa01468dca548e71 : FAILURE in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/116b81b93a6a4aaba4e6ed31bbe2ef22 : SUCCESS in 6m 21s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/e84291b0d3b849b18dca5fb6e7791ccf : FAILURE in 21m 58s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/23cd0fa052d0477089485c90e74124c8 : RETRY_LIMIT in 2m 57s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/09f390d24d73472eb7f4b87494d368e7 : FAILURE in 22m 10s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/7219288f95e342f1a78e1824189c22f3 : SUCCESS in 1h 11m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/1b812be0fcb14040b4dc86dc01767c68 : FAILURE in 29m 51s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-victoria https://zuul.opendev.org/t/openstack/build/54bd88620f4a4fc0b84af6795dbfd101 : FAILURE in 41m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/caf05651412c4b90ba70657d25ee3f2f : SUCCESS in 37m 10s","accounts_in_message":[],"_revision_number":2},{"id":"80b9c47ebad40c9fa42ed859887e55e741ccd6f1","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-17 07:39:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad6fbbf48ec5de0d0bc6dd75c05af65cdf6de91a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 07:51:18.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/890784 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"09ab0b554a7334a3832f806144bdcb16d3940cdc","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-08-17 12:05:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"551c0ae5608b49cc2faeefea28dc410a2dabf810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 13:41:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bf23c3f40bb46488945339fb2a7d933\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f244395642ff46e1a46ea8402434f8ca : FAILURE in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8905c66444db4dc5882d75801cad04bb : FAILURE in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da03aee4a9014954a18b546f275c6c6b : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8bea12743f7483a973d728c786867d7 : FAILURE in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f240c6a2abf449cab92b73303b8cf31 : SUCCESS in 6m 54s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/fee44f07bdca4e25b5974e55be101704 : FAILURE in 41m 18s\n- barbican-dogtag-tox-functional https://zuul.opendev.org/t/openstack/build/0c59913ab54d4c52b2f7a330d94f9628 : RETRY_LIMIT in 3m 03s (non-voting)\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/92b332764e2c4b1ebd6e5e66125bbc22 : FAILURE in 28m 16s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/34ef69eaa0004881a55339beaafdef43 : SUCCESS in 1h 04m 53s (non-voting)\n- barbican-tempest-plugin-simple-crypto-victoria https://zuul.opendev.org/t/openstack/build/8f4de41130244221bb50dcb959bd0ca4 : FAILURE in 39m 25s\n- barbican-tempest-plugin-simple-crypto-ipv6-only-victoria https://zuul.opendev.org/t/openstack/build/505510e15218423fb1d096be242df267 : SUCCESS in 45m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1219a1cef4af4a2c8a63812e108028cd : SUCCESS in 33m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4167d4f0974c37e11fa0d17cf7645818d66c8536","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-09-06 19:51:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3495a6014365a008d8420ae788faba543d9af92b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-26 18:20:04.000000000","message":"Abandoned\n\nstable/victoria branch of openstack/barbican is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/victoria if you want to further work on this patch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc330f8ff4cc8a50ddea249cbdb2977a9b444664","revisions":{"6df0aad962af7721cf287d51105151a19c93f215":{"kind":"REWORK","_number":1,"created":"2023-07-31 14:44:33.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/890141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/41/890141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/41/890141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/41/890141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/41/890141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/41/890141/1"}}},"commit":{"parents":[{"commit":"8ef263b880a0959785bcc8a5512a7aa9f6121c04","subject":"Fix Story 2010258 (CVE-2022-3100)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/8ef263b880a0959785bcc8a5512a7aa9f6121c04"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-07-31 14:43:38.000000000","tz":120},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-07-31 14:44:32.000000000","tz":120},"subject":"[DNM] Testing newer version of tempest - victoria","message":"[DNM] Testing newer version of tempest - victoria\n\nDepends-On: Ib8f31e28b50b46ed148f2a816b5239ab3f531973\nChange-Id: Id7811a15fb00041544dc47c2e63bd33c50a771b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6df0aad962af7721cf287d51105151a19c93f215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6df0aad962af7721cf287d51105151a19c93f215"}]},"branch":"refs/heads/stable/victoria"},"fc330f8ff4cc8a50ddea249cbdb2977a9b444664":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-08 11:46:50.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/890141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/41/890141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/41/890141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/41/890141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/41/890141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/41/890141/2"}}},"commit":{"parents":[{"commit":"8ef263b880a0959785bcc8a5512a7aa9f6121c04","subject":"Fix Story 2010258 (CVE-2022-3100)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/8ef263b880a0959785bcc8a5512a7aa9f6121c04"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-07-31 14:43:38.000000000","tz":120},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-08-08 11:46:50.000000000","tz":0},"subject":"[DNM] Testing newer version of tempest - victoria","message":"[DNM] Testing newer version of tempest - victoria\n\nDepends-On: Ia4a30d12de1a58b93a06979188e662edeef21ec6\nChange-Id: Id7811a15fb00041544dc47c2e63bd33c50a771b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/fc330f8ff4cc8a50ddea249cbdb2977a9b444664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/fc330f8ff4cc8a50ddea249cbdb2977a9b444664"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
