)]}'
{"id":"openstack%2Fbarbican~902513","triplet_id":"openstack%2Fbarbican~master~I75fc942c9994ee1f7f998a061ab7748c5ad7edbc","project":"openstack/barbican","branch":"master","topic":"bug-2042949","attention_set":{},"removed_from_attention_set":{"36531":{"account":{"_account_id":36531,"name":"Robert Jansen","email":"robert.jansen@nl.team.blue","username":"linuzz2"},"last_update":"2023-12-06 08:47:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I75fc942c9994ee1f7f998a061ab7748c5ad7edbc","subject":"add documentation","status":"ABANDONED","created":"2023-12-04 10:20:11.000000000","updated":"2023-12-06 08:47:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4970d5059d8ebe71eee7f03f0270c97477dc8849","_number":902513,"virtual_id_number":902513,"owner":{"_account_id":36531,"name":"Robert Jansen","email":"robert.jansen@nl.team.blue","username":"linuzz2"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-04 11:13:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-04 11:13:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb89bdfba33f2eb7934e91fb6c52ad2caff156c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36531,"name":"Robert Jansen","email":"robert.jansen@nl.team.blue","username":"linuzz2"},"date":"2023-12-04 10:20:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51596b14a5b723dcadfd6669a31b183e3473de24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 11:13:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17562cc4ec204c62bfbc212e9576c7d0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc6e72cf6f6e41e491f1099fb880aa8d : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0458b54d6694df88f6c4a9003b6cbc6 : FAILURE in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51d6a5174c584d94adde0dd365c1bd20 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f0a8830f5c848f6a5457fc2cf4355aa : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c7e4dbe1db24af3921b2adce0d0ea6b : SUCCESS in 6m 26s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/8caee23692ff406bbd40fe34c1ee2d5f : SUCCESS in 22m 07s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/0023eb18030940698e650c1228195bd7 : SUCCESS in 24m 25s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/7fa773ed3740446da7d354936ac33e79 : FAILURE in 50m 14s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/081ec03e826e492b8fd627b57154a398 : SUCCESS in 44m 00s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/014a4639d07f4ec98f89e65a5f631709 : FAILURE in 43m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/b9ce40f3daa94632ac55fa10bf501a26 : SUCCESS in 27m 49s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/56fb19990a3b4906af7f68d8eb14aaf2 : SUCCESS in 52m 23s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/817004c03d424bcab7e299e7bd661af1 : SUCCESS in 44m 12s\n- barbican-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7d33f2b48aa849d8bd1767a83a4a3526 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4970d5059d8ebe71eee7f03f0270c97477dc8849","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36531,"name":"Robert Jansen","email":"robert.jansen@nl.team.blue","username":"linuzz2"},"date":"2023-12-06 08:47:41.000000000","message":"Abandoned\n\nwrong commit","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c76b70919292c31f6a69f30f9c1ee318c3cf25dc","revisions":{"c76b70919292c31f6a69f30f9c1ee318c3cf25dc":{"kind":"REWORK","_number":1,"created":"2023-12-04 10:20:11.000000000","uploader":{"_account_id":36531,"name":"Robert Jansen","email":"robert.jansen@nl.team.blue","username":"linuzz2"},"ref":"refs/changes/13/902513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/13/902513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/902513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/902513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/902513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/13/902513/1"}}},"commit":{"parents":[{"commit":"90a1d5cc55783ba24b7114cea40a2b724d0e9da6","subject":"Merge \"Add tempest to devstack how-to\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/90a1d5cc55783ba24b7114cea40a2b724d0e9da6"}]}],"author":{"name":"Robert Jansen","email":"robert@cloudvps.com","date":"2023-11-07 14:25:01.000000000","tz":60},"committer":{"name":"Robert Jansen","email":"robert@cloudvps.com","date":"2023-12-04 10:19:24.000000000","tz":60},"subject":"add documentation","message":"add documentation\n\nChange-Id: I75fc942c9994ee1f7f998a061ab7748c5ad7edbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/c76b70919292c31f6a69f30f9c1ee318c3cf25dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/c76b70919292c31f6a69f30f9c1ee318c3cf25dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
