)]}'
{"id":"openstack%2Fbarbican~920305","triplet_id":"openstack%2Fbarbican~master~I9758f83763ff7870750c8e74ee546a63f0e29491","project":"openstack/barbican","branch":"master","topic":"secret-movement","attention_set":{"32624":{"account":{"_account_id":32624,"name":"Rajiv Mucheli","email":"rajiv.mucheli@gmail.com","username":"rajivmucheli"},"last_update":"2024-05-23 17:20:13.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9758f83763ff7870750c8e74ee546a63f0e29491","subject":"Secret Transfer","status":"NEW","created":"2024-05-23 16:31:26.000000000","updated":"2024-06-04 10:19:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84474b2eaa619963835fdfacaca783463c2f7680","_number":920305,"virtual_id_number":920305,"owner":{"_account_id":32624,"name":"Rajiv Mucheli","email":"rajiv.mucheli@gmail.com","username":"rajivmucheli"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-23 17:20:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-23 17:20:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-04 10:19:47.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"CC"}],"messages":[{"id":"e85b8a209139e3b898e77faceb300eaedc61d7f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32624,"name":"Rajiv Mucheli","email":"rajiv.mucheli@gmail.com","username":"rajivmucheli"},"date":"2024-05-23 16:31:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"840814a5bfe9ea76e7e170d0e91d5823d5cb54b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32624,"name":"Rajiv Mucheli","email":"rajiv.mucheli@gmail.com","username":"rajivmucheli"},"date":"2024-05-23 16:31:43.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"842cb36a9f9252aced0292fdfcca38f5e14ff822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 17:20:13.000000000","message":"Patch Set 2: Verified-1\n\n(59 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/decc7e87782044559b255cb3281b0282\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43b69d8ee8464b3f85c89b3a8be9fd52 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/179d6f43b4934d8fb8032f1b0dcefd84 : FAILURE in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3fb724900d84eb285a3abd844c06cdb : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/647446011dd44e1d82407d92b68cdbfe : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89bbae42b40a48798e8fd1e7d293990c : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1788d002b5a3413f94674e345a73b6f5 : SUCCESS in 6m 56s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/3fdf1dd25bef4dcaa506a691b09cfda1 : SUCCESS in 35m 12s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/a187c75c97824a86acc1c609362630be : SUCCESS in 29m 24s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/849e07c5a5574d39931f6da92b2df464 : FAILURE in 43m 27s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8d5d17e6847d4368b713aa8dac837ffa : SUCCESS in 36m 12s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/e5b0207c0f304be1912d3962871faea6 : SUCCESS in 27m 33s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/e7b8c57c12594f4dbbe7f0af9c0650cc : SUCCESS in 28m 15s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/ee4f0b6472114300acf0f5e9b312e636 : SUCCESS in 30m 16s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c6aa63d89c74f878b874024c4d0c579 : SUCCESS in 34m 40s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/6fd6b5a0b5664de4b2d601350f8008b4 : SUCCESS in 34m 42s\n- barbican-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/aaa181580881410cabe54360a5827067 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"84474b2eaa619963835fdfacaca783463c2f7680","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-06-04 10:19:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9c5b7bbac31f5f924f9d760f3a518b408f944b40","revisions":{"065d134506b60547e0b42088d383ad2b2ac95687":{"kind":"REWORK","_number":1,"created":"2024-05-23 16:31:26.000000000","uploader":{"_account_id":32624,"name":"Rajiv Mucheli","email":"rajiv.mucheli@gmail.com","username":"rajivmucheli"},"ref":"refs/changes/05/920305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/05/920305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/05/920305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/05/920305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/05/920305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/05/920305/1"}}},"commit":{"parents":[{"commit":"184c2db308eab808b4710faf8ce7cfd40adbd5e6","subject":"Merge \"Remove unused test utils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/184c2db308eab808b4710faf8ce7cfd40adbd5e6"}]}],"author":{"name":"rajivmucheli","email":"rajiv.mucheli@gmail.com","date":"2024-05-23 16:30:38.000000000","tz":330},"committer":{"name":"rajivmucheli","email":"rajiv.mucheli@gmail.com","date":"2024-05-23 16:30:38.000000000","tz":330},"subject":"Secret Transfer","message":"Secret Transfer\n\nSecret transfer is a openstack feature to allows users to migrate their secrets from one user to another effectively and seamlessly.\n\nChange-Id: I9758f83763ff7870750c8e74ee546a63f0e29491\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/065d134506b60547e0b42088d383ad2b2ac95687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/065d134506b60547e0b42088d383ad2b2ac95687"}]},"branch":"refs/heads/master"},"9c5b7bbac31f5f924f9d760f3a518b408f944b40":{"kind":"NO_CHANGE","_number":2,"created":"2024-05-23 16:31:43.000000000","uploader":{"_account_id":32624,"name":"Rajiv Mucheli","email":"rajiv.mucheli@gmail.com","username":"rajivmucheli"},"ref":"refs/changes/05/920305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/05/920305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/05/920305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/05/920305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/05/920305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/05/920305/2"}}},"commit":{"parents":[{"commit":"184c2db308eab808b4710faf8ce7cfd40adbd5e6","subject":"Merge \"Remove unused test utils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/184c2db308eab808b4710faf8ce7cfd40adbd5e6"}]}],"author":{"name":"rajivmucheli","email":"rajiv.mucheli@gmail.com","date":"2024-05-23 16:30:38.000000000","tz":330},"committer":{"name":"rajivmucheli","email":"rajiv.mucheli@gmail.com","date":"2024-05-23 16:31:28.000000000","tz":330},"subject":"Secret Transfer","message":"Secret Transfer\n\nSecret transfer is a openstack feature to allows users to migrate their secrets from one user to another effectively and seamlessly.\n\nChange-Id: I9758f83763ff7870750c8e74ee546a63f0e29491\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/9c5b7bbac31f5f924f9d760f3a518b408f944b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/9c5b7bbac31f5f924f9d760f3a518b408f944b40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
