)]}'
{"id":"openstack%2Fbarbican~933461","triplet_id":"openstack%2Fbarbican~master~Ic2009a2a55622bb707e884d6a960c044b2248f52","project":"openstack/barbican","branch":"master","topic":"bug/2036506","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-11-14 18:55:09.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-11-14 18:23:05.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"Ic2009a2a55622bb707e884d6a960c044b2248f52","subject":"Configure mechanism for wrapping pKEKs","status":"MERGED","created":"2024-10-25 20:54:30.000000000","updated":"2024-11-14 18:56:24.000000000","submitted":"2024-11-14 18:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933461-bug/2036506","meta_rev_id":"d085d5397fff4f4c5ca383b2237a4b3cb394949b","_number":933461,"virtual_id_number":933461,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 18:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2024-11-14 18:23:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2024-11-14 16:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2024-11-14 18:23:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-25 21:42:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-13 15:37:38.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-11-14 16:47:43.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"}],"messages":[{"id":"76698bf86d2756ae72e27dbc04d69be00f453fb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-10-25 20:54:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17bd1e47d8dd8b259283e0eb2733fde47ad22895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 21:42:47.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fc94a8226e942e69ad6b3539d85999d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c662d9af94e94b30806cf115557b444e : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df03cdb7eaf0486786416bfe9829bc06 : FAILURE in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77dc38af0be240c4b743a91acf9a5840 : FAILURE in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60f44aa5d65e484f82b0a4987e57526c : FAILURE in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1439b4a4525c4e3590dd6c4b87605994 : SUCCESS in 4m 40s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/b400401c7f08450085c66f5c288f36a8 : SUCCESS in 31m 48s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/22384b2214d3412499210d6213ca70c0 : SUCCESS in 33m 35s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/79d478dd2b6b40b49775cae8dcf72fa6 : FAILURE in 29m 00s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7613f16df4cc4dd9aa806383571db29b : SUCCESS in 33m 51s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/de1fe78b567e446eb2d4ab3fa99f93fe : SUCCESS in 33m 25s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d344b6f10caf41289aa749a780c5f145 : SUCCESS in 19m 44s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/ce3c8c3bdbad458d847a25d3c215e49a : SUCCESS in 47m 32s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5d855f72fc8046a19bb27d5bb67ed54c : SUCCESS in 32m 01s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/d8fe5a4969594569b8aa0a2142202ccc : SUCCESS in 32m 12s","accounts_in_message":[],"_revision_number":1},{"id":"cb2485afbb73838fc1832b4cbb3346c1d8b6f26e","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-10-29 18:00:18.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"65231e5a9f91def41a93fd9bc857e1f06fae6d40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-10-30 19:18:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a12f149c06ecaf343436677cca36a8035133a944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 20:13:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/619e71b9c25842769b9011e1d0c947af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/727d00ca4e9c43bea81e199094adae93 : FAILURE in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb40791c8bfa486aa7ce3731412954f0 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34139a7c940a4dd59fa8f78e0fe41cdd : FAILURE in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8db92634a5404d55b3db0ecd56af1904 : FAILURE in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310a0065877d4c40a9cc52d2b5b1bed0 : SUCCESS in 6m 28s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/02f63374bb404b3a80441b9bc92c0294 : SUCCESS in 22m 56s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/912c94f5fcad4f74b2a13279e7b72017 : SUCCESS in 26m 24s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/3d9fc5627bb840388f4e830822cdc66d : FAILURE in 37m 24s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3fb8fa80f2474cd6a0802e2e40cb5f70 : SUCCESS in 24m 57s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/464e190960fc47b38c7d7c61c27064c5 : SUCCESS in 26m 26s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/78d6bf70e79d409286b981a23a06ecb6 : SUCCESS in 28m 00s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/32ee3fcad69e4474a0358ca3a58a7c7f : SUCCESS in 32m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ddec1c02d767415b822f8d630d02d098 : SUCCESS in 52m 55s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/b6c0b335b52b4882af617f936ee5e882 : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":2},{"id":"89a9fcc07a2278ff5db3c79e3dbfc0e1f36b05e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-10-31 19:38:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8627c236890850a859b45fc3b927bb614121e40d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 20:24:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34b3339b20784f99a9c1fb4e7db7562c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9817fb4c61b54700858197ea7c17e457 : FAILURE in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2c8cee95c2f4a30a5ff97b9e839f885 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a35ae9364c0444d5a51df7ebeaf48198 : FAILURE in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e51c9faae0e241bcafcc0e163eda02e9 : FAILURE in 15m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4625245bfda4464aa34c18610f8cbfb : SUCCESS in 6m 46s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/bd541b73916844a0be6f4094404bcb22 : SUCCESS in 22m 49s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/81665e3c8a564f1e94b70e7d0d551a12 : SUCCESS in 24m 23s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/6209d67f3b1c4775a263e50ddd0b1287 : FAILURE in 42m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/eda5e58b3265441193ed339fd94fe51d : SUCCESS in 17m 36s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/e13855b55dd743ae8afa744a958b2a17 : SUCCESS in 22m 15s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/a7ed67a50e9d4ffe8832bb84723a2e90 : SUCCESS in 18m 53s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/323fd8805acd4dc98d327c3ff98323e3 : SUCCESS in 32m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c3118425eceb4851b1808da78cd83339 : SUCCESS in 33m 19s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/bf743bcf70cb40dd9406be5c2d19ed45 : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":3},{"id":"277dfd19d29a339b5d7d1297c0aab60b0220e314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-11-04 20:52:59.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5925a9a970ba619a203c4fe9436095158d278290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 21:42:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/edd51c4db82e4c21898a217b7e727c83\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f52765a191a94f0498b362a8a6effbe2 : FAILURE in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a6adf0b5f7c476eb76b8ac1ee7d39b5 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/196b919675fb44428a0ac5835949283c : FAILURE in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e025a3ee925f4d0091a17b61018714d2 : FAILURE in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e116f5ba29640f3a87095798e56b8e5 : SUCCESS in 5m 49s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/dd2f92c4d8744097b7434ea4a0299fa6 : SUCCESS in 20m 35s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/b789a91cb1e847169f9e86f77b14e56e : SUCCESS in 19m 33s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/117e51b7232540f68a2661813fd6113e : FAILURE in 29m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8b0a5725b6e44b0fbc66ad1c230832ad : SUCCESS in 22m 07s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/ce707bedac15486285827d07e155cb28 : SUCCESS in 20m 14s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/85c9bb9581c1445eb56fc92d5d534f2b : SUCCESS in 21m 08s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/9bc04a3b879e4356a03f48b4d34cf608 : SUCCESS in 22m 39s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8a4e3c03b0564d99b5b54d18bdd83fd9 : SUCCESS in 39m 58s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/f6223f5747b141fba4fa37f1652e7cc7 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":4},{"id":"b4afdec344e1d34526a46a6083c8090958f58d9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-11-06 04:22:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"065050203cc56ac30d3b6569b071f2015334a8ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 05:16:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6061250e01ab4912b7cdeb02e30180f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86d562645f914993a34cb555ea419e92 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c3e0446935a4e3db2b1b01118390070 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96d6258acb9941d4a163d6f17ec19c9d : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c089ccf15f1b48ed8bfd4e125ce22a3d : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c272bc121c644ea59efc6dba75dded76 : SUCCESS in 5m 58s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/f286c5faa3884bbeb56339da9b78cb3b : SUCCESS in 19m 17s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/f1cbc63d95b24e4296ace4b595e30427 : SUCCESS in 21m 21s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/5c98ec0a17c043388d52bf57668be836 : FAILURE in 35m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6a2b4f74169d47648f1b38a38ff64373 : SUCCESS in 46m 46s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/1ebe490fa357482db827e284e2575de4 : SUCCESS in 30m 52s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/310c1545a8c94777b1498179e129fbf7 : SUCCESS in 20m 42s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/a2860108316241a0be56daf2fa85289d : SUCCESS in 27m 16s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bf54d49135b64fa2bc067f24117be743 : SUCCESS in 37m 20s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/dc4c6db499994536960bc425961ad09e : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":5},{"id":"928cd5d816e5738c9494c6ab31112cc8fde2d5a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-11-08 19:21:37.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"53a3bf8723de7b532cf1ad8744fe6b54001a4654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 20:10:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d55350ba4884737b9a1820bd3d4dd4c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/144bd99a79b24b3aaced944ede969090 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90388fc68f024597b4d63fba45569822 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8da0e241a3d44e26beca1d089271fb0c : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31a0b2f529da4243b867d06720a42f3a : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d169f03fc4ac4afa91819d3bfa8c13f1 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77ee020cf6c94e9f9ebf486ed4718cbd : SUCCESS in 5m 37s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/88d98c49ee304620bddd277806d3725e : SUCCESS in 29m 09s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/c69a3d01b1054d968695d1ff19d7a789 : SUCCESS in 28m 00s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/2ed6749c38974d2394f9e9e7ae3dd840 : FAILURE in 35m 45s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d8fd13fcdfeb462084d7cc1b32dd6521 : SUCCESS in 21m 34s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/7f36baca9af14d9dab5c7dcb805cc663 : SUCCESS in 22m 02s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/a8e435d2efb04f77a66bd8c97a2b5f4b : SUCCESS in 21m 40s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/d5e11ff55fd04cc985e0087db8c771d0 : SUCCESS in 48m 29s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/96f620e8174f44a1a80da22752579a21 : SUCCESS in 33m 04s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/c80d372c03da49e58c2a697386fdbf76 : SUCCESS in 31m 26s","accounts_in_message":[],"_revision_number":6},{"id":"f15058c3818ac8fd860b5c5384ba1cab7b438eea","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-11-13 15:37:38.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"76099274be499cca5cda40c87850bd083cb26adf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-11-13 20:43:08.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"31235636651c73896d5a6d059bfe1eaf1135808a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 21:34:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86035cd3cc3844dfa0bce5cac4b959b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac6662dd78974b31a925ded68d5a9ce6 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e5883bb718a4758a90000c3ac76ace4 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51fb9bbb3aa94649ac963e334106409d : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99b03b5ecef64618aa5cb8bdd9340052 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31635e41204940d1b1fb2e2dc180fa40 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/763952f160fd429e9a9301997465b759 : SUCCESS in 5m 29s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/8cf9b509bec747d38822c11134b8e6f6 : SUCCESS in 22m 35s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/0d76a7fbd0014bfbb69bec5ed1f415a6 : SUCCESS in 30m 14s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/3f5f40ec30ae456ba9b92f50bd841de5 : FAILURE in 35m 28s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fc7cab5afc5a49d2a42800edba468c8a : SUCCESS in 31m 14s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/5b7d8061d1ba4a8bb9b8b74859031152 : SUCCESS in 35m 05s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/15c09fb8708a42efa36b597211503ab5 : SUCCESS in 24m 17s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/0a2f72d6792248098afcaf91467e82f8 : SUCCESS in 41m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/754eac27aabe402fa8516f582f21f070 : SUCCESS in 42m 40s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/c2381a0821f8490583856af027e655f8 : SUCCESS in 50m 27s","accounts_in_message":[],"_revision_number":7},{"id":"fafca87bc7c909f00bea15aabc0abcf00211ebd4","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2024-11-14 16:47:43.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"a43b065d7ec91ff9b5801bc051270f11ced5dd07","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-11-14 18:23:05.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"ac4d0f12ccd487966a5b71606ad77d93a292e59e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 18:23:35.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a626ea15ff218753697c2d34dd61c88ba2722c82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 18:55:09.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fa2b1b8be84462498b31a8235eae1bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f97f641c70fb451496434c1351f7648a : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c45237f9701241b9b7edc3e3052536d3 : SUCCESS in 6m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/061c7343d8a44837829659f072bbf029 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1994caccb9b4f489385642f44b92663 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98a37d39a96248578ace68819fcdc746 : SUCCESS in 6m 19s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/7899a8e7760a4aa6aab9fe02466b52ce : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":7},{"id":"5c6b0a4aa6d8a673f91097621c8be1f977873f5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 18:55:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"d085d5397fff4f4c5ca383b2237a4b3cb394949b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 18:56:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/817e51c25a6d4fa79dee3a4cf3818af6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5cdd4ade2d24566a7a1f60aa0dca0b6 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d2200c1e3604419a7b65ec96f3467f1 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0d4101fa5da52f242ab0a52955f67769b23485a1","revisions":{"5f58c10b9a7af8f39b281f8fd85770355902be00":{"kind":"REWORK","_number":1,"created":"2024-10-25 20:54:30.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/61/933461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/61/933461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/61/933461/1"}}},"commit":{"parents":[{"commit":"97ce039f266f4a9ffef3798b8d0f6ac71783fd40","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/97ce039f266f4a9ffef3798b8d0f6ac71783fd40"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-10-25 20:45:58.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-10-25 20:54:18.000000000","tz":-240},"subject":"Configure mechanism for wrapping pKEKs","message":"Configure mechanism for wrapping pKEKs\n\nThe PKCS#11 backend key-wraps (encrypts) the project-specific Key\nEncryption Keys (pKEKs) using the master encryption key (MKEK).\n\nThe mechanism for wrapping/unwrapping the keys was hard-coded to use\nCKM_AES_CBC_PAD.  This patch refactors the pkcs11 module to make this\nmechanism configurable.\n\nThis is necessary to fix Bug #2036506 because some PKCS#11 devices and\nsoftware implementations no longer allow CKM_AES_CBC_PAD to be used for\nkey wrapping.\n\nSupported key wrap mechanisms now include:\n\n* CKM_AES_CBC_PAD\n* CKM_AES_KEY_WRAP_PAD\n* CKM_AES_KEY_WRAP_KWP\n\nCloses-Bug: #2036506\nChange-Id: Ic2009a2a55622bb707e884d6a960c044b2248f52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/5f58c10b9a7af8f39b281f8fd85770355902be00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/5f58c10b9a7af8f39b281f8fd85770355902be00"}]},"branch":"refs/heads/master"},"4565521c380af30d266c5f00ad2cd6f69cec8312":{"kind":"REWORK","_number":2,"created":"2024-10-30 19:18:49.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/61/933461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/61/933461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/61/933461/2"}}},"commit":{"parents":[{"commit":"97ce039f266f4a9ffef3798b8d0f6ac71783fd40","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/97ce039f266f4a9ffef3798b8d0f6ac71783fd40"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-10-25 20:45:58.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2024-10-30 19:18:45.000000000","tz":0},"subject":"Configure mechanism for wrapping pKEKs","message":"Configure mechanism for wrapping pKEKs\n\nThe PKCS#11 backend key-wraps (encrypts) the project-specific Key\nEncryption Keys (pKEKs) using the master encryption key (MKEK).\n\nThe mechanism for wrapping/unwrapping the keys was hard-coded to use\nCKM_AES_CBC_PAD.  This patch refactors the pkcs11 module to make this\nmechanism configurable.\n\nThis is necessary to fix Bug #2036506 because some PKCS#11 devices and\nsoftware implementations no longer allow CKM_AES_CBC_PAD to be used for\nkey wrapping.\n\nSupported key wrap mechanisms now include:\n\n* CKM_AES_CBC_PAD\n* CKM_AES_KEY_WRAP_PAD\n* CKM_AES_KEY_WRAP_KWP\n\nCloses-Bug: #2036506\nChange-Id: Ic2009a2a55622bb707e884d6a960c044b2248f52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/4565521c380af30d266c5f00ad2cd6f69cec8312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/4565521c380af30d266c5f00ad2cd6f69cec8312"}]},"branch":"refs/heads/master"},"6affe57a8cac6262d885ccd716c870c026a9c525":{"kind":"REWORK","_number":3,"created":"2024-10-31 19:38:59.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/61/933461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/61/933461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/61/933461/3"}}},"commit":{"parents":[{"commit":"97ce039f266f4a9ffef3798b8d0f6ac71783fd40","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/97ce039f266f4a9ffef3798b8d0f6ac71783fd40"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-10-25 20:45:58.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2024-10-31 19:38:38.000000000","tz":0},"subject":"Configure mechanism for wrapping pKEKs","message":"Configure mechanism for wrapping pKEKs\n\nThe PKCS#11 backend key-wraps (encrypts) the project-specific Key\nEncryption Keys (pKEKs) using the master encryption key (MKEK).\n\nThe mechanism for wrapping/unwrapping the keys was hard-coded to use\nCKM_AES_CBC_PAD.  This patch refactors the pkcs11 module to make this\nmechanism configurable.\n\nThis is necessary to fix Bug #2036506 because some PKCS#11 devices and\nsoftware implementations no longer allow CKM_AES_CBC_PAD to be used for\nkey wrapping.\n\nSupported key wrap mechanisms now include:\n\n* CKM_AES_CBC_PAD\n* CKM_AES_KEY_WRAP_PAD\n* CKM_AES_KEY_WRAP_KWP\n\nCloses-Bug: #2036506\nChange-Id: Ic2009a2a55622bb707e884d6a960c044b2248f52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6affe57a8cac6262d885ccd716c870c026a9c525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6affe57a8cac6262d885ccd716c870c026a9c525"}]},"branch":"refs/heads/master"},"f905279b457c4d404dd67e4566e4bd45b220c2cb":{"kind":"REWORK","_number":4,"created":"2024-11-04 20:52:59.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/61/933461/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/61/933461/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/61/933461/4"}}},"commit":{"parents":[{"commit":"97ce039f266f4a9ffef3798b8d0f6ac71783fd40","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/97ce039f266f4a9ffef3798b8d0f6ac71783fd40"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-10-25 20:45:58.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-11-04 20:52:25.000000000","tz":-300},"subject":"Configure mechanism for wrapping pKEKs","message":"Configure mechanism for wrapping pKEKs\n\nThe PKCS#11 backend key-wraps (encrypts) the project-specific Key\nEncryption Keys (pKEKs) using the master encryption key (MKEK).\n\nThe mechanism for wrapping/unwrapping the keys was hard-coded to use\nCKM_AES_CBC_PAD.  This patch refactors the pkcs11 module to make this\nmechanism configurable.\n\nThis is necessary to fix Bug #2036506 because some PKCS#11 devices and\nsoftware implementations no longer allow CKM_AES_CBC_PAD to be used for\nkey wrapping.\n\nSupported key wrap mechanisms now include:\n\n* CKM_AES_CBC_PAD\n* CKM_AES_KEY_WRAP_PAD\n* CKM_AES_KEY_WRAP_KWP\n\nCloses-Bug: #2036506\nChange-Id: Ic2009a2a55622bb707e884d6a960c044b2248f52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f905279b457c4d404dd67e4566e4bd45b220c2cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f905279b457c4d404dd67e4566e4bd45b220c2cb"}]},"branch":"refs/heads/master"},"b80ceeff2cd23255a8ac9cf677c2c4d019ffd354":{"kind":"REWORK","_number":5,"created":"2024-11-06 04:22:34.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/61/933461/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/61/933461/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/61/933461/5"}}},"commit":{"parents":[{"commit":"2ec87245850b72e0c594e41f260a8de68fc81fa1","subject":"Merge \"Remove Python 3.8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/2ec87245850b72e0c594e41f260a8de68fc81fa1"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-10-25 20:45:58.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-11-06 04:21:53.000000000","tz":-300},"subject":"Configure mechanism for wrapping pKEKs","message":"Configure mechanism for wrapping pKEKs\n\nThe PKCS#11 backend key-wraps (encrypts) the project-specific Key\nEncryption Keys (pKEKs) using the master encryption key (MKEK).\n\nThe mechanism for wrapping/unwrapping the keys was hard-coded to use\nCKM_AES_CBC_PAD.  This patch refactors the pkcs11 module to make this\nmechanism configurable.\n\nThis is necessary to fix Bug #2036506 because some PKCS#11 devices and\nsoftware implementations no longer allow CKM_AES_CBC_PAD to be used for\nkey wrapping.\n\nSupported key wrap mechanisms now include:\n\n* CKM_AES_CBC_PAD\n* CKM_AES_KEY_WRAP_PAD\n* CKM_AES_KEY_WRAP_KWP\n\nCloses-Bug: #2036506\nChange-Id: Ic2009a2a55622bb707e884d6a960c044b2248f52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b80ceeff2cd23255a8ac9cf677c2c4d019ffd354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b80ceeff2cd23255a8ac9cf677c2c4d019ffd354"}]},"branch":"refs/heads/master"},"27debbf065a34b1268b62ad449a315c85cc7c6d1":{"kind":"REWORK","_number":6,"created":"2024-11-08 19:21:37.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/61/933461/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/61/933461/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/61/933461/6"}}},"commit":{"parents":[{"commit":"2ec87245850b72e0c594e41f260a8de68fc81fa1","subject":"Merge \"Remove Python 3.8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/2ec87245850b72e0c594e41f260a8de68fc81fa1"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-10-25 20:45:58.000000000","tz":-240},"committer":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-11-08 19:21:33.000000000","tz":-300},"subject":"Configure mechanism for wrapping pKEKs","message":"Configure mechanism for wrapping pKEKs\n\nThe PKCS#11 backend key-wraps (encrypts) the project-specific Key\nEncryption Keys (pKEKs) using the master encryption key (MKEK).\n\nThe mechanism for wrapping/unwrapping the keys was hard-coded to use\nCKM_AES_CBC_PAD.  This patch refactors the pkcs11 module to make this\nmechanism configurable.\n\nThis is necessary to fix Bug #2036506 because some PKCS#11 devices and\nsoftware implementations no longer allow CKM_AES_CBC_PAD to be used for\nkey wrapping.\n\nSupported key wrap mechanisms now include:\n\n* CKM_AES_CBC_PAD\n* CKM_AES_KEY_WRAP_PAD\n* CKM_AES_KEY_WRAP_KWP\n\nCloses-Bug: #2036506\nChange-Id: Ic2009a2a55622bb707e884d6a960c044b2248f52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/27debbf065a34b1268b62ad449a315c85cc7c6d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/27debbf065a34b1268b62ad449a315c85cc7c6d1"}]},"branch":"refs/heads/master"},"0d4101fa5da52f242ab0a52955f67769b23485a1":{"kind":"REWORK","_number":7,"created":"2024-11-13 20:43:08.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/61/933461/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/61/933461/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/61/933461/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/61/933461/7"}}},"commit":{"parents":[{"commit":"2ec87245850b72e0c594e41f260a8de68fc81fa1","subject":"Merge \"Remove Python 3.8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/2ec87245850b72e0c594e41f260a8de68fc81fa1"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-10-25 20:45:58.000000000","tz":-240},"committer":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-11-13 20:42:30.000000000","tz":-300},"subject":"Configure mechanism for wrapping pKEKs","message":"Configure mechanism for wrapping pKEKs\n\nThe PKCS#11 backend key-wraps (encrypts) the project-specific Key\nEncryption Keys (pKEKs) using the master encryption key (MKEK).\n\nThe mechanism for wrapping/unwrapping the keys was hard-coded to use\nCKM_AES_CBC_PAD.  This patch refactors the pkcs11 module to make this\nmechanism configurable.\n\nThis is necessary to fix Bug #2036506 because some PKCS#11 devices and\nsoftware implementations no longer allow CKM_AES_CBC_PAD to be used for\nkey wrapping.\n\nSupported key wrap mechanisms now include:\n\n* CKM_AES_CBC_PAD\n* CKM_AES_KEY_WRAP_PAD\n* CKM_AES_KEY_WRAP_KWP\n\nCloses-Bug: #2036506\nChange-Id: Ic2009a2a55622bb707e884d6a960c044b2248f52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/0d4101fa5da52f242ab0a52955f67769b23485a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/0d4101fa5da52f242ab0a52955f67769b23485a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
