)]}'
{"id":"openstack%2Fbarbican~933513","triplet_id":"openstack%2Fbarbican~master~I12f38ce31a876f780a53fc727c4b1d32e7ce8de2","project":"openstack/barbican","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-30 15:30:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I12f38ce31a876f780a53fc727c4b1d32e7ce8de2","subject":"Add note about requirements lower bounds","status":"MERGED","created":"2024-10-26 15:37:04.000000000","updated":"2024-10-30 15:31:39.000000000","submitted":"2024-10-30 15:30:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933513","meta_rev_id":"69a9e60234d7f781a038c6644fb1026f41ba7803","_number":933513,"virtual_id_number":933513,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-30 15:30:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":2,"date":"2024-10-30 15:02:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":1,"date":"2024-10-30 15:02:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-26 16:18:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-30 15:02:23.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"3d468a93b691ff02fa5e077eb842f8faad405b6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 15:37:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a01d8cf6d725b6ea87e04782a67aa9b2c4e293e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 16:18:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/465702f50b1c4d1185ef7112d4fb3ae8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9d7f003ca71540cc987dd182a6c8666e : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fda365843b04ed1bff3626ddacbc472 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a45423d03bec49229ebe659a2b656059 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b0bad3fed7f43d3bdd01fef9fe1278f : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df7fe33c87f84f218c3680543dcf31d1 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75fedd6682214989b18a02c8de55fd6a : SUCCESS in 6m 35s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/29aaed8df53542f398a6b9bf43482906 : SUCCESS in 18m 28s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/9e62bf8089b0487e8eacaf4bc2ede6ef : SUCCESS in 28m 25s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/1a4a31241fde4ef395232c68d936ba2e : FAILURE in 29m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ab3963f8deb94dd79c6d2d5b9a536ce9 : SUCCESS in 18m 46s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/04a0179fd70c4794b71ff1cc485f06b5 : SUCCESS in 36m 00s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/6a7401dc25f149779c13e5bf0f7dfaf9 : SUCCESS in 18m 34s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/e902d49fa9a24bacb41110d76a868ae3 : SUCCESS in 41m 00s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e24c83a95fb74daa80f74c8cf98fa9f7 : SUCCESS in 31m 36s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/24d6b3845b9c4cdcaa919fcc49f4b7f8 : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e258d1e300104be0f041b9ed60c2c1e938fd69fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 18:30:18.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6256ce792be4732d04f34a2482f2b56cd0f90b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 19:15:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd36d80214d641b38795e539f1c2f05b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/888162c0d1af4013bec86309121b37cb : SUCCESS in 1m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb32e3d9ec7b4e5a934b4e3191677c63 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a9c4bd245694df18c0f6fccf866ba30 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bc3ab2b50504887a93c4d3729a0dec5 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2f4626e3d5f426a81e91d3af6fd09b7 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ae18c79620436d9073d19a779b10d4 : SUCCESS in 5m 50s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/e654f0b782e44a11a234e73e6c0eab27 : SUCCESS in 23m 40s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/53c260b7adc5491caade80b1991cd3a1 : SUCCESS in 25m 25s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/708f93d16ea54a91b575602eca8283d4 : FAILURE in 34m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/55908748c6d0437db68444815e9c4c49 : SUCCESS in 19m 23s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/6f5a097d10874531b97861a0aab6d9d1 : SUCCESS in 32m 25s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/dd75c5ff2628455c989888f6d7f699da : SUCCESS in 32m 00s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/757bd9962ce34be2b3a2f7d8cb755c8e : SUCCESS in 38m 26s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5bd7b3dc413e4d3d9a64a02ccdb4ad91 : SUCCESS in 38m 44s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/403ef154b0fc440ea454ffdf659d513e : SUCCESS in 41m 15s","accounts_in_message":[],"_revision_number":2},{"id":"bb5d834292a364b36ef63213519ec0cc581689fa","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-10-30 15:02:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ca23c5ce5d775cb8c4be9d7258dae955206c91b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 15:02:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"19bbcbbd60a8aa78033622bbf4a2f5b3e127abca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 15:30:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b780e9863b7a4002a801c5283c19bd70\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4447e99bbd7e4782a8656e00018859b9 : SUCCESS in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16a81e8095dd4ede8111942a4994f8b4 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31f433b504b94ab3bdc1ac9176b052b2 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95725e43230a49c483d44a68925c2560 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6893851930814fafa69059a0ad682d89 : SUCCESS in 3m 57s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/9c4af1c661a342f78593f20ba8ad1335 : SUCCESS in 19m 58s","accounts_in_message":[],"_revision_number":2},{"id":"f1fdfcf7bd1ed71dae5e3c89d113a0ae240cf996","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 15:30:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"69a9e60234d7f781a038c6644fb1026f41ba7803","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 15:31:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f8cbb8e70274d3a8f3bb19d14e02cb1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99030eed833446594ec63c8e715a873 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"52f4d5ca17a321f81919ed858c3facc32befa69f","revisions":{"b017260e9270d9efbfb4d2441522efb9e0ed1c16":{"kind":"REWORK","_number":1,"created":"2024-10-26 15:37:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/933513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/13/933513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/933513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/933513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/933513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/13/933513/1"}}},"commit":{"parents":[{"commit":"e1f7b30e9381932fe26f31d4a1f675be7d9242d9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/e1f7b30e9381932fe26f31d4a1f675be7d9242d9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 15:36:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 15:36:55.000000000","tz":540},"subject":"Add note about requirements lower bounds","message":"Add note about requirements lower bounds\n\nSince lower constraints job was removed the bounds are not actually\ntested. Add a note to explain the bounds are maintained on bast effort\nbasis.\n\nChange-Id: I12f38ce31a876f780a53fc727c4b1d32e7ce8de2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b017260e9270d9efbfb4d2441522efb9e0ed1c16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/b017260e9270d9efbfb4d2441522efb9e0ed1c16"}]},"branch":"refs/heads/master"},"52f4d5ca17a321f81919ed858c3facc32befa69f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-26 18:30:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/933513/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/13/933513/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/933513/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/933513/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/13/933513/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/13/933513/2"}}},"commit":{"parents":[{"commit":"e1f7b30e9381932fe26f31d4a1f675be7d9242d9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/e1f7b30e9381932fe26f31d4a1f675be7d9242d9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 15:36:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 18:30:18.000000000","tz":0},"subject":"Add note about requirements lower bounds","message":"Add note about requirements lower bounds\n\nSince lower constraints job was removed the bounds are not actually\ntested. Add a note to explain the bounds are maintained on best effort\nbasis.\n\nChange-Id: I12f38ce31a876f780a53fc727c4b1d32e7ce8de2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/52f4d5ca17a321f81919ed858c3facc32befa69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/52f4d5ca17a321f81919ed858c3facc32befa69f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
