)]}'
{"id":"openstack%2Fbarbican~937293","triplet_id":"openstack%2Fbarbican~master~If52404cd32ded7ef33b98878ea62d27b15c00460","project":"openstack/barbican","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2025-04-11 16:13:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If52404cd32ded7ef33b98878ea62d27b15c00460","subject":"DNM: Testing gate health","status":"ABANDONED","created":"2024-12-06 20:42:02.000000000","updated":"2025-04-11 16:13:21.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0e385f9d401a34d06de6aa5c12e93d0d0db5a3d7","_number":937293,"virtual_id_number":937293,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-22 21:13:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 21:44:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8b91bf417c2bab1479d7253e0c456c0cb85f239","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-12-06 20:42:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eee1cf997fa37cc445b7dd266169242a7504032a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 21:44:50.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f8f66dca7b9422dbdecf0eefd003800\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cd752749959480689723f697dd56c4d : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6adf63b0b0246baa3aff1a45d7f4d1c : FAILURE in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ef213a021ac45209698e68c02a86d0c : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/612e089990e14dbfbf3c9a425fcdb6a4 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd52278307f14d7bad882bdf6e465eab : SUCCESS in 6m 39s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/5c7b314e11064c5a99f042b3592d7414 : SUCCESS in 26m 06s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/d61c4be262c741ada8511a48b0567c29 : SUCCESS in 25m 05s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/2a96378ddc5a4b21bad5b89f47603d6d : FAILURE in 35m 55s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/00ee444b62c244c6b2e56afe72312434 : SUCCESS in 24m 36s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/e668b222b07c4dfea1da325c9a53c83e : SUCCESS in 18m 51s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/4f76fcc087ab466ea0dfe9846c469a98 : SUCCESS in 29m 59s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/4ad88ad1b8504aac9cd0669d4d97b473 : SUCCESS in 37m 58s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/f63d1940a37b49f697f093a6dbf5be80 : SUCCESS in 35m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18a4aaeb2e174ba4842121b2cead5da3 : FAILURE in 55m 01s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/28908111045a43c49ed579cd42c33fc1 : FAILURE in 42m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f935a7b43ace017dcc5d169337cb6bede96aab31","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-12-11 16:44:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e8deb899733ddc42cd3503a0acd09b688bc769b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 17:50:13.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/880e06e468b043d9b869ba7452d8e4c7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beff1fb442024b6887d689c5299efd49 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff99d35456514e4e994ad5f9c5fb0c19 : FAILURE in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cf4f3fc4c414a32ae8d2a876c1a3c76 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18368438bf4f493ba0b5be63690b982c : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce648c80025c40a1b7bfc87e6c537365 : SUCCESS in 6m 14s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/445a75f0d01546f5bf8e3b69dbc90033 : SUCCESS in 23m 52s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/514c018a0c964f169001894715a20bc0 : SUCCESS in 25m 36s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/f23bd33af9174cc79e52f8c06c1dc95d : FAILURE in 50m 43s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d84b7f288ca842b7bd9311debdcc1f3c : SUCCESS in 38m 25s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/994046d0e9a843a8bd76c017be753d93 : SUCCESS in 34m 27s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/fe191b98249a40a6b44057291625ce8d : SUCCESS in 33m 35s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/3f40850bd5f2492892b559b2373f1d2c : SUCCESS in 24m 41s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/082cbfd8803a46b3bf61ffc4a232214a : FAILURE in 43m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17d442f4f6c249cba6ead8f583bc4e3b : FAILURE in 40m 34s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/d6c44c25491247a294468468db3d28db : FAILURE in 53m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b40a6e7b3d3ba0950da210e17ac4d52f27eae7eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-01-22 20:15:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9a4b20ccbb8fa0ea7947638901bc6144c3ec54d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 21:13:56.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bbbb7ca8ebf4add800365c3c5e9f4e0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9babc693baad44378de129251c127d6d : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a34ee319018546e9b492c6f6a45d21c8 : FAILURE in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40b1421b8e414b669bd171ed9bb868d3 : SUCCESS in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50b80cd5adec4ecaa31b89201222faf8 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0fd90a906445e7a2e71e0df2fc8b59 : SUCCESS in 6m 49s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/6ce265e718724effa0b63cb559890b65 : SUCCESS in 37m 49s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/e63e15ffaf864e9593957af60fbb7ca1 : SUCCESS in 38m 42s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/4608120d797c4901a20f5e4504f0b654 : FAILURE in 31m 27s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a26db4fd83d848938e88367027991cbc : SUCCESS in 20m 32s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/1f862c72982147519a9c3b0dfdafa044 : SUCCESS in 27m 22s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/c2d004018b7944218025c61eb605a087 : SUCCESS in 21m 37s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/d6f021b737e94a7996aab9d490db364b : SUCCESS in 31m 28s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/e6bd0a9f27344f2bbd807e8d0853a12d : SUCCESS in 28m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c630271cdb5c4047887f48a0c1f97ad0 : SUCCESS in 56m 17s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/c1433e640b6345ff9f19de958cb13ddc : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":2},{"id":"0e385f9d401a34d06de6aa5c12e93d0d0db5a3d7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-04-11 16:13:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6eccaba8628544b4b6cab8f8023113595c8481db","revisions":{"f26b4a807a29b462a08669deaf446318496e54f4":{"kind":"REWORK","_number":1,"created":"2024-12-06 20:42:02.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/93/937293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/93/937293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/93/937293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/93/937293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/93/937293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/93/937293/1"}}},"commit":{"parents":[{"commit":"0589f7fb346b03bbfa4801756a839eaddb11c120","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/0589f7fb346b03bbfa4801756a839eaddb11c120"}]}],"author":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-12-06 20:41:37.000000000","tz":-300},"committer":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-12-06 20:41:37.000000000","tz":-300},"subject":"DNM: Testing gate health","message":"DNM: Testing gate health\n\nChange-Id: If52404cd32ded7ef33b98878ea62d27b15c00460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f26b4a807a29b462a08669deaf446318496e54f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/f26b4a807a29b462a08669deaf446318496e54f4"}]},"branch":"refs/heads/master"},"6eccaba8628544b4b6cab8f8023113595c8481db":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-01-22 20:15:34.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/93/937293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/93/937293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/93/937293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/93/937293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/93/937293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/93/937293/2"}}},"commit":{"parents":[{"commit":"55c758eb7c119d043a3b1365f4864308e60dd978","subject":"Merge \"Remove unused versionbuild script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/55c758eb7c119d043a3b1365f4864308e60dd978"}]}],"author":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2024-12-06 20:41:37.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2025-01-22 20:15:34.000000000","tz":0},"subject":"DNM: Testing gate health","message":"DNM: Testing gate health\n\nChange-Id: If52404cd32ded7ef33b98878ea62d27b15c00460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6eccaba8628544b4b6cab8f8023113595c8481db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/6eccaba8628544b4b6cab8f8023113595c8481db"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
