)]}'
{"id":"openstack%2Fbarbican~945643","triplet_id":"openstack%2Fbarbican~master~I6f01608548defd5182137d7287009ec7ab1756ea","project":"openstack/barbican","branch":"master","topic":"raise-exc","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2025-04-04 15:30:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6f01608548defd5182137d7287009ec7ab1756ea","subject":"Raise HTTPUnauthorized in ContextMiddleware","status":"MERGED","created":"2025-03-26 18:01:22.000000000","updated":"2025-04-04 15:31:46.000000000","submitted":"2025-04-04 15:30:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945643-raise-exc","meta_rev_id":"11791ca16ab290823b284786bcebb7d14ae20eee","_number":945643,"virtual_id_number":945643,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-04 15:30:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2025-04-04 14:51:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2025-03-28 14:33:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2025-04-04 14:51:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13636,"name":"Anton A","display_name":"Anton","email":"arefyevanton@gmail.com","username":"aarefiev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 18:01:22.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":13636,"name":"Anton A","display_name":"Anton","email":"arefyevanton@gmail.com","username":"aarefiev"},"state":"CC"},{"updated":"2025-03-26 18:41:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-28 14:33:00.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-04-04 14:51:29.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"51c533d4795e53fc4fd5f6fccefa8de61d40533a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-03-26 18:01:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5454451641bd12a02f3ceaa7ee32418a7148b14d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 18:41:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5399ddae0ea4b87ac987a33392903da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b40838482a74ce0bb0deecac31a5921 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/052ed39205f0454fbfe4716453212391 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/245701da0d36439b9a72cbff9c53f992 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c77f26869be0405f9dba1e3793519ca7 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc81b1f307f14678b26767b9e691ef93 : SUCCESS in 9m 32s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/47e99b296e2f45d7893f73cade3c471a : SUCCESS in 25m 39s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/54e2763a7c794122a00575a87ca3f41f : SUCCESS in 26m 07s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/477ab5110f0f414b821f0b8239c80d53 : FAILURE in 35m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/251e538a19b74a7ab47dc8bbdb70a662 : SUCCESS in 29m 20s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/deb21a66d4cd411f8897113db812cca9 : SUCCESS in 23m 58s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/c44b612a9cd549e999786c4f544061ee : SUCCESS in 29m 09s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/fd3db4a5439c4159915fa7a10963304a : SUCCESS in 27m 03s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/1db91c7b230c4492aeb40debe61f4d40 : SUCCESS in 33m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/495ab30a4b9745a2a9f9e4c65ed368b5 : SUCCESS in 36m 04s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/1eb4aee74cc44f97b6b4e1c59d7f4af7 : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5efa0f1d48f02e9896ff749dadf8f732f140a556","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-03-28 14:33:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06b845df704492c08be8110526f01bc6802af6d1","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-04-04 14:51:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"038d6dde2ac953e31c193f5dff3a9e51fff1a527","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 14:52:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"067217b082230501465b4d0f4be43b453451506a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 15:30:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bf3eccae5ed48cca57a8b0ddc6cefd2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2210705bf9024f7aba30da86d376e540 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12f15a31e33542f7b7d8dcd3d61f7507 : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ad177fe16ef45fc8625b155a456b006 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada90c2a2b9648bb98fe88be030e2f80 : SUCCESS in 9m 15s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/b4b333a903c24cd59a453e132730c49e : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ae429a96bd61ad59bc3d4fd68680da2a1694350e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 15:30:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"11791ca16ab290823b284786bcebb7d14ae20eee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 15:31:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee874eb6b02b4c7ab2a72d2d6752b1e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e75e6227fa35430f85df5075fa2579df : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff2b33b772ec4755bee336a8d7c88e653a6ab9d4","revisions":{"ff2b33b772ec4755bee336a8d7c88e653a6ab9d4":{"kind":"REWORK","_number":1,"created":"2025-03-26 18:01:22.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/43/945643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/43/945643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/43/945643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/43/945643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/43/945643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/43/945643/1"}}},"commit":{"parents":[{"commit":"da7482eab00828c7a953b59f30df719ff84748ab","subject":"Merge \"Add simple_crypto new_pkek subcommand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/da7482eab00828c7a953b59f30df719ff84748ab"}]}],"author":{"name":"Arefiev Anton","email":"aarefiev@mirantis.com","date":"2024-04-25 11:53:13.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-03-26 17:59:05.000000000","tz":0},"subject":"Raise HTTPUnauthorized in ContextMiddleware","message":"Raise HTTPUnauthorized in ContextMiddleware\n\nin this place exsception should be raised, not returned.\n\nChange-Id: I6f01608548defd5182137d7287009ec7ab1756ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ff2b33b772ec4755bee336a8d7c88e653a6ab9d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ff2b33b772ec4755bee336a8d7c88e653a6ab9d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
