)]}'
{"id":"openstack%2Fbarbican~947049","triplet_id":"openstack%2Fbarbican~master~Iac1a1e2149387a53d3b57add63aee8641026ce78","project":"openstack/barbican","branch":"master","topic":"bug/2064583","attention_set":{},"removed_from_attention_set":{"37876":{"account":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"last_update":"2026-05-15 13:30:56.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2026-05-15 13:30:56.000000000","reason":"Change was submitted"},"35125":{"account":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"last_update":"2026-05-15 12:54:26.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Iac1a1e2149387a53d3b57add63aee8641026ce78","subject":"Updated pagination logic to preserve URL filters","status":"MERGED","created":"2025-04-12 06:08:42.000000000","updated":"2026-05-15 13:33:22.000000000","submitted":"2026-05-15 13:30:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947049-bug/2064583","meta_rev_id":"4ba88a54a9295bb003ac5d68ced9fd8d34f6520d","_number":947049,"virtual_id_number":947049,"owner":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-15 13:30:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"all":[{"value":2,"date":"2026-05-15 12:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":2,"date":"2025-12-15 13:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-04 11:39:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"all":[{"value":1,"date":"2026-05-15 12:54:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-12 06:49:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-12 07:44:38.000000000","updated_by":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"},{"updated":"2025-04-14 12:00:54.000000000","updated_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2025-12-15 13:15:49.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2026-05-04 11:39:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"e4a7ed7a196c552121bb20c7c905f75958056fa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"date":"2025-04-12 06:08:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bad0a56bd02bb2ebdc6306ad2c75b2732841f92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 06:49:21.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/036ea9dc18e04c058bc56648a3c2113d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19ecdb13c0624072b6c3c92c08de52d4 : FAILURE in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fafd41d4c8c43efb843b692e3ee2265 : FAILURE in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f07ddf9bb804817a719f89065aa9887 : FAILURE in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1273b69bd2d643a6917aafab8319753e : FAILURE in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf1a0dbec24b4378bad1e6e7c905139f : FAILURE in 8m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b58b9df8a2db4717a6c69f98bef31673 : SUCCESS in 4m 28s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/bf39c38b7c8145bc9824ec1be0681ab1 : FAILURE in 15m 00s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/33ae90f0d63c4ff2bdd3b87f90511fce : FAILURE in 18m 39s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/d1d2e3ae901f426e88cfca6ac0383c2f : FAILURE in 33m 14s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3c8292594246490e9951064353905877 : FAILURE in 16m 32s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/6b41f5b0ed7c4c778a48758d18b73ef2 : FAILURE in 12m 02s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/f8d6a1c1fee841199044a2400b293bad : FAILURE in 30m 21s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/b2f1ad0d59fd4e379098260e1e86f293 : FAILURE in 20m 09s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/f41ed2f45de649e58e2cc875e0460150 : FAILURE in 22m 37s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/94c167bd53144b22af8174738a9bede4 : FAILURE in 29m 12s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/82036b436e0a40559ae5cc25dd56970f : FAILURE in 29m 03s","accounts_in_message":[],"_revision_number":1},{"id":"59e037912027d9e02ea73f76645596bc7725a348","author":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"date":"2025-04-12 07:44:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bfde77f4318d8723be2e1ee2d9d9886d8c5df74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"date":"2025-04-12 08:02:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3616549e4444f4d93ff1070022b08a64cf5cf9ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"date":"2025-04-12 08:14:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9f3d508f1d6cab0a7e0cc18f061fbd3b86d34428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 09:02:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d98fd16edbd54427b90f8f9115b3fa2a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01bffc52a15c4dfda9a0a1c1f05f752b : FAILURE in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d3a304addb49f886aa5ed0de9156f7 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd9165bbf6ae40c4ba5d532bf95bddfb : FAILURE in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6751e21f36434d3d897f519fdb094e68 : FAILURE in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ead5c3a9eb8d4451970267fa2b5beae5 : FAILURE in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a35f937c8f774425b087e8ffa8965a52 : SUCCESS in 7m 57s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/30aa9dd2cc634b1a98f1b0ca8e8ac5b7 : FAILURE in 31m 21s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/f22b764897e34e31941f0bd3751ee3da : FAILURE in 29m 38s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/9c2a3b7b5c684c998696d833c91ee87d : FAILURE in 44m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8a1c918f889f4aec85cc9d5d95595a89 : FAILURE in 31m 25s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/8eae493b27414a2ca288a0091d443163 : FAILURE in 27m 42s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/d19c3d634c9444bfbfa79a29bfa9e8ef : FAILURE in 32m 26s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/9dcf445cb8fb48c985464a5bd4d97922 : FAILURE in 21m 46s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/a0decf57bad443bdaa0b78822c6b6e8f : FAILURE in 27m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ad7c7f05e5f44faaf7c0f9d65fac83c : SUCCESS in 37m 04s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/ef647af19f114f8589aeb06858dac96c : SUCCESS in 36m 46s","accounts_in_message":[],"_revision_number":3},{"id":"dfcfae004567d859f468afa39dbba86a7ac2c948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"date":"2025-04-12 12:05:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"926dc7b62d6f5928b4568680aca5ae26a3b42a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 12:40:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6076285c827347c5ac9171671a1c7d44\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aba3874d11ae41439f07b5018407c326 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/716b6627064245c2845776b9bf2e5e06 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa868ce554e948258a9fd8fa0dc370c4 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06d5a44ad96040858270afd501dcb381 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/432d2d9970b143d8b335b865c66fa29c : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc59b3c797e4ae4923409ce7ad19a63 : SUCCESS in 6m 47s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/2f9fbe62cbdd4b529f6fdad3045a15fb : SUCCESS in 12m 03s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/2eacc7b6b5a64e49a50d62b687e3d5c2 : SUCCESS in 22m 25s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/adb2b4d4f8b54c259ca6ab08604ad964 : FAILURE in 20m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4bfe84385a46411384a3deba31da73be : SUCCESS in 21m 07s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/80d063b626bf4a819adcdbe1a4ad0935 : SUCCESS in 19m 21s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/125a312bc6084e5ea1c1124ae352f211 : SUCCESS in 21m 43s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/35f0e8f0e37546ca87cc9f2c0fca3d56 : SUCCESS in 20m 30s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/74e5142a03b742ffb1e58d9a3f865321 : SUCCESS in 26m 44s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e674dde872aa46e0bc0dcfd868388eec : SUCCESS in 22m 35s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/29420eb0033f45d3842b706657400baa : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":4},{"id":"3388c61dfbd518e0b686ebaf64855bd630b6252a","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-04-14 12:00:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0e38e1385c6318cf287bcda83fe95f308c285f78","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-04-14 12:01:04.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c0e71aa61497190f99962d8baaa2e0e9b7157790","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"date":"2025-04-14 16:00:57.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"54e7f0b75276948d84b1a0130aa6488c85e1e6b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 16:40:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c144ce9cb0d34422959eff8a5e5c3e22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22194a7d3fbe4677b45a01505c4b50d5 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/983ecaf7a1924948b21e96748e816166 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9c8acb3e3f44c38b7a434c3bb09cf02 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82082ca27e3d41cbb9cc25b277ad642b : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b66ef5a899a4203a47a20223f163340 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d993784664024fb688e74d9215898808 : SUCCESS in 4m 08s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/ee9d218645914f8a91fc88f3457849ab : SUCCESS in 13m 12s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/6d14c3e248314da8b02aa965655c9244 : SUCCESS in 13m 38s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/66125fcf36154f4f90cadf3abbbe8821 : FAILURE in 30m 06s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2c8b98ac7f784aa28885ebf1978cc5ef : SUCCESS in 14m 34s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/b93ce9a8afc34947af6e1183aa2e7279 : SUCCESS in 25m 25s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/4b755d6666334a85bdb0a11247d1007b : FAILURE in 28m 49s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/a5522305808d4958ab64b55ffdb6a787 : SUCCESS in 31m 47s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/b0441f45ae274997a0f04d3bd06c158e : SUCCESS in 16m 49s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5ecf0c5e3df24a47a39d62239f0334cd : SUCCESS in 37m 21s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/7252673e6129431c8d283ccb693070b3 : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":5},{"id":"0573d67f020905f39c87b8b7eb50a202ccfa3543","author":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"date":"2025-04-15 04:51:00.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0ded3cf1aace0b36fb2a9ad88615b3ba3604f1d8","author":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"date":"2025-04-15 05:16:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e965ba1ff777cc661b79219a2e492058ae427d4c","author":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"date":"2025-04-15 05:27:03.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"08abe6c0cd1a74f5c7380872a2bef30b2c2a6675","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"date":"2025-04-15 05:28:23.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0f2a168d1cf2770d7ce859dfd4a0894725b7ac96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 06:06:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9be2b3881d742309c7532e087a4900b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/891937a34a7d4445a64d5e8d977880fe : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b6d46baa66a4245b20c0bf7b94849ad : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c10642ebbe084706bc08010616f200cd : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a0aaa13d51484adbbef2ea8a4dc96d2a : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3a2359e199f4e50ace198c2e6d4aefd : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90457591e4444aa5bc71e7392f67fd36 : SUCCESS in 7m 48s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/06626e4113994a7ea2715998957a9c65 : SUCCESS in 24m 02s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/22f98badd06846408172405dcc6b58ec : SUCCESS in 30m 59s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/79f73cbcb8e2439f9f2c3475dce1cedf : FAILURE in 31m 08s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/44b6ce39d544439faa9e042987a566ab : SUCCESS in 32m 08s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/dabce1a28d824888a4505c654bd5a0bd : SUCCESS in 13m 55s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/64be376fcd414a6f83b744be7819dd67 : SUCCESS in 17m 36s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/aa1c93ead94a470897b61162a2840756 : SUCCESS in 15m 50s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/6a503e4d6b01435596a5ee044bf5d9bf : SUCCESS in 20m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c33aa05534046dfb67c727b5e216b19 : SUCCESS in 34m 33s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/cfc2abb0195546dfa74fa01091b49974 : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":6},{"id":"25bf45cf7d7f06b2b9f7a7e8961f25fe1b461613","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-12-15 13:15:49.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"a7104b9704f1a989c742ef829788ff57b2f6f43c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 11:39:31.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9210bba8c6bf26f187e416e9e43d9ce7ce356096","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2026-05-15 12:54:26.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nCorrectly preserves query parameters across paginated responses. Logic is sound - filters out limit/offset and re-appends the rest.","accounts_in_message":[],"_revision_number":6},{"id":"759fb46ea3cc01905806c81c3c8c687671c1c7d5","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2026-05-15 12:54:27.000000000","message":"Patch Set 6:\n\nCorrectly preserves query parameters across paginated responses. Logic is sound - filters out limit/offset and re-appends the rest.","accounts_in_message":[],"_revision_number":6},{"id":"36768603e9fb19a3acf9c88ef67822b953d51567","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2026-05-15 12:54:28.000000000","message":"Patch Set 6:\n\nCorrectly preserves query parameters across paginated responses. Logic is sound - filters out limit/offset and re-appends the rest.","accounts_in_message":[],"_revision_number":6},{"id":"39fcb97e0835c512d7d6997fecfe2b56000e4cec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 12:55:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"906b45cdd2d912b12eb86179c15bbaa3072f854a","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2026-05-15 12:55:49.000000000","message":"Patch Set 6:\n\nCorrectly preserves query parameters across paginated responses. Logic is sound - filters out limit/offset and re-appends the rest.","accounts_in_message":[],"_revision_number":6},{"id":"06c20f820eef4866ee32eb0177413fbb1ad27f23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 13:30:56.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04ee2726c9f24a40b835c0e70d13976d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ffdb0ed8ff84fe89de6728cc603424c : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8337c1ae984545c4abee7266e8b993fd : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8998d51645464d71bf24f63e4edd8d34 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f65fddbff3c44133aee978ba60e2e26b : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/091beeea7aac4fe68f9280f6c507de38 : SUCCESS in 5m 45s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/1e3bbd97947e4ebcae0a9383f0efcefe : SUCCESS in 27m 20s","accounts_in_message":[],"_revision_number":6},{"id":"c7ec8da54dd7ee1026e7fdfaec1fab0b87dbbb39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 13:30:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"4ba88a54a9295bb003ac5d68ced9fd8d34f6520d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 13:33:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/029daaa5602d4377bb3bc56bdf90cffd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a4c7e7dd27f45faa64cb686477e178e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"efe7040bca0819233a556638f3c112030f0f1400","revisions":{"113e174c1cffe54e3502d9f8f0217620ee4c2868":{"kind":"REWORK","_number":1,"created":"2025-04-12 06:08:42.000000000","uploader":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"ref":"refs/changes/49/947049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/49/947049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/49/947049/1"}}},"commit":{"parents":[{"commit":"62a1d6956f5ff335a8beee0d8a960391c8aac0f0","subject":"Merge \"Raise HTTPUnauthorized in ContextMiddleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/62a1d6956f5ff335a8beee0d8a960391c8aac0f0"}]}],"author":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-12 06:02:40.000000000","tz":60},"committer":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-12 06:04:33.000000000","tz":60},"subject":"Updated pagination logic to preserve URL filters","message":"Updated pagination logic to preserve URL filters\n\nAs reported on https://bugs.launchpad.net/barbican/+bug/2064583\nThe Barbican pagination logic does not preserve URL filters with example\n$ curl -s \u0027https://keymanager-XXX/v1/secrets?acl_only\u003dfalse\u0026secret_type\u003dpassphrase\u0027\n-H \"X-Auth-Token: ${OS_AUTH_TOKEN}\" | jq -r | grep next\n\"next\": \"https://keymanager-XXX:443/v1/secrets?limit\u003d10\u0026offset\u003d10\"\n\nThe issue was because the pagination logic only handled limit and offset\nleaving out other query_param which made it unavailable when next or\nprev is called.\n\nThe commit modifies secret retrieval on secrets,py and href helper\nfunctions on hrefs.py by:\n    - Extra any additional query parameters from the original request by\n\texcluding limit and offset already handled by next and prev\n    - Encode and append the filtered parameters to the pagination URLs\n\nThe committed changes will ensure that subsequent pages in the paginated\nresult maintained the original filtering criteria, providing a\nconsistent and expected API behaviour.\n\nCloses-Bug: #2064583\nChange-Id: Iac1a1e2149387a53d3b57add63aee8641026ce78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/113e174c1cffe54e3502d9f8f0217620ee4c2868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/113e174c1cffe54e3502d9f8f0217620ee4c2868"}]},"branch":"refs/heads/master"},"a5ce18dc29a7c703c15e9245d275ae038bc1a277":{"kind":"REWORK","_number":2,"created":"2025-04-12 08:02:52.000000000","uploader":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"ref":"refs/changes/49/947049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/49/947049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/49/947049/2"}}},"commit":{"parents":[{"commit":"62a1d6956f5ff335a8beee0d8a960391c8aac0f0","subject":"Merge \"Raise HTTPUnauthorized in ContextMiddleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/62a1d6956f5ff335a8beee0d8a960391c8aac0f0"}]}],"author":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-12 06:02:40.000000000","tz":60},"committer":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-12 08:00:18.000000000","tz":60},"subject":"Updated pagination logic to preserve URL filters","message":"Updated pagination logic to preserve URL filters\n\nAs reported on https://bugs.launchpad.net/barbican/+bug/2064583\nThe Barbican pagination logic does not preserve URL filters with example\n$ curl -s \u0027https://keymanager-XXX/v1/secrets?acl_only\u003dfalse\u0026secret_type\u003dpassphrase\u0027\n-H \"X-Auth-Token: ${OS_AUTH_TOKEN}\" | jq -r | grep next\n\"next\": \"https://keymanager-XXX:443/v1/secrets?limit\u003d10\u0026offset\u003d10\"\n\nThe issue was because the pagination logic only handled limit and offset\nleaving out other query_param which made it unavailable when next or\nprev is called.\n\nThe commit modifies secret retrieval on secrets,py and href helper\nfunctions on hrefs.py by:\n    - Extra any additional query parameters from the original request by\n\texcluding limit and offset already handled by next and prev\n    - Encode and append the filtered parameters to the pagination URLs\n\nThe committed changes will ensure that subsequent pages in the paginated\nresult maintained the original filtering criteria, providing a\nconsistent and expected API behaviour.\n\n- Fixed the pep8 issue.\n\nCloses-Bug: #2064583\nChange-Id: Iac1a1e2149387a53d3b57add63aee8641026ce78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/a5ce18dc29a7c703c15e9245d275ae038bc1a277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/a5ce18dc29a7c703c15e9245d275ae038bc1a277"}]},"branch":"refs/heads/master"},"c08bb0dd129164e05ea8111bf773a756c39841d8":{"kind":"REWORK","_number":3,"created":"2025-04-12 08:14:58.000000000","uploader":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"ref":"refs/changes/49/947049/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/49/947049/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/49/947049/3"}}},"commit":{"parents":[{"commit":"62a1d6956f5ff335a8beee0d8a960391c8aac0f0","subject":"Merge \"Raise HTTPUnauthorized in ContextMiddleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/62a1d6956f5ff335a8beee0d8a960391c8aac0f0"}]}],"author":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-12 06:02:40.000000000","tz":60},"committer":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-12 08:14:08.000000000","tz":60},"subject":"Updated pagination logic to preserve URL filters","message":"Updated pagination logic to preserve URL filters\n\nAs reported on https://bugs.launchpad.net/barbican/+bug/2064583\nThe Barbican pagination logic does not preserve URL filters with example\n$ curl -s \u0027https://keymanager-XXX/v1/secrets?acl_only\u003dfalse\u0026secret_type\u003dpassphrase\u0027\n-H \"X-Auth-Token: ${OS_AUTH_TOKEN}\" | jq -r | grep next\n\"next\": \"https://keymanager-XXX:443/v1/secrets?limit\u003d10\u0026offset\u003d10\"\n\nThe issue was because the pagination logic only handled limit and offset\nleaving out other query_param which made it unavailable when next or\nprev is called.\n\nThe commit modifies secret retrieval on secrets,py and href helper\nfunctions on hrefs.py by:\n    - Extra any additional query parameters from the original request by\n\texcluding limit and offset already handled by next and prev\n    - Encode and append the filtered parameters to the pagination URLs\n\nThe committed changes will ensure that subsequent pages in the paginated\nresult maintained the original filtering criteria, providing a\nconsistent and expected API behaviour.\n\n- Fixed the pep8 issue.\n\nCloses-Bug: #2064583\nChange-Id: Iac1a1e2149387a53d3b57add63aee8641026ce78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/c08bb0dd129164e05ea8111bf773a756c39841d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/c08bb0dd129164e05ea8111bf773a756c39841d8"}]},"branch":"refs/heads/master"},"ae6d3c97747f92bbd6ac74eb56d11af3629e3ecc":{"kind":"REWORK","_number":4,"created":"2025-04-12 12:05:16.000000000","uploader":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"ref":"refs/changes/49/947049/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/49/947049/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/49/947049/4"}}},"commit":{"parents":[{"commit":"62a1d6956f5ff335a8beee0d8a960391c8aac0f0","subject":"Merge \"Raise HTTPUnauthorized in ContextMiddleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/62a1d6956f5ff335a8beee0d8a960391c8aac0f0"}]}],"author":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-12 06:02:40.000000000","tz":60},"committer":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-12 12:04:35.000000000","tz":60},"subject":"Updated pagination logic to preserve URL filters","message":"Updated pagination logic to preserve URL filters\n\nAs reported on https://bugs.launchpad.net/barbican/+bug/2064583\nThe Barbican pagination logic does not preserve URL filters with example\n$ curl -s \u0027https://keymanager-XXX/v1/secrets?acl_only\u003dfalse\u0026secret_type\u003dpassphrase\u0027\n-H \"X-Auth-Token: ${OS_AUTH_TOKEN}\" | jq -r | grep next\n\"next\": \"https://keymanager-XXX:443/v1/secrets?limit\u003d10\u0026offset\u003d10\"\n\nThe issue was because the pagination logic only handled limit and offset\nleaving out other query_param which made it unavailable when next or\nprev is called.\n\nThe commit modifies secret retrieval on secrets,py and href helper\nfunctions on hrefs.py by:\n    - Extra any additional query parameters from the original request by\n\texcluding limit and offset already handled by next and prev\n    - Encode and append the filtered parameters to the pagination URLs\n\nThe committed changes will ensure that subsequent pages in the paginated\nresult maintained the original filtering criteria, providing a\nconsistent and expected API behaviour.\n\n- Fixed the pep8 issue.\n\nCloses-Bug: #2064583\nChange-Id: Iac1a1e2149387a53d3b57add63aee8641026ce78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ae6d3c97747f92bbd6ac74eb56d11af3629e3ecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ae6d3c97747f92bbd6ac74eb56d11af3629e3ecc"}]},"branch":"refs/heads/master"},"dfe27a2aa402ccd0dbab910278c1c8774376a57f":{"kind":"REWORK","_number":5,"created":"2025-04-14 16:00:57.000000000","uploader":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"ref":"refs/changes/49/947049/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/49/947049/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/49/947049/5"}}},"commit":{"parents":[{"commit":"62a1d6956f5ff335a8beee0d8a960391c8aac0f0","subject":"Merge \"Raise HTTPUnauthorized in ContextMiddleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/62a1d6956f5ff335a8beee0d8a960391c8aac0f0"}]}],"author":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-12 06:02:40.000000000","tz":60},"committer":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-14 15:59:50.000000000","tz":60},"subject":"Updated pagination logic to preserve URL filters","message":"Updated pagination logic to preserve URL filters\n\nAs reported on https://bugs.launchpad.net/barbican/+bug/2064583\nThe Barbican pagination logic does not preserve URL filters with example\n$ curl -s \u0027https://keymanager-XXX/v1/secrets?acl_only\u003dfalse\u0026secret_type\u003dpassphrase\u0027\n-H \"X-Auth-Token: ${OS_AUTH_TOKEN}\" | jq -r | grep next\n\"next\": \"https://keymanager-XXX:443/v1/secrets?limit\u003d10\u0026offset\u003d10\"\n\nThe issue was because the pagination logic only handled limit and offset\nleaving out other query_param which made it unavailable when next or\nprev is called.\n\nThe commit modifies secret retrieval on secrets,py and href helper\nfunctions on hrefs.py by:\n    - Extra any additional query parameters from the original request by\n\texcluding limit and offset already handled by next and prev\n    - Encode and append the filtered parameters to the pagination URLs\n\nThe committed changes will ensure that subsequent pages in the paginated\nresult maintained the original filtering criteria, providing a\nconsistent and expected API behaviour.\n\n- Fixed the pep8 issue.\n\nCloses-Bug: #2064583\nChange-Id: Iac1a1e2149387a53d3b57add63aee8641026ce78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/dfe27a2aa402ccd0dbab910278c1c8774376a57f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/dfe27a2aa402ccd0dbab910278c1c8774376a57f"}]},"branch":"refs/heads/master"},"efe7040bca0819233a556638f3c112030f0f1400":{"kind":"NO_CHANGE","_number":6,"created":"2025-04-15 05:28:23.000000000","uploader":{"_account_id":37876,"name":"Theresa James","display_name":"Tessy James","email":"tessyjames28@gmail.com","username":"TessyJames"},"ref":"refs/changes/49/947049/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/49/947049/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/49/947049/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/49/947049/6"}}},"commit":{"parents":[{"commit":"62a1d6956f5ff335a8beee0d8a960391c8aac0f0","subject":"Merge \"Raise HTTPUnauthorized in ContextMiddleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/62a1d6956f5ff335a8beee0d8a960391c8aac0f0"}]}],"author":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-12 06:02:40.000000000","tz":60},"committer":{"name":"Theresa James","email":"tessyjames28@gmail.com","date":"2025-04-15 05:27:51.000000000","tz":60},"subject":"Updated pagination logic to preserve URL filters","message":"Updated pagination logic to preserve URL filters\n\nAs reported on https://bugs.launchpad.net/barbican/+bug/2064583\nThe Barbican pagination logic does not preserve URL filters with example\n$ curl -s \u0027https://keymanager-XXX/v1/secrets?acl_only\u003dfalse\u0026secret_type\u003dpassphrase\u0027\n-H \"X-Auth-Token: ${OS_AUTH_TOKEN}\" | jq -r | grep next\n\"next\": \"https://keymanager-XXX:443/v1/secrets?limit\u003d10\u0026offset\u003d10\"\n\nThe issue was because the pagination logic only handled limit and offset\nleaving out other query_param which made it unavailable when next or\nprev is called.\n\nThe commit modifies secret retrieval on secrets,py and href helper\nfunctions on hrefs.py by:\n    - Extra any additional query parameters from the original request by\n\texcluding limit and offset already handled by next and prev\n    - Encode and append the filtered parameters to the pagination URLs\n\nThe committed changes will ensure that subsequent pages in the paginated\nresult maintained the original filtering criteria, providing a\nconsistent and expected API behaviour.\n\n- Fixed the pep8 issue.\n\nCloses-Bug: #2064583\nChange-Id: Iac1a1e2149387a53d3b57add63aee8641026ce78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/efe7040bca0819233a556638f3c112030f0f1400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/efe7040bca0819233a556638f3c112030f0f1400"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
