)]}'
{"id":"openstack%2Fbarbican~956553","triplet_id":"openstack%2Fbarbican~master~I74ea6fa954f3c3e313eb0afed487a5a4e2dd7eb2","project":"openstack/barbican","branch":"master","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2025-09-10 08:48:43.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-19 14:33:48.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2025-09-19 14:09:07.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"I74ea6fa954f3c3e313eb0afed487a5a4e2dd7eb2","subject":"Use static KEK in devstack","status":"MERGED","created":"2025-08-05 12:21:29.000000000","updated":"2025-09-19 14:34:39.000000000","submitted":"2025-09-19 14:33:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956553","meta_rev_id":"04a6b60c3d034b6b8cac2d520cb15250efa23aaf","_number":956553,"virtual_id_number":956553,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-19 14:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2025-09-10 08:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2025-09-15 15:13:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":2,"date":"2025-09-19 14:09:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-19 14:13:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":1,"date":"2025-09-19 14:09:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-05 13:13:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-12 15:13:08.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2025-08-12 15:13:08.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-09-15 15:13:08.000000000","updated_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"},{"updated":"2025-09-19 14:13:20.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"4140dea90421ac2474a2b290671ffccca3093955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-05 12:21:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"508a6296f9f50367a726dc8f30cf5d05c3c4c060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 13:13:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170057d9d36245f7b9820e5d0a2e9198\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35c238829765452b9b28e63e57d18fe6 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffd4f38762f74d319f3df376f5127d44 : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b1ae7653b624d51ba3b59e61c94f449 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd91840c2f3240189e73d5a752e0c7de : SUCCESS in 6m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12541f1d47034019b02643fff27dd19a : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38315d98abf3422397d8423583ae7841 : SUCCESS in 7m 35s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/aad7d4d2aa514c3385c7097f38e76c47 : SUCCESS in 30m 57s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/3c1dbe5e5a41486d869d1289a16f22b6 : SUCCESS in 30m 54s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/6ff1fa95e07b44eeb3708fbca429d2d1 : FAILURE in 46m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/673df445bca44755ba72083590c4981a : SUCCESS in 31m 03s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/d84a6d35f60042f1842bfdeae9e9164f : SUCCESS in 13m 53s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/6710e96f88df4bf8ada83c1d044ea059 : SUCCESS in 33m 01s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/85db20eeda734c55a8cc0c2ec92befa0 : SUCCESS in 32m 17s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/753de04df9b34ad4bcf8d7b7ff1b728e : POST_FAILURE in 15m 29s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/62ee6fce2b924b279131e5e9b33d4777 : SUCCESS in 35m 31s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/f7d619e3e56147c29c78af561c938583 : SUCCESS in 35m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ee4f6eece7652748df8ed457530d919d35b81dd3","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-09-10 08:48:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"812e4246baf91ef3df9d14415efa3d0b3602017b","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-09-15 15:13:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dd06688654c31f3dc4a9f6050e528383bf9edf81","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-09-19 14:09:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22ea3fb661cc0f7bf9ec7891ab7a85ebe0a0987f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 14:09:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"533f1437ef42379ade69799eb6a9921446b4d511","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-19 14:13:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ecb244a34551bd9516e6cbdab152e2f654f0dd28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 14:33:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83bc577a435145fc87234de321fd4d1b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/615593b53d52448ba7476789a926bb7b : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c64c1c106de42688218e2813dba6e70 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d8980d2130746df91008ef06122c4fe : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98977aafc7724f9383c2c7daa1218208 : SUCCESS in 4m 20s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/a92a5bc0459245078d681e7f122a617d : SUCCESS in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e1e1a7d9b39e8ebea65e54167fbaf151d0f83414","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 14:33:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"04a6b60c3d034b6b8cac2d520cb15250efa23aaf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 14:34:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c10eebf880074992940f0f263dabd217\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16ab7a56f0cd4241886033b7f468cabf : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07244ffaa8b258b1955da12d7db0b8e24900e52d","revisions":{"07244ffaa8b258b1955da12d7db0b8e24900e52d":{"kind":"REWORK","_number":1,"created":"2025-08-05 12:21:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/956553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/53/956553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/53/956553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/53/956553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/53/956553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/53/956553/1"}}},"commit":{"parents":[{"commit":"bfd1e72bf070a6a169882ad0abb88687f443a7da","subject":"Exclude barbican/wsgi/api.py from coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bfd1e72bf070a6a169882ad0abb88687f443a7da"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-05 12:20:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-05 12:21:17.000000000","tz":540},"subject":"Use static KEK in devstack","message":"Use static KEK in devstack\n\nCommit 5250c5f368204083fd36bbed5273b61ebe9e6de4 changed how kek is\ngenerated in devstack, but it broke grenade which requires that\nconsistent kek is used before/after upgrade.\n\nChange-Id: I74ea6fa954f3c3e313eb0afed487a5a4e2dd7eb2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/07244ffaa8b258b1955da12d7db0b8e24900e52d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/07244ffaa8b258b1955da12d7db0b8e24900e52d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
