)]}'
{"id":"openstack%2Fbarbican~978352","triplet_id":"openstack%2Fbarbican~master~Idec7f6bfda49b2d841519536375fd3871da52e1f","project":"openstack/barbican","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-29 15:29:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idec7f6bfda49b2d841519536375fd3871da52e1f","subject":"Do not override host_ref by default","status":"MERGED","created":"2026-03-01 14:40:27.000000000","updated":"2026-05-29 15:30:12.000000000","submitted":"2026-05-29 15:29:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978352","meta_rev_id":"028b5c3307d5efa095580b8a0674aea7cbce1067","_number":978352,"virtual_id_number":978352,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-29 15:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"all":[{"value":2,"date":"2026-05-26 10:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":2,"date":"2026-05-29 14:51:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":1,"date":"2026-05-29 14:51:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-01 15:50:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-26 10:08:19.000000000","updated_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"},{"updated":"2026-05-29 14:51:00.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"}],"messages":[{"id":"ce767963030c300553a2d93ae1cf16cba7faa329","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 14:40:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f70fcc90cfb40df7899c8c1ccd709f86db74433","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-01 14:45:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02acc8b0d38891315caf8326f217e6be3c88ae7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 15:50:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7192e9a8ff0144df9ab90faa2846ad3c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d270b209c51345ab9970bfbb0a9fc85e : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d70e1f0af42448f87493459785aaba5 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7a7ff7d7a3e4b9c8b38948c59cb98bb : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e38dbf6593a46328e59c663a79c478a : SUCCESS in 5m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e247858e66514ebab2797857b4c65694 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90371d07a5da4839bd1e9303a452dafb : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe702fd321cc4f17b539940f69394ae7 : SUCCESS in 4m 08s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/1d6e4fb7737d41e5b70e561a333f8349 : FAILURE in 30m 00s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/f9b6dcf88dba4e5d9aa9e1fb2af209cb : FAILURE in 32m 52s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/245a0664b3844879ba9824bfaf0716d8 : FAILURE in 48m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/cfb0b6bb0dee4f92b61e312c354a1a58 : SUCCESS in 37m 12s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/585e3fd6a66a4dc3b942b6f8ed905d68 : FAILURE in 10m 50s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/c5c7b77c4a1a4a0ba08311728eda8ed9 : SUCCESS in 34m 04s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/ccbf3348083c4467a77842f6fdbe3650 : SUCCESS in 32m 06s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/c8e48823f6b944ca8769ae25a17e8835 : POST_FAILURE in 8m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1be756d13adc44508953f85c107fb15c : SUCCESS in 1h 02m 45s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/7dba5013ec1245ec87fcc8e3885604ff : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":2},{"id":"973675300c50633b4570d5f7ab3f0916b92aa3fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-02 14:54:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"54274baf20a0eeb85cc67c1e37866f589c99c29f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:14:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ba8820a3870481da45c8fc7ec69940d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93d545c40e29450ea35b59129555154d : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/591acd0705934ea5bd97c9aa5f7b83ef : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/722035052329484da8a2ac98c0d6e19e : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b23abb88e5b445c6a04fb32e2d74858b : SUCCESS in 5m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d14ba969b2624434acef9f0d19c3be29 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b21fb3de13b4e49b977937d145de826 : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27331f433e124fba965b58db77c11ed8 : SUCCESS in 6m 01s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/1fa2f9b3b7a54973bc00d9d5509962c5 : SUCCESS in 17m 28s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/6ee81da9934241dcb06ba88386491bd9 : SUCCESS in 32m 31s (non-voting)\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/8500ba3c5daf40639d4440ff144c0eef : FAILURE in 44m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/923ccc16ff14493f9188fd046629e747 : SUCCESS in 35m 27s\n- barbican-tempest-plugin-simple-crypto-jammy https://zuul.opendev.org/t/openstack/build/c9cb4cf11a4042cc85a9d7d73c7eebde : FAILURE in 10m 39s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/70c7dc7d5cf44d68acec63bc708e5ce2 : SUCCESS in 34m 52s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/76d6dd37dc9140bf82563f07000c8531 : SUCCESS in 36m 39s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/7c522a3644024d5b95dcaba8c0926843 : POST_FAILURE in 20m 37s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7df16aca6f7b43a5a419a3aec5bb483a : SUCCESS in 38m 43s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/84d867e894014d8e88f543252a590a11 : SUCCESS in 35m 48s","accounts_in_message":[],"_revision_number":3},{"id":"047d5434c48118df0085da42f8927e959dbc54f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-17 02:38:24.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"b21e3e19aa8a8393ff69ad45533b279edae1f94c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-17 02:45:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"58388deae90a524e0737f1fd04ac22458f4cd2af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-17 03:36:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1118fda33928ee95fe106008b432c1ab8d9335c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-17 03:38:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"142ad86a6f4ade99e62f977336281ec9e9ef455f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 04:21:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fe894679af54a53b76c7fc6af276c82\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f595817fcf248ffb9618b9895ba3d09 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9db1d0a6b2e483ca8b475ccf3470e5b : SUCCESS in 2m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8206780cabcb4845ab5072f231435f20 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/baa9aeca1a864f20a246c0e2bc2864f3 : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ea3e334d3f8c4f9d94688bf33b551e79 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/527f6031604e4f2a966ab29857e0f611 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cade839777c044028427a1472aac42db : SUCCESS in 3m 15s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/1721a63c264645558568a089c6bb7e68 : SUCCESS in 17m 06s\n- barbican-vault-tox-functional https://zuul.opendev.org/t/openstack/build/8ead952e433e43d3ad443d63de80c953 : SUCCESS in 17m 15s\n- barbican-openbao-tox-functional https://zuul.opendev.org/t/openstack/build/b192bd952fd5470b8c648715192d1e32 : SUCCESS in 31m 17s\n- barbican-grenade https://zuul.opendev.org/t/openstack/build/775de567071c44c286e4423a53670091 : SUCCESS in 41m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fb1c89d7bd2d4294b729d253d47b2883 : SUCCESS in 32m 02s\n- barbican-tempest-plugin-simple-crypto-secure-rbac https://zuul.opendev.org/t/openstack/build/7c2ca4fd58e44ca09105c8f83b4e9d7c : SUCCESS in 25m 36s\n- barbican-tempest-plugin-simple-crypto-ipv6-only https://zuul.opendev.org/t/openstack/build/78e19faf1a7047ba8be761d8297793f3 : SUCCESS in 14m 33s\n- barbican-tox-functional-fips https://zuul.opendev.org/t/openstack/build/e437aa582dd84eedb8ebb070c825a680 : RETRY_LIMIT in 1m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/340cf7e84136418d8d5792edb36b8b45 : SUCCESS in 34m 42s\n- octavia-v2-dsvm-tls-barbican-secure-rbac https://zuul.opendev.org/t/openstack/build/a5357d12a29741edae343e6650724390 : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":7},{"id":"30c02e422076d26e1f62affbb0b8b05d4b197167","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2026-05-26 10:08:19.000000000","message":"Patch Set 7: Code-Review+2\n\nPS7 is clean and CI passes. Removing the hardcoded localhost default is long overdue — the OptionalURI type with deprecation warning for empty string is the right approach. LGTM.","accounts_in_message":[],"_revision_number":7},{"id":"1b61874dad227084c2892fad25401295fc959e62","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2026-05-29 14:51:00.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"120426e450623950b73e57ccf8e1f2270a2976a6","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2026-05-29 14:51:11.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"6150e2b81c51e5e072eef63fd43597d166545a5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 14:51:59.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"eb442bdcd3cc46b31b29c51a84032a73190e778f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:29:09.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c0b50b2b89e4e7b9591a82d4914ea43\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c53d061e19c467f89021458a8e15e90 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ecd9b8d297b4c5ba3eb595177b17d96 : SUCCESS in 6m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/41b02c76abc7422a9a6783720e87c985 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59367ef3839d4788a749c1be9557d000 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e671b1044689401c9bbf28ad21a6bec9 : SUCCESS in 5m 13s\n- barbican-tox-functional https://zuul.opendev.org/t/openstack/build/1995e06abb9e49d19a0875a459e36030 : SUCCESS in 32m 13s","accounts_in_message":[],"_revision_number":7},{"id":"e2767cb411eec86e196b850067443c85df0bf363","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:29:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"028b5c3307d5efa095580b8a0674aea7cbce1067","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:30:12.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc52fb25933946fb8b3f37dd9e0ee9dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/527f83517d7245848c8f8302c69c005a : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fe831fbe37344b3bc27616ec8321127 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"70b20d3f96167a7a19064a28ce71d78764a1d1c3","revisions":{"51c2bcdab504a2c5057b174909ce4c49b7c06137":{"kind":"REWORK","_number":1,"created":"2026-03-01 14:40:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/978352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/52/978352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/52/978352/1"}}},"commit":{"parents":[{"commit":"bbc7d50b3640d13d0ae34d04135066703d4fb62f","subject":"Drop unused sphinxcontrib-blockdiag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bbc7d50b3640d13d0ae34d04135066703d4fb62f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 13:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 14:15:07.000000000","tz":540},"subject":"Do not override host_ref by default","message":"Do not override host_ref by default\n\nThe current default value is not at all suitable for real deployments.\nThis default value disables detection of actual endpoint url according\nrequest environments.\n\nChange-Id: Idec7f6bfda49b2d841519536375fd3871da52e1f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/51c2bcdab504a2c5057b174909ce4c49b7c06137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/51c2bcdab504a2c5057b174909ce4c49b7c06137"}]},"branch":"refs/heads/master"},"18ecb036538e3920d264c89df381740e5c09ad36":{"kind":"REWORK","_number":2,"created":"2026-03-01 14:45:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/978352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/52/978352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/52/978352/2"}}},"commit":{"parents":[{"commit":"bbc7d50b3640d13d0ae34d04135066703d4fb62f","subject":"Drop unused sphinxcontrib-blockdiag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bbc7d50b3640d13d0ae34d04135066703d4fb62f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 13:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 14:45:00.000000000","tz":540},"subject":"Do not override host_ref by default","message":"Do not override host_ref by default\n\nThe current default value is not at all suitable for real deployments.\nThis default value disables detection of actual endpoint url according\nrequest environments.\n\nChange-Id: Idec7f6bfda49b2d841519536375fd3871da52e1f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/18ecb036538e3920d264c89df381740e5c09ad36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/18ecb036538e3920d264c89df381740e5c09ad36"}]},"branch":"refs/heads/master"},"c06dfa75713a89035be6cda22f2513f73b43db02":{"kind":"REWORK","_number":3,"created":"2026-03-02 14:54:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/978352/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/52/978352/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/52/978352/3"}}},"commit":{"parents":[{"commit":"bbc7d50b3640d13d0ae34d04135066703d4fb62f","subject":"Drop unused sphinxcontrib-blockdiag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/bbc7d50b3640d13d0ae34d04135066703d4fb62f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 13:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-02 14:53:20.000000000","tz":540},"subject":"Do not override host_ref by default","message":"Do not override host_ref by default\n\nThe current default value is not at all suitable for real deployments.\nThis default value disables detection of actual endpoint url according\nrequest environments.\n\nChange-Id: Idec7f6bfda49b2d841519536375fd3871da52e1f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/c06dfa75713a89035be6cda22f2513f73b43db02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/c06dfa75713a89035be6cda22f2513f73b43db02"}]},"branch":"refs/heads/master"},"53a02d5f69e02e7ed3b82809f8bcc48551bc3330":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-05-17 02:38:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/978352/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/52/978352/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/52/978352/4"}}},"commit":{"parents":[{"commit":"2d1ea8606784cb7e1147e869a79f6be7e8466236","subject":"Merge \"Don\u0027t allow deleting secrets with consumers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/2d1ea8606784cb7e1147e869a79f6be7e8466236"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 13:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-17 02:38:24.000000000","tz":0},"subject":"Do not override host_ref by default","message":"Do not override host_ref by default\n\nThe current default value is not at all suitable for real deployments.\nThis default value disables detection of actual endpoint url according\nrequest environments.\n\nChange-Id: Idec7f6bfda49b2d841519536375fd3871da52e1f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/53a02d5f69e02e7ed3b82809f8bcc48551bc3330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/53a02d5f69e02e7ed3b82809f8bcc48551bc3330"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"c06dfa75713a89035be6cda22f2513f73b43db02","theirs":"2d1ea8606784cb7e1147e869a79f6be7e8466236","contains_conflicts":false}},"fcdd2c2f60ffa93ee5e4532d27e39f468ec189c0":{"kind":"REWORK","_number":5,"created":"2026-05-17 02:45:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/978352/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/52/978352/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/52/978352/5"}}},"commit":{"parents":[{"commit":"ceb9b85d05514eaf2952944171ee51630e5256d1","subject":"Merge \"sqlalchemy: Use built-in declarative\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ceb9b85d05514eaf2952944171ee51630e5256d1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 13:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-17 11:44:07.000000000","tz":540},"subject":"Do not override host_ref by default","message":"Do not override host_ref by default\n\nThe current default value is not at all suitable for real deployments.\nThis default value disables detection of actual endpoint url according\nrequest environments.\n\nChange-Id: Idec7f6bfda49b2d841519536375fd3871da52e1f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/fcdd2c2f60ffa93ee5e4532d27e39f468ec189c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/fcdd2c2f60ffa93ee5e4532d27e39f468ec189c0"}]},"branch":"refs/heads/master"},"af5f042d0200c16858d12f0309f9bff1a097965f":{"kind":"REWORK","_number":6,"created":"2026-05-17 03:36:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/978352/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/52/978352/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/52/978352/6"}}},"commit":{"parents":[{"commit":"ceb9b85d05514eaf2952944171ee51630e5256d1","subject":"Merge \"sqlalchemy: Use built-in declarative\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ceb9b85d05514eaf2952944171ee51630e5256d1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 13:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-17 12:30:54.000000000","tz":540},"subject":"Do not override host_ref by default","message":"Do not override host_ref by default\n\nThe current default value is not at all suitable for real deployments.\nThis default value disables detection of actual endpoint url according\nrequest environments.\n\nChange-Id: Idec7f6bfda49b2d841519536375fd3871da52e1f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/af5f042d0200c16858d12f0309f9bff1a097965f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/af5f042d0200c16858d12f0309f9bff1a097965f"}]},"branch":"refs/heads/master"},"70b20d3f96167a7a19064a28ce71d78764a1d1c3":{"kind":"REWORK","_number":7,"created":"2026-05-17 03:38:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/978352/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/barbican","ref":"refs/changes/52/978352/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/barbican refs/changes/52/978352/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/barbican refs/changes/52/978352/7"}}},"commit":{"parents":[{"commit":"ceb9b85d05514eaf2952944171ee51630e5256d1","subject":"Merge \"sqlalchemy: Use built-in declarative\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/ceb9b85d05514eaf2952944171ee51630e5256d1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-01 13:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-17 12:32:42.000000000","tz":540},"subject":"Do not override host_ref by default","message":"Do not override host_ref by default\n\nThe current default value is not at all suitable for real deployments.\nThis default value disables detection of actual endpoint url according\nrequest environments.\n\nChange-Id: Idec7f6bfda49b2d841519536375fd3871da52e1f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/70b20d3f96167a7a19064a28ce71d78764a1d1c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/barbican/commit/70b20d3f96167a7a19064a28ce71d78764a1d1c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
