)]}'
{"id":"openstack%2Fbashate~729134","triplet_id":"openstack%2Fbashate~master~Ic97beaec35ff633380688beb93318273ab6ca15c","project":"openstack/bashate","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"Ic97beaec35ff633380688beb93318273ab6ca15c","subject":"Switch to newer openstackdocstheme and reno versions","status":"MERGED","created":"2020-05-19 06:39:26.000000000","updated":"2020-06-02 12:42:17.000000000","submitted":"2020-06-02 12:40:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729134-1591101643750-94f34792","meta_rev_id":"cedcc62aea8f18090a28e90d5fcef14e2abddbd2","_number":729134,"virtual_id_number":729134,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-02 12:40:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":2,"date":"2020-06-02 12:31:09.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2020-05-22 17:33:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2020-05-23 08:15:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":1,"date":"2020-06-02 12:31:09.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 17:33:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-23 08:15:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-02 12:31:09.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-06-02 12:40:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d193933689f9097dab49b683073ef2636698dbf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-19 06:39:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e15e6dee0064eccb7cb2a8accfa728966b62157","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-19 06:40:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e06d66f82f6670e5ea05e2076729105ae63fddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 06:55:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b1e7133e62af469e927cfdc8c81400c0 : FAILURE in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31563ada7d1417e980574b43a042cb8 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ffb70964802411aac558347b0702574 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3307682eb165465e809b282fd2e56ca6 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71022f5e233e42479edc1f0e166a8800 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/351e58e69bfc48db89a66f6fd61003ac : SUCCESS in 7m 29s\n- bashate-devstack https://zuul.opendev.org/t/openstack/build/4fdc6818000e450e84b60a9a6b01cc00 : SUCCESS in 3m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2ffdf5e1798c69a3d75d920c96355f6056df63c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-19 07:06:08.000000000","message":"Patch Set 1:\n\npy35 drama here too","accounts_in_message":[],"_revision_number":1},{"id":"1c885fb2c5b2a51355f90761c93bb77a2f57212f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-19 08:21:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cc13e33627949068f640743418f118d89ffd3b7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 08:27:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bd71de18557f423db6fadca475b5261f : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf7a8db28f24bea8c001507038f0fa0 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c9afd1ab06440fd8f4b5056389ba46e : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff8b05b5c00b4357b184a7e6987dc784 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea47b9e61a924268a662ca6efa1f6c8c : FAILURE in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24093e75ca4f468e9e9b4bbf825bdbc7 : FAILURE in 3m 39s\n- bashate-devstack https://zuul.opendev.org/t/openstack/build/d7cca0df6f5c48f4aefd34c8b5566305 : SUCCESS in 3m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0e1a3968c6e90d1a5e74dd12afe95acc7bfe2f91","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-19 08:30:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9b7123ad6200d76ee78e86709b87103b47646fe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 08:39:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9655fd64b36c4d36bbb9a9c033ef7103 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac9d8fd5c40462a8e4ccc0948d92ec0 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9d01047328848a781cb45f34d2350d4 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/821a4dcf828b4afbbb71e39fa76666f0 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5a3222d97e44dc89321333d7a335d9c : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef300a9eeb8e4d24bd4f6409df7cabc6 : SUCCESS in 3m 24s\n- bashate-devstack https://zuul.opendev.org/t/openstack/build/0ee7ab0ac09c4d9982bf43bfefaa2d24 : SUCCESS in 2m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba04c668e34a2cbfe627ae47279e4d7b165bd59d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-19 12:26:23.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a039055273d1cd957684174b969b5d5ac3f5407b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-21 07:20:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"623df57a2ca2d8a5ec9e1b3e88c5cc3785a68cce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 07:30:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/db08788a85d94a228c3b797610b110b9 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb2e2bcd9ee5460b81fea3a0eeda954c : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc2f269644ba40ec93b813926b36851d : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/110bbfc92f004e43a50c4f61a2b4c471 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9167669be6ca4dc6ba5dcb57f4dc4f60 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b9a75f182b741e38766f505cfc8a28d : SUCCESS in 5m 08s\n- bashate-devstack https://zuul.opendev.org/t/openstack/build/a845a2ea1a964830aed4587f28067f48 : SUCCESS in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9b97d2eaa96d6f91ea9d7bd97aecc9809a803e33","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-22 17:20:44.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"07865ab71068f2142d06b2e4149948371fb92cc6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-22 17:31:28.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"edd28d84c9fbaff1b915034f636458fc248066f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-22 17:33:24.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"39c92641041e4b9425b9809ba34495790fe4d890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 17:42:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f2331e251e3242fa9096a1d78378a7f5 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5fccb0fd449449a8b158db163c0f0ba : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1256aef44481470d9f6895536104b83b : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92d9bb2c00b048b6beff4dee595266a0 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed01e0fa29e24163a033bf3f83b28ca4 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e06634890ac14770ae7a4fd499b20a0d : SUCCESS in 4m 01s\n- bashate-devstack https://zuul.opendev.org/t/openstack/build/6c80a9dde7464992b9b3313552d9c6bb : SUCCESS in 2m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e28d8989d285030fe9bcafedf4a89015a2d7bedd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-23 08:15:26.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e46aaf4bb7a7077378ac84bbb3d9ffcc767d5fa8","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-06-02 12:31:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b346b314d5013f1a538883e2120835bdc3288d01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 12:31:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b7b487fbdd990bc1134f79a38d1e989dd553264a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 12:40:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6e5bac56407d41f5b1e55f5d9d829141 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4aa32e1f7a148fb8f1581ee94eac6d0 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9978a3c9b6994d33a9e922b865f45abd : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae1a9d2dbc514dbe899a7b1833b75a59 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cadf2aae94242f89aa5048af353460d : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd3dcaf7291e406d924b97c7de2d2f2f : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":5},{"id":"1010d82d3ba83c5bbcaa9ef1fca5f46d39453866","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 12:40:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"12d5f5624497ec775110ae7a72b499eb86a5d8de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 12:42:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af123976ad6f4bd7af43c21c6589d45c : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acfdf981444440a2af2aaea50f16d775 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6a3769e0770d5c71622652e7a61898745cf94510","revisions":{"adc73110e46c345bfef92f8e04be7c34939559e0":{"kind":"REWORK","_number":1,"created":"2020-05-19 06:39:26.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/34/729134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bashate","ref":"refs/changes/34/729134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bashate refs/changes/34/729134/1"}}},"commit":{"parents":[{"commit":"e57a429706e75b4c8cdc9456a7b76ba40549066d","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/e57a429706e75b4c8cdc9456a7b76ba40549066d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-19 06:37:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-19 06:39:08.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.0 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nChange-Id: Ic97beaec35ff633380688beb93318273ab6ca15c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/adc73110e46c345bfef92f8e04be7c34939559e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/adc73110e46c345bfef92f8e04be7c34939559e0"}]},"branch":"refs/heads/master"},"821fb3ea4d847280f2304146e7c07be9d3904535":{"kind":"REWORK","_number":2,"created":"2020-05-19 08:21:52.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/34/729134/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bashate","ref":"refs/changes/34/729134/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bashate refs/changes/34/729134/2"}}},"commit":{"parents":[{"commit":"e57a429706e75b4c8cdc9456a7b76ba40549066d","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/e57a429706e75b4c8cdc9456a7b76ba40549066d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-19 06:37:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-19 08:21:35.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.0 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nCreate separate doc/requirements.txt file and use it, this allows\nto continue testing with py35 and build docs that require py36.\n\nChange-Id: Ic97beaec35ff633380688beb93318273ab6ca15c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/821fb3ea4d847280f2304146e7c07be9d3904535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/821fb3ea4d847280f2304146e7c07be9d3904535"}]},"branch":"refs/heads/master"},"e21554f69696e3616860e80a03ea27e18a525e5d":{"kind":"REWORK","_number":3,"created":"2020-05-19 08:30:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/34/729134/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bashate","ref":"refs/changes/34/729134/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bashate refs/changes/34/729134/3"}}},"commit":{"parents":[{"commit":"e57a429706e75b4c8cdc9456a7b76ba40549066d","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/e57a429706e75b4c8cdc9456a7b76ba40549066d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-19 06:37:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-19 08:30:17.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.0 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nCreate separate doc/requirements.txt file and use it, this allows\nto continue testing with py35 and build docs that require py36.\n\nChange-Id: Ic97beaec35ff633380688beb93318273ab6ca15c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/e21554f69696e3616860e80a03ea27e18a525e5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/e21554f69696e3616860e80a03ea27e18a525e5d"}]},"branch":"refs/heads/master"},"0968689f792e6eaff42ea4d83f1b8c50328ae8b6":{"kind":"REWORK","_number":4,"created":"2020-05-21 07:20:10.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/34/729134/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bashate","ref":"refs/changes/34/729134/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bashate refs/changes/34/729134/4"}}},"commit":{"parents":[{"commit":"e57a429706e75b4c8cdc9456a7b76ba40549066d","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/e57a429706e75b4c8cdc9456a7b76ba40549066d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-19 06:37:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-21 07:20:05.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.0 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nCreate separate doc/requirements.txt file and use it, this allows\nto continue testing with py35 and build docs that require py36.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: Ic97beaec35ff633380688beb93318273ab6ca15c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/0968689f792e6eaff42ea4d83f1b8c50328ae8b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/0968689f792e6eaff42ea4d83f1b8c50328ae8b6"}]},"branch":"refs/heads/master"},"6a3769e0770d5c71622652e7a61898745cf94510":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-05-22 17:31:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/34/729134/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bashate","ref":"refs/changes/34/729134/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bashate refs/changes/34/729134/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bashate refs/changes/34/729134/5"}}},"commit":{"parents":[{"commit":"e57a429706e75b4c8cdc9456a7b76ba40549066d","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/e57a429706e75b4c8cdc9456a7b76ba40549066d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-19 06:37:44.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-05-22 17:31:24.000000000","tz":0},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nCreate separate doc/requirements.txt file and use it, this allows\nto continue testing with py35 and build docs that require py36.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: Ic97beaec35ff633380688beb93318273ab6ca15c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/6a3769e0770d5c71622652e7a61898745cf94510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bashate/commit/6a3769e0770d5c71622652e7a61898745cf94510"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
