)]}'
{"id":"openstack%2Fbifrost~665851","triplet_id":"openstack%2Fbifrost~master~Iebca5c4cc8dfdecf7c561b3612691b62ca4ac0c8","project":"openstack/bifrost","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"Iebca5c4cc8dfdecf7c561b3612691b62ca4ac0c8","subject":"Build pdf doc","status":"MERGED","created":"2019-06-18 02:57:22.000000000","updated":"2019-09-06 10:44:45.000000000","submitted":"2019-09-06 10:40:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665851-1567766417062-1c9b4c9a","meta_rev_id":"86df12822b8856a0dafb232fec4bbe7addf4bdc1","_number":665851,"virtual_id_number":665851,"owner":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-09-06 10:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2019-09-06 03:39:39.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":2,"date":"2019-09-06 10:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":1,"date":"2019-09-06 10:06:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 11:11:19.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-09-06 03:39:39.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2019-09-06 10:06:01.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-09-06 10:40:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57952edcd6aef2a8f6a9e7d097d6928e38c59dc8","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-06-18 02:57:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6330dc136e80c9cb850515123c6139ef4b6dafed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 03:25:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/51/665851/1/check/openstack-tox-pep8/4a4dc25/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/51/665851/1/check/openstack-tox-py27/0080118/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/51/665851/1/check/openstack-tox-py36/d8e4595/ : SUCCESS in 3m 47s\n- openstack-tox-py37 http://logs.openstack.org/51/665851/1/check/openstack-tox-py37/9e230e1/ : SUCCESS in 4m 46s\n- requirements-check http://logs.openstack.org/51/665851/1/check/requirements-check/18a27f3/ : SUCCESS in 3m 28s\n- openstack-tox-docs http://logs.openstack.org/51/665851/1/check/openstack-tox-docs/7530fe6/html/ : SUCCESS in 3m 13s\n- build-openstack-releasenotes http://logs.openstack.org/51/665851/1/check/build-openstack-releasenotes/f674347/html/ : SUCCESS in 3m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/665851/1/check/openstack-tox-lower-constraints/713895d/ : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5051e7bd27076061c38fcfb242e1598eb6aacd61","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-06-18 13:03:32.000000000","message":"Patch Set 1:\n\nunfortunately this won\u0027t work smoothly as we have svg images in the docs :/\nChapter 1.\n! Bad character code (-1).\n\\Gread@generic@aux ...^^I10 \\catcode \\endlinechar \n                                                  5 \\@makeother \\:\\@makeothe...\nl.100 ...e6613d3e758bfb5ff9622d6f07/bifrost}.svg}}\n                                                  ^^M","accounts_in_message":[],"_revision_number":1},{"id":"247ab84592320e3cfd8f7e3da90d272fa0e6a974","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-06-19 01:40:23.000000000","message":"Patch Set 1:\n\nWeird, my local running is fine, after having a check, the svg is not downloaded in my environment :)\n\nthe badge image is a common issue, I believe there will be a general solution.\n\nWARNING: Could not fetch remote image: https://governance.openstack.org/tc/badges/bifrost.svg [HTTPSConnectionPool(host\u003d\u0027governance.openstack.org\u0027, port\u003d443): Max retries exceeded with url: /tc/badges/bifrost.svg (Caused by NewConnectionError(\u0027\u003curllib3.connection.VerifiedHTTPSConnection object at 0x7fb378a1af98\u003e: Failed to establish a new connection: [Errno 101] Network is unreachable\u0027,))]","accounts_in_message":[],"_revision_number":1},{"id":"d6779685a6ebb0ff00c3806cc6d298e6be919ffe","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-06-25 01:31:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"07ae7ad4bd20ac08ff1fdae2ed62576d4caa6759","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 05:05:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/51/665851/2/check/openstack-tox-pep8/6a0e94a/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/51/665851/2/check/openstack-tox-py27/32fbbde/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/51/665851/2/check/openstack-tox-py36/8a78845/ : SUCCESS in 4m 59s\n- openstack-tox-py37 http://logs.openstack.org/51/665851/2/check/openstack-tox-py37/16c8990/ : SUCCESS in 4m 48s\n- requirements-check http://logs.openstack.org/51/665851/2/check/requirements-check/ed1c73f/ : SUCCESS in 2m 48s\n- openstack-tox-docs http://logs.openstack.org/51/665851/2/check/openstack-tox-docs/4d604ed/ : FAILURE in 4m 17s\n- build-openstack-releasenotes http://logs.openstack.org/51/665851/2/check/build-openstack-releasenotes/12113aa/html/ : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/665851/2/check/openstack-tox-lower-constraints/ff6d390/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"ba9704a583cafefacdaa87dd748d482cd5881678","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-01 11:11:19.000000000","message":"Patch Set 2:\n\nhey Kaifeng, to be able to pass CI we need to add imagemagick to binddep.txt for dpkg systems","accounts_in_message":[],"_revision_number":2},{"id":"93b6c3112e132e1d856abd976c1ea8b1798583ce","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-07-01 14:57:25.000000000","message":"Patch Set 2:\n\n\u003e hey Kaifeng, to be able to pass CI we need to add imagemagick to\n \u003e binddep.txt for dpkg systems\n\nyep, for testing locally, you will need that, I haven\u0027t planed to land pdf patches yet.","accounts_in_message":[],"_revision_number":2},{"id":"f54e76854fcaad6283cbb21f6efcabaa85a70e7b","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-09-05 06:47:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"321b22bec191bbf3dd9bbb1e8936b4470695d7ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 07:39:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac49b7d4ccb48538b03090e8073e2f1 : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce86952d6c1d43e398e9b02164d8fcf8 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4734eee74c9341d18a40e0c9762741cd : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfcc09b78cc945f3842d9a81f51e8300 : SUCCESS in 5m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5a7dcc232298467d88e6f2d01cfb9102 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/264f11d6a6674d02ae4a357dfa4b3e69 : FAILURE in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc8dbd9bed364bfca1e9743b47248ad8 : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d460b50b06bd4e79a82e8f5a5fbbe515 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"2da575802cb2ae0f98927b11084790a4ec9488c4","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-09-06 02:14:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"635e0a9b92c3409d4df202a5375025769471fad4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:31:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d233fda1c6af4e4495da78fe81849be4 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb3fab400d074f0f953d991a732c4bb4 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaaae10239634d129f6a1672c1e46134 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5402afa5ea50465caf29444de8e05758 : SUCCESS in 5m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8cda97f83ce1411883fe2bc9faf996a6 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6608eee2ba684e08b682877ad971a605 : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2261e67180014596b5a76e3a2f04dab2 : SUCCESS in 5m 41s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/90eef8e596ea46e68c12c17e28c8ce13 : SUCCESS in 33m 53s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/4a796f921e874ae5b031d08ab334c681 : SUCCESS in 21m 11s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/b4909d986fd14080b83fd1e7d8860f97 : RETRY_LIMIT in 3m 11s (non-voting)\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b95dc195ce6e4265b36eac495b5107d6 : SUCCESS in 29m 12s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/b4ebc9035e184ab5878eba5fc1d1f265 : RETRY_LIMIT in 5m 56s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/1b003cde3205442cbef9efe5b08373c5 : FAILURE in 46m 45s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/c0ed29845ce94269bd9b880da0b5c962 : FAILURE in 56m 04s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b4b6b4a777584f24943337c084bdd934 : SUCCESS in 34m 22s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/fe61400593bb4429a7de26b7a6b0d347 : RETRY_LIMIT in 1m 40s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/60f4f99bd26b49a49dc3a03590a99865 : TIMED_OUT in 1h 00m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45f2ab5cb27a409c90ba2c27e9dcc6ae : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":4},{"id":"e445f9a41209254860ca72da05437725bb7c2c94","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-09-06 03:39:39.000000000","message":"Patch Set 4: Code-Review+2\n\nThank you Kaifeng.","accounts_in_message":[],"_revision_number":4},{"id":"bb53401a442582831fa0b86f81e44a751a16887c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-06 10:06:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f9a9b7508310be77cb275e74fc70581ebc16c9eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 10:06:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a18d67a792d2105caf2cfa33bca0478f5e07e8d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 10:40:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ed2e70a0c1342ec8212408cc1fa0c17 : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c96cc53b2cc491484a8fa212da42e2a : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b24ff61c10444b41a4380cd980b089d8 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87d06bfbe062484ba89da0b47267981c : SUCCESS in 4m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b8fbbbed65214bdcb75d2154f54a7b8d : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14d240ecfd2142b4a7e3d20bfbed5852 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13c9175f292342df8618c75321dd5e37 : SUCCESS in 4m 16s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/488f69c292564d9e9f2e9c7c7045099e : SUCCESS in 19m 49s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d0920572a9754b108353756f7f9bcdaf : SUCCESS in 23m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3457c057bb934fb39b973b5f66be8913 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":4},{"id":"f20f0981f203613cdc407da619fa8f429a7ced0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 10:40:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8b007c59cc7abdc79e17193d71931bd37e5f0b34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 10:44:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ffe9e6cb1144e6e9bd822e6708a47dd : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0bc77a53c08f24b97324e1e3628fa043c7735af2","revisions":{"f857c4e7dc88ef25efe95dcc4cac07ceb8636451":{"kind":"REWORK","_number":1,"created":"2019-06-18 02:57:22.000000000","uploader":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"ref":"refs/changes/51/665851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/51/665851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/51/665851/1"}}},"commit":{"parents":[{"commit":"563dab994c1849542a1aca9b8a4dce0d2aaf102a","subject":"Update Python 3 test runtimes for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/563dab994c1849542a1aca9b8a4dce0d2aaf102a"}]}],"author":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-18 02:51:04.000000000","tz":480},"committer":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-18 02:51:04.000000000","tz":480},"subject":"[PoC] Enable creating pdf doc","message":"[PoC] Enable creating pdf doc\n\nChange-Id: Iebca5c4cc8dfdecf7c561b3612691b62ca4ac0c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f857c4e7dc88ef25efe95dcc4cac07ceb8636451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f857c4e7dc88ef25efe95dcc4cac07ceb8636451"}]},"branch":"refs/heads/master"},"f3a892f5a2eb05f5dc490fa34a216da45416d300":{"kind":"REWORK","_number":2,"created":"2019-06-25 01:31:13.000000000","uploader":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"ref":"refs/changes/51/665851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/51/665851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/51/665851/2"}}},"commit":{"parents":[{"commit":"b8f8c88bf4f7b737344b24880c7ff965ab73eebb","subject":"Merge \"Deprecate the Nginx data store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b8f8c88bf4f7b737344b24880c7ff965ab73eebb"}]}],"author":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-18 02:51:04.000000000","tz":480},"committer":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-25 01:24:34.000000000","tz":480},"subject":"[PoC] Enable creating pdf doc","message":"[PoC] Enable creating pdf doc\n\nChange-Id: Iebca5c4cc8dfdecf7c561b3612691b62ca4ac0c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f3a892f5a2eb05f5dc490fa34a216da45416d300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f3a892f5a2eb05f5dc490fa34a216da45416d300"}]},"branch":"refs/heads/master"},"7ee4f3dc70c12586c2a3b7b08cfd4eb0aaf7b484":{"kind":"REWORK","_number":3,"created":"2019-09-05 06:47:25.000000000","uploader":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"ref":"refs/changes/51/665851/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/51/665851/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/51/665851/3"}}},"commit":{"parents":[{"commit":"b8f8c88bf4f7b737344b24880c7ff965ab73eebb","subject":"Merge \"Deprecate the Nginx data store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b8f8c88bf4f7b737344b24880c7ff965ab73eebb"}]}],"author":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-18 02:51:04.000000000","tz":480},"committer":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-09-05 06:45:19.000000000","tz":480},"subject":"Build pdf doc","message":"Build pdf doc\n\nThe is one of community goals that each project could produce a\nsingle PDF file. The pdf should be in the output of openstack-tox-docs\njob.\n\nTeX packages are required to build PDF locally, following is recommended:\n\n* inkscape\n* texlive-latex-base\n* texlive-latex-extra\n* texlive-fonts-recommended\n\nMore about the goal:\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\nhttps://etherpad.openstack.org/p/train-pdf-support-goal\nhttps://etherpad.openstack.org/p/pdf-goal-train-common-problems\n\nChange-Id: Iebca5c4cc8dfdecf7c561b3612691b62ca4ac0c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7ee4f3dc70c12586c2a3b7b08cfd4eb0aaf7b484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7ee4f3dc70c12586c2a3b7b08cfd4eb0aaf7b484"}]},"branch":"refs/heads/master"},"0bc77a53c08f24b97324e1e3628fa043c7735af2":{"kind":"REWORK","_number":4,"created":"2019-09-06 02:14:53.000000000","uploader":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"ref":"refs/changes/51/665851/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/51/665851/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/51/665851/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/51/665851/4"}}},"commit":{"parents":[{"commit":"b8f8c88bf4f7b737344b24880c7ff965ab73eebb","subject":"Merge \"Deprecate the Nginx data store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b8f8c88bf4f7b737344b24880c7ff965ab73eebb"}]}],"author":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-18 02:51:04.000000000","tz":480},"committer":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-09-06 02:13:32.000000000","tz":480},"subject":"Build pdf doc","message":"Build pdf doc\n\nThe is one of community goals that each project could produce a\nsingle PDF file. The pdf should be in the output of openstack-tox-docs\njob.\n\nTeX packages are required to build PDF locally, following is recommended:\n\n* inkscape\n* texlive-latex-base\n* texlive-latex-extra\n* texlive-fonts-recommended\n\nMore about the goal:\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\nhttps://etherpad.openstack.org/p/train-pdf-support-goal\nhttps://etherpad.openstack.org/p/pdf-goal-train-common-problems\n\nChange-Id: Iebca5c4cc8dfdecf7c561b3612691b62ca4ac0c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0bc77a53c08f24b97324e1e3628fa043c7735af2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0bc77a53c08f24b97324e1e3628fa043c7735af2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
