)]}'
{"id":"openstack%2Fbifrost~681509","triplet_id":"openstack%2Fbifrost~master~I2fe16e572ff0705248d842d93ea178bf0f8b35b4","project":"openstack/bifrost","branch":"master","topic":"add_missing_ubuntu_deps","hashtags":[],"change_id":"I2fe16e572ff0705248d842d93ea178bf0f8b35b4","subject":"Add dependencies for ubuntu","status":"MERGED","created":"2019-09-11 14:45:49.000000000","updated":"2019-09-12 08:26:04.000000000","submitted":"2019-09-12 08:24:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681509-1568276666877-a07f33e3","meta_rev_id":"5bf6bef7ce84616f399fd0ab3aae311969c74e5f","_number":681509,"virtual_id_number":681509,"owner":{"_account_id":30911,"name":"Jan Horstmann","email":"horstmann@osism.tech","username":"jhorstmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-09-12 08:24:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-09-11 17:28:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-09-12 05:42:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-09-12 07:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-09-12 07:51:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 17:28:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-12 05:42:48.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-09-12 07:51:23.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-09-12 08:24:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"589e7beb556094541b52f346eeddddc4d432db72","author":{"_account_id":30911,"name":"Jan Horstmann","email":"horstmann@osism.tech","username":"jhorstmann"},"date":"2019-09-11 14:45:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e4bab7fc3f54a18a7c2a0170f45433969bcb4bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 15:29:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a982c050a9384105bb2d18bc52b86a92 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/636efcf5267e4efda24d2e04abfa3544 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91635611678e4f05b4b8beb6a6dc1754 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db3c6bb13c524dc6bf4e7791426f369d : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae98e30425a454ebe3b3cc26b0ab93d : SUCCESS in 6m 16s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/093f7e06d5cf41098e8dcdea469c5e76 : SUCCESS in 20m 48s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3a7a6ce49053478c990fbec88b325473 : SUCCESS in 24m 45s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/c30149faf29f4e94b9e5940991896f58 : SUCCESS in 19m 34s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6bd3d1af46954534a323e63a09923414 : FAILURE in 28m 31s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/4e8fe9d9b55a4905b6267f84179acb1a : SUCCESS in 26m 53s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/3b11d16f7c3a4a8788a02cafc2729df8 : FAILURE in 32m 22s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/6c9ebb72b3f548b98c815ce2c381bb51 : SUCCESS in 36m 57s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6089ca391c8a4c8ca39f67050a3ec204 : SUCCESS in 33m 40s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/f20a5d02349744a7a183df89b04927e6 : SUCCESS in 32m 47s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/1f2d2eb5fb5e44f5a88db9b69debfeff : SUCCESS in 25m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49a54c862bd547dfbcbe1cfae09d6092 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"55f5181cffde200f44962211b5f463795324c2b6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-11 17:28:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8cee19d5681f085872f96abdd4d540ca3f0c306c","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-09-12 05:42:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7900e09817ea79f051c48df4ce0b6937d92943f6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-09-12 07:51:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"4e3d65899807ecf516f3bb6dd65a7f0b8aa4ac1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 07:51:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"259b57883469dd56bc968dd69ccbfabcc57f3d62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 08:24:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cc5237264361210aa9c5136d179476d9fa2e125a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 08:24:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d07df9fb5b46e2be0ddbd97deb0653 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ffbe172eeeb4d9cbba3761afa530f10 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a670923e0d01406bb0f3ef7d996e9e4d : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/713eabf4cbe74eea8b86586016f821c7 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a5e14bd9a5746ccbf5da371ab681036 : SUCCESS in 6m 50s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9d763cef676b47f88ed28599b675542c : SUCCESS in 26m 53s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3ec7022dc61e4ae79858a08ecbe3685f : SUCCESS in 26m 41s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/39a9f6f9ba1248cba60a48595f8bc773 : SUCCESS in 19m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e2da13767b74664859112b663f5d01b : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d85cbf43e91c2ffa92e7851c9c8659f631571e6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 08:26:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f8c45299f804486b8c8df721a4ff6f7 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12e7305818fc9608c890ee42206dfc11eeb20aa4","revisions":{"12e7305818fc9608c890ee42206dfc11eeb20aa4":{"kind":"REWORK","_number":1,"created":"2019-09-11 14:45:49.000000000","uploader":{"_account_id":30911,"name":"Jan Horstmann","email":"horstmann@osism.tech","username":"jhorstmann"},"ref":"refs/changes/09/681509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/09/681509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/681509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/681509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/681509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/09/681509/1"}}},"commit":{"parents":[{"commit":"4351a9b76a9ed0eeb8f7ea3e68e3a93a5f932b51","subject":"Merge \"Revert \"Make the opensuse-15 jobs non-voting\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4351a9b76a9ed0eeb8f7ea3e68e3a93a5f932b51"}]}],"author":{"name":"Jan Horstmann","email":"j.horstmann@mittwald.de","date":"2019-09-11 14:38:25.000000000","tz":120},"committer":{"name":"Jan Horstmann","email":"j.horstmann@mittwald.de","date":"2019-09-11 14:38:25.000000000","tz":120},"subject":"Add dependencies for ubuntu","message":"Add dependencies for ubuntu\n\nBifrost sets up iptables rules. Especially in container\nenvironments this package may be missing, so bifrost should\ninstall it as a requirement.\nAdditionally iscsi deployments rely on hexdump, which is part of\nbsdmainutils, which may be missing as well.\n\nChange-Id: I2fe16e572ff0705248d842d93ea178bf0f8b35b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/12e7305818fc9608c890ee42206dfc11eeb20aa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/12e7305818fc9608c890ee42206dfc11eeb20aa4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
