)]}'
{"id":"openstack%2Fbifrost~703506","triplet_id":"openstack%2Fbifrost~master~I643d4d3ee9f75cc7525ddf0551a03483906c433b","project":"openstack/bifrost","branch":"master","topic":"700912","hashtags":[],"change_id":"I643d4d3ee9f75cc7525ddf0551a03483906c433b","subject":"Enable Fasttrack","status":"MERGED","created":"2020-01-20 21:28:44.000000000","updated":"2020-01-22 15:50:00.000000000","submitted":"2020-01-22 15:47:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703506-1579708027309-c94174cc","meta_rev_id":"2631d5b493edae3b086a5b0c99c4dc15bd01f896","_number":703506,"virtual_id_number":703506,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 15:47:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-01-22 15:25:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-22 10:04:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2020-01-22 15:25:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 10:04:28.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-01-22 15:25:08.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-01-22 15:47:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9436be9662b81d8c9e87a3e78ee1e4e1785beb63","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-01-20 21:28:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6dc626c50968d70faee29115a60f9b5569557058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 21:45:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/339d1cc33d68409e84079e1b32c22b32 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/673856588b46420080bc29580e920317 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3f7a0cec70b4eb2a227258d4770bc03 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5716f37c58124491b4c564b868450b32 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6759b5c1e65945d8b53cddf84ccc9457 : SUCCESS in 4m 55s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/84da1a9231644a2fab8f26a9572e207d : FAILURE in 11m 46s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/312ef0ac04cc4b19a260f9dbe64d992a : FAILURE in 9m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/516dfde2a7234d0a8a8294a58f7adf11 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"72f0b90966dc687632e327f9ad7031b2064afff1","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-01-20 21:46:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ab22967b33c8e4dbec1bba932c70fccc6dfa644d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 22:21:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/903b2734ab034a4ebcedc941ed3a0e4e : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65b1a3d2924a48e2a693105b3a33f645 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/757e18cc06b54938ad0c768424149845 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37d38ad6153c4d199523e7aa7bdd1186 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4fa5374e02047249f89562cfaac1b56 : SUCCESS in 4m 53s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/87742a545d5a471f94eaf36d181f1718 : FAILURE in 19m 11s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05b424ca082e4ef7ade55192703553fd : FAILURE in 14m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9f8dcc9bd1c4914988a4b3736a4ef46 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":2},{"id":"ecf937b4eec88ee318fa623a40c0eaf639ce981c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-01-20 22:51:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b9a57680671348bbfcd6ff2c1fb99345bdc8ee20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 23:17:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f285f34fd3d42e79270ea44b323306b : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8adb120bd6184af282c0a1f8e717e184 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17f389568ea44c6f81bc3de52c030f8e : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d79d4e35f284496584eb8245dcea00e3 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90659053450a4a54aad67d0306105dc3 : SUCCESS in 4m 09s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d9f9ad494f39414088c366e6826a1b34 : FAILURE in 20m 02s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7741415430364b41b97b2483e68951ca : SUCCESS in 17m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f94f458c3ea4197845554625b396489 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":3},{"id":"4b85a844f87ec9a64f32071b40cb398270e77663","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-01-22 10:04:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9eaa1b1ca23737f8e86d267de96b111b1b907695","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-22 15:25:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ede724898c9c71da47ad3f9f4712ceac46057f34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:25:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"91e10fadf4e679a6c1d4dd9a2c5a608095583e00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:47:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4c7dc85046814dbeb2e8a6c040893f9ce433f595","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:47:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/496892575ede46309082202c3463a90a : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfff6883f5d943ad80b64c2086e942fa : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/859a53a69bb643d39ed90e95e4432a65 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4fe29ea0d1a45889b845cb3fbce10a3 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85a0ad2b19ef429aa96533ab6dc8504e : SUCCESS in 4m 47s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9dfd59413d5434a8712f640c29f2f0b : SUCCESS in 19m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e5ba5850dda4a098e4cb5d4b70a5e3a : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"661812159933d22a0e6571d7dc5f137a434b40a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:50:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3fc9ada118c4e2e9da5e6ed90129f87 : SUCCESS in 2m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77641c2d03404e66889ae35827028661 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eefd7739d9f641de1e377fd21a7d812c879a2455","revisions":{"d658752137a327e401a64f97ccf06e3b76304ebe":{"kind":"REWORK","_number":1,"created":"2020-01-20 21:28:44.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/06/703506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/06/703506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/06/703506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/06/703506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/06/703506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/06/703506/1"}}},"commit":{"parents":[{"commit":"735f692753ba728922671aa85b2c9d14d91175f9","subject":"Add a listen-address to the dnsmasq bind","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/735f692753ba728922671aa85b2c9d14d91175f9"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-01-20 21:27:32.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-01-20 21:27:32.000000000","tz":-480},"subject":"Enable Fasttrack","message":"Enable Fasttrack\n\nDefaults bifrost to leverage fast track. This will enable\nbifrost\u0027s CI to identify failures related to code changes\naround fast track... which is ultimately a little more\ndifficult to test in the OpenStack oriented CI jobs.\n\nChange-Id: I643d4d3ee9f75cc7525ddf0551a03483906c433b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d658752137a327e401a64f97ccf06e3b76304ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d658752137a327e401a64f97ccf06e3b76304ebe"}]},"branch":"refs/heads/master"},"71eef085dd860df2741715b8a9e442ee96706481":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-20 21:46:55.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/06/703506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/06/703506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/06/703506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/06/703506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/06/703506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/06/703506/2"}}},"commit":{"parents":[{"commit":"9230e0ef751973cf48ba49edadc98bcd7731ca9f","subject":"Add a listen-address to the dnsmasq bind","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9230e0ef751973cf48ba49edadc98bcd7731ca9f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-01-20 21:27:32.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-01-20 21:46:55.000000000","tz":0},"subject":"Enable Fasttrack","message":"Enable Fasttrack\n\nDefaults bifrost to leverage fast track. This will enable\nbifrost\u0027s CI to identify failures related to code changes\naround fast track... which is ultimately a little more\ndifficult to test in the OpenStack oriented CI jobs.\n\nChange-Id: I643d4d3ee9f75cc7525ddf0551a03483906c433b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/71eef085dd860df2741715b8a9e442ee96706481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/71eef085dd860df2741715b8a9e442ee96706481"}]},"branch":"refs/heads/master"},"eefd7739d9f641de1e377fd21a7d812c879a2455":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-20 22:51:40.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/06/703506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/06/703506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/06/703506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/06/703506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/06/703506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/06/703506/3"}}},"commit":{"parents":[{"commit":"dc6c45c80579fc9c637ac88c5acdd52d97283e04","subject":"Add a listen-address to the dnsmasq bind","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/dc6c45c80579fc9c637ac88c5acdd52d97283e04"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-01-20 21:27:32.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-01-20 22:51:40.000000000","tz":0},"subject":"Enable Fasttrack","message":"Enable Fasttrack\n\nDefaults bifrost to leverage fast track. This will enable\nbifrost\u0027s CI to identify failures related to code changes\naround fast track... which is ultimately a little more\ndifficult to test in the OpenStack oriented CI jobs.\n\nChange-Id: I643d4d3ee9f75cc7525ddf0551a03483906c433b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/eefd7739d9f641de1e377fd21a7d812c879a2455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/eefd7739d9f641de1e377fd21a7d812c879a2455"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
