)]}'
{"id":"openstack%2Fbifrost~703954","triplet_id":"openstack%2Fbifrost~master~I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d","project":"openstack/bifrost","branch":"master","topic":"test-libvirt","hashtags":[],"change_id":"I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d","subject":"Prepare to switch CentOS 8 job voting again","status":"MERGED","created":"2020-01-23 09:47:13.000000000","updated":"2020-03-09 15:06:48.000000000","submitted":"2020-03-09 15:04:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703954-1583766257699-46c24852","meta_rev_id":"20f62e9c596a2aa849f310128e823e3fcab3ea66","_number":703954,"virtual_id_number":703954,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31556,"name":"Georgy Karataev","email":"georgkarat@gmail.com","username":"gkarat"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 15:04:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":31556,"name":"Georgy Karataev","email":"georgkarat@gmail.com","username":"gkarat"},{"value":2,"date":"2020-03-09 14:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-03-06 13:58:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":31556,"name":"Georgy Karataev","email":"georgkarat@gmail.com","username":"gkarat"},{"value":1,"date":"2020-03-09 14:36:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":31556,"name":"Georgy Karataev","email":"georgkarat@gmail.com","username":"gkarat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 13:48:11.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-02-02 09:22:11.000000000","updated_by":{"_account_id":31556,"name":"Georgy Karataev","email":"georgkarat@gmail.com","username":"gkarat"},"reviewer":{"_account_id":31556,"name":"Georgy Karataev","email":"georgkarat@gmail.com","username":"gkarat"},"state":"REVIEWER"},{"updated":"2020-03-06 13:58:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-09 14:36:06.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-03-09 15:04:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d766f1259621a74e65ab5683c3bc3286baa1a546","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-23 09:47:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0965252d0227efd27bd74725959064cba77a9145","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-01-23 10:03:50.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM, needs rebasing.","accounts_in_message":[],"_revision_number":1},{"id":"8755bfa3837bfb9331149c381db8e820edecf089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 10:13:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3282bbc2c94e4426acf798c632a3026a : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eb6482b60d4411a9f4a7a680e8b0be7 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8b88287b21d49ed9d47b5fcedfedd44 : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bcba6d27f254d938d1eb113eb579932 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0b242eb370848c8bd5ce89f715a7c0c : SUCCESS in 6m 43s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b39a88eb7b164393b255f45257b93347 : SUCCESS in 19m 59s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/9b285cde2fb94542bf98b54b525cf5af : FAILURE in 6m 09s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fbd97dca4a2479d81380ce0de9f16c9 : FAILURE in 7m 34s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2d5b4e2bdb943938d5c72150e2eacc1 : FAILURE in 9m 11s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c34155597f6c4c7ebceaa26be587b143 : FAILURE in 11m 28s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/2aa15614198946499ff6e5a17d598e70 : FAILURE in 9m 41s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/e91cb29f17d34b40be322ffde451e943 : FAILURE in 7m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"824573aa62b47ef445ea09c49754062aa33c419a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-23 10:15:19.000000000","message":"Patch Set 1:\n\nI should\u0027ve put a big brilliant WIP in front of this :D","accounts_in_message":[],"_revision_number":1},{"id":"394acba56c10bb9e3313fb329535e2d12b944b00","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-23 10:39:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7bb4eccb06a13c586347f6c33f98b345b8b9252f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-23 10:39:53.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"674e7b7f3d1494c7c3f2bde7babc29e99862d675","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-23 10:52:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9b1f96a8f196fa77d26b86ddbd2f37b409e7d25e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 11:13:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb55fab967b14546bcaa3c5d5ae15d55 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d36f37289ba477583be1608e1782070 : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bf36bd423f84128b120da44725db1c9 : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e746e6ec708049f9ab24b22816cb6b44 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1877c3812b74b3389d660354d0ab822 : SUCCESS in 8m 10s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92a76a3d3dcd4605b7d8170c617962e8 : SUCCESS in 19m 02s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/47f1a5b25e5542ab89eee556d6249c88 : FAILURE in 6m 58s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/132a10f4f6d24d25bab3f59b182bff35 : FAILURE in 6m 46s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2c1a4d0d7d94b48980face40d5fccf9 : FAILURE in 6m 16s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a046c40175848a189d4bd81134297f1 : FAILURE in 10m 31s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/82f790d20b8d4b61ac4a92a40cb3451c : FAILURE in 8m 33s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/d07cf50ff03e4126be7aef60e1b4873d : FAILURE in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a8c943f09d6407a4748c4f4f117c9418c3f58698","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-23 14:42:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3ba6bf30f4111d803615eb3b533cdfb2cd099500","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-23 14:43:14.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"4232680e265e2cd27f23154d646ec0140c9fa01b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 15:06:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33780b6ff2c441eda8bbf817a9ce200f : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbe4d9f0ff3949d9b13ed7fa4736734f : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68ee176c323d4611a32f1559d77c22e1 : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4eaa903b5f0421ea08a68ed51ff93b4 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/118d6d59eb394de89556a50d7af33ecc : SUCCESS in 7m 13s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f602de9890a4d368543f0dfe91c81d7 : SUCCESS in 21m 49s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/29aaafc75f4d46728008d2f40a2e4b4e : FAILURE in 7m 33s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca1e7b7d32d64be0a6ba59e905c777ea : FAILURE in 8m 13s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95cdf62648cd4384a6f7c39a78d32826 : FAILURE in 7m 58s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01c7f7168c264bd788d0684173fdcc1d : FAILURE in 12m 59s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/5436259526824d72aeaebb367fe91be9 : FAILURE in 7m 02s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/72a581a9dc7b4541800b2160e1a5c111 : FAILURE in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f21e1338f32052725a35522c5b91165f1b0d6cc2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-23 17:08:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5b32d5f0f61d7fc0de74a3ed67d5944eec519725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 17:49:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/599b5b2f84a14ccab7045dfd9bc7309d : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49e7c54ed8c84999aede9f590be20141 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3b8e29ccdf44553902049df1f3eb67d : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3949f6b0dbc840a18ff44f6753f84ee9 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e0943b41e34ffd8560055e63dfc7c7 : SUCCESS in 12m 09s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98d8f78a502745779102984a9c468a2c : SUCCESS in 19m 13s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/7979154fe88c405a840951a07a5837c1 : FAILURE in 9m 54s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1404a457843046a3b96b28a6058053d8 : FAILURE in 5m 36s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ac0b981b15a4a59adc3f5a87c3b5ea3 : FAILURE in 5m 49s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3aaa0926d4143148b051104aaccb007 : FAILURE in 11m 22s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/b992640ba73f4dc28cb1722e201e9f6d : FAILURE in 10m 50s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/47cfd72afc194d738b0f87c2d2863739 : FAILURE in 13m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e22e2eae06bb7eb00b47110856694415f5766825","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-24 09:54:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e07081d679403c55e4fed1a262d5cd87834d82c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 10:26:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363b0870bb824b1d82f614449870b7c5 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33c9b45d46d4401cb8fd019e66bfd42c : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cd47eced04d487780a0c7d7a3ca35c3 : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96bad9fca62d47b4a99f32e211456015 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6596e3ab625244799733259e23d4d991 : SUCCESS in 7m 00s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbdb13b494344ab9b2d049200e162ab4 : SUCCESS in 20m 23s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/338944391a6341e0afbee57104f6ab66 : FAILURE in 15m 08s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1764aa3f46794b5482a39810513137a8 : FAILURE in 6m 23s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91a2245957824541b9169129cca4969d : FAILURE in 13m 43s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee06bd7f84a5437aba07939a8e6e929a : FAILURE in 11m 43s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/fc48a091bcdb4399aacee558ffdbb800 : FAILURE in 9m 46s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/050f4b254d0d486e91c7bdeb40c64a98 : FAILURE in 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b36ae6e3f99e693e3cd796c187f154bead058c2b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-24 13:57:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fff506b0b0a2dbdabeb6baf7ad59e4a107770e82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 14:20:31.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0105bfb293574caca92de5007826fb72 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1f35655b89e4af6a7cb16b09f11c73e : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26c518cfa9474bedb1ad5cd3e22c4a6d : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5d2195cfdbb4acd809ec17fea706b5d : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9974b4df0e2f4ee1a94df08f72961dc3 : SUCCESS in 6m 44s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1c1c1374a7a4e17b2678d371c23d293 : SUCCESS in 20m 30s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/9acca1f46edd4aaa97cc574de1703b5c : FAILURE in 17m 28s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f57a7c0696a948febf2d42132d3a71bb : FAILURE in 6m 57s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4a5b42d67e64131a6e38588ad77e71c : FAILURE in 10m 21s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03f92c7546fa46cfb9a9cb23071beadc : FAILURE in 11m 00s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/71a770c63b404843b93cc07b8656d13d : FAILURE in 10m 10s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/cf19897cc6fb4a45bd36374f1ed35fe4 : FAILURE in 14m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a3b6d7f76916f53c3f76e5ef3fbe6de91f2b8c8c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-24 14:43:11.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7ba6a5efa7ecedecd343c4757e1a545b63b87265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 15:05:59.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/843338f7a3fe42578b22f14c98fe3e9c : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06678196ea85455ba1beaef8cbee9c02 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43fc2b14eadf4e939eb9866fbe0646aa : SUCCESS in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13174b8379854188abe3e981e0bcc168 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/266c9f2435724b81b10ebcd83cabcbda : SUCCESS in 7m 12s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa3f633864474104a0bdc4bd4eff98a1 : SUCCESS in 18m 02s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/1b6c451534d64e60a9af134dc29a9c9e : FAILURE in 8m 17s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42929332fb3a4054b72bdbd8208b0867 : FAILURE in 11m 35s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7523b5edea4c4a5c90b3a5f50aaf4951 : FAILURE in 6m 08s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/987b5236612c49acaa33c94baf459403 : FAILURE in 12m 38s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/0a31b96df7d84b898abf4ae5b8965bc0 : FAILURE in 7m 51s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/cd6c1bded44a4507b91a8d8d4b3fd1f7 : FAILURE in 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d6a59a7c0bad5c8c5be126befced7e577bd02eef","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-24 17:48:49.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"36b9f17b3d7d1e2176aada005512be0e45ab5084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 18:11:42.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c4b7cb67dd4a108637f1082a2e5812 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2039772011de415085371a70d5d281c0 : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c116568e2ce4cd6abce1b10bdd64d2c : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17b5643789074159814f0dd36fda8770 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cec0d831a6a943439f79e7cd5849dc3d : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40db99d0cb4247f8a0fd5d642437f6d8 : SUCCESS in 8m 12s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e29fec0c2084c3ab16f93cbc92a06b6 : SUCCESS in 19m 11s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/9468f68f08084cd2a267c8493b4d39a6 : FAILURE in 8m 30s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d223ce8a0e984796be86e76e99b9d433 : FAILURE in 8m 08s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbc00734193843f09277228f4f620de0 : FAILURE in 7m 15s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8929fe69fc4c4d7aa5b993b42fcf735b : FAILURE in 10m 52s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/da06540502224501b6b3381b9766f853 : FAILURE in 12m 26s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/7a76697a316d43129bfb624f9bffac4d : FAILURE in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c25c517a68770633265468389f1291c6c549bd3c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-27 11:18:14.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"eac1d5b2f73b530fdf70a73670dd7667102464d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 11:35:00.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ebefdca97f419e9dbb95bdacc74af1 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/797f776ed00941f891b4b2dbfa715214 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1248bcdc54664c299b106d42b58881be : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17c388d95aee44cda236b2b3d4e201b1 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53ba050d2a4b42189d9919058fcfeee1 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed1b9761afbf4a70a5dbdd42a8748d5f : SUCCESS in 7m 39s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17d80b159e344570b5c09087ab6a4809 : SUCCESS in 15m 29s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/107ecce91f924edc981830f8fc2e47ba : FAILURE in 10m 33s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c65661d9394e4d479a1e9e11b1a38a88 : FAILURE in 5m 32s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8743b79907d04ccca1b9daa97bebab0f : FAILURE in 6m 11s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43568b1bda3949e0a22ef2587e181e2f : FAILURE in 10m 04s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/6f6c904f7537418a8b984c78076668c0 : FAILURE in 9m 02s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/872fefc7c6584d24a40f6dbd0e056b53 : FAILURE in 8m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"555fa66baa08e4a825aba45ba808252c7ef841ef","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-27 13:43:53.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b8b02026352c9a00775e531c391a8330dd56a1f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:07:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebdbb8b980b341069dc237ba4b899803 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7776cd2320744cdb80d011584d464d36 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dab4809a854e4b4b8591affc6fb880e3 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0608521fb0e848039b06e66bc594c1c8 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95eb612184be4b6ca489f851f6a59ffb : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbcd4ec512824015bf428897a82b70f2 : SUCCESS in 6m 07s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/284f8126ef8245f39d756e44f07be0bf : SUCCESS in 16m 38s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/42d7e70fa41a4fc68aa10a392c579e3e : FAILURE in 7m 23s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71611c8c134b4e6786c6f20644f5ed5b : FAILURE in 5m 30s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d7b93c601604e74a6a0fd7a9b9cafb3 : FAILURE in 6m 05s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64ecb22e757446bab27a6d9831c92ebe : FAILURE in 10m 13s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/727a063731d6459487ab7d983f5411ba : FAILURE in 6m 34s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/b5c58eaf260d4c14a74658987c00d037 : FAILURE in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"26fa2039b7a0c52eeae079bb3649595bff802c7b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-27 14:14:52.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7bf421a33fcd46587a6c67662b34edc03871ac6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:41:34.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8acb4b25e2844837a7ac32f50f12de4b : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/703191aafc0f4a988a1a634e4f5d9027 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f430725aa6a94ac2acee074ca50e79f5 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff6a53885f5b4fcfb359b74231d1410c : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba99dd4efded45feafc5c48a0988985d : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da47cf0156a495dbdc009e6f0a63d18 : SUCCESS in 7m 15s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/587f709d0e214929a59fb370d40ee20e : SUCCESS in 24m 04s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/f2983b6eccec4a7ebb7b582bae53c3b2 : FAILURE in 21m 11s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59d2ea8a25e94ed0a7663fc6e8e6a8ca : FAILURE in 6m 00s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6968ce9aa02c422fb436b91e12fd6621 : FAILURE in 7m 33s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4228f39ce51c4020827ceccb0206b5ff : FAILURE in 10m 40s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/79d5f9aded2144c2972d5c6178626749 : FAILURE in 7m 52s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/fa537b4f3fa94ebf8d577ff91b32d6ec : FAILURE in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"261b9242073a610f83219f1c989352078f17bf4b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-27 16:03:12.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"c399e3bc44337c4d5aa17c42db03c1a5587a1503","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-27 16:03:52.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"1bb07d555ef1421c5e18e56ec6ec2214ef7ac0ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 16:29:40.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b77f79a52614162bad40d01d46a2d06 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0572e105ee674d4c9e9ea3ad2a586b16 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa3e24b8453944688d738b7fa011c1a4 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46b182be365a4025b4179eacab8d42b2 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60321811400a4530b23be26af9831cf9 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a28a6a98cc08406689b6315be014a564 : SUCCESS in 7m 17s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3279c7ccf6f4bccad3f462516825a83 : SUCCESS in 19m 06s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/96166b8b9f4343419bfc58ed961726db : FAILURE in 22m 56s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fe879ee90774117b960ebd4a4529e70 : FAILURE in 7m 06s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0939dfa764a84bec8e4c0faec02de29e : FAILURE in 7m 08s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c90859858ce545e3bd53e214950ced33 : FAILURE in 10m 06s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/5905bc20cefd46dd9b6f789b88976aeb : FAILURE in 8m 53s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/3fdfa75ad21a470eb8891fa17997a003 : FAILURE in 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8da066b8bc0202703ff1cb4631e98703d2fb85a1","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-27 17:51:46.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"8e45c49fcb6a6451cd6a7acfc8957f99e487e8a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 18:07:55.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/226803061fec4e8392991c8d40bb5935 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f927a04320fb4610a1ee0afca9df4a28 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c4677cec84d4e4dad406ac5580a085e : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d57e423dc334f4ba3232cbc440c2bab : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ab13ec441a84abfa45e96575d1c3d5a : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88b21d1662874edcb802b1c64a01bc7f : SUCCESS in 7m 40s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2e257a5eef04261a5941b3426918e6b : FAILURE in 8m 44s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/81f84a72abec4071a6cf30a0416f4db4 : FAILURE in 9m 51s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aad00dd92f314a90a57988c13dfea04a : FAILURE in 7m 25s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa1ee4407dd54fd49fe208e93a0fe5e6 : FAILURE in 6m 35s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e201941bf0644eeb95f6879c4373893a : FAILURE in 6m 41s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/0d4a56c3a5dc435d8f232a896b7a2bb5 : FAILURE in 9m 19s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/4fe92ab3629449608a71f103b5c65132 : FAILURE in 8m 54s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"44aa9be61232b22484e110d1f9d35a5480d54ab6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-27 18:20:40.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"3629804cba6e9064241a3f738b0895f23223a5b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 18:51:46.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b96554e6b834677b364298e5f66d0d3 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e2d1bc3673b4a13aa71c103f1153b92 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98ffc0881416495aad22177b1951871d : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa311b5e95d44243bd740726285ad593 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ab9a162a9f040e1881d5e55445464e4 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/733730d9a14d477fa6ec5946845c1ef2 : SUCCESS in 7m 51s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3397dbcf13764eeb93eb1387ce0fb4a8 : SUCCESS in 17m 35s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/3a76b63bff094aebad80462108a5ee67 : FAILURE in 22m 54s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8e04bfedb584c829f2ced55853850fb : FAILURE in 7m 23s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de1b1502de904e01a58394345e9a176b : FAILURE in 5m 59s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4abdc2cdd95473197e6c9bccabf2cea : FAILURE in 9m 50s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/85e5f6eeb95343b096388d7eb6df1c82 : FAILURE in 8m 17s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/596e533546f842a89263e0dc146a5cc4 : FAILURE in 14m 32s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c426a0aed900fd5a5a7e4a09603274b8c6821877","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-27 20:43:42.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"844fb6efa857907707d3b699d0acf04ffecf131a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-27 20:47:34.000000000","message":"Uploaded patch set 19: Commit message was updated.","accounts_in_message":[],"_revision_number":19},{"id":"6cfe7540e919763229f51bed4c27ee17a74ff602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 21:18:56.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b28b87778aa549c0b310592737ef840f : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09b3dcea91b34a71b53ed722ff445047 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6be9710a1b1e4e1aa27308931ea7c1b1 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cc9bbb0ab7d406d9ebb3caa623a95c8 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51389a2295f747a7a646db4cf9a42968 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6874ffe33e24a5b94ba2b1694c8cc9a : SUCCESS in 9m 14s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e5337ed5fd64eb39a93538a45078126 : SUCCESS in 17m 20s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/da06f9371ee64890bdf56d41a4d82586 : FAILURE in 25m 00s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23daf3be20454813bf6936b9f8f9cda2 : FAILURE in 5m 59s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2096575d444d41a6b4250197227d77b1 : FAILURE in 6m 23s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9b3886f853a42618a21cffe6a07be28 : FAILURE in 19m 00s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/7aa4045158c345c887800814f7958daa : FAILURE in 7m 46s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/5bdc19e146094d36b3f82a5d226c9722 : FAILURE in 15m 07s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"b2dfec00e6e257f8bc8c4b71d1b7caa61aa23ed5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-28 10:26:48.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"fa8cbdf2bc5476d070341a6671a1d6b8ecb22e85","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-28 10:42:38.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"5198a4a72be98ce2961f80052aad2092e6f5fd31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 11:12:36.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aea24a4830948a698f581e979d9173a : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1033981948a4eb2ad6b2ce002f3581c : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85596590a5864eb8a2259c1b7e4fe6a3 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9617ac1a53df4c9daa6b910460bc245b : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8de8c85a148344e3992143ad5423b66b : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055ac7ad2b354c72b91998055d4b3577 : SUCCESS in 6m 49s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ebae687fe3f46e9a394338427089fe2 : SUCCESS in 21m 02s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/8963eb534ce543a3831636a2d466eea3 : FAILURE in 23m 23s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3eaeeabae7a349338b6d93eaa726c6c5 : FAILURE in 6m 49s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aebc48cd690d4c5390c24acc724c600f : FAILURE in 7m 19s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e930e8e4ebf742748a992ddaa8998edc : FAILURE in 14m 56s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/54c192d4b9064209833b5494ddf293e5 : FAILURE in 7m 15s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/ca8a71c0dd9b4e2b929e9b7913a7c460 : FAILURE in 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"cb8b2abe28299aeb608babf8852029c87114275b","author":{"_account_id":31556,"name":"Georgy Karataev","email":"georgkarat@gmail.com","username":"gkarat"},"date":"2020-01-29 13:46:46.000000000","message":"Patch Set 21:\n\n(5 comments)","accounts_in_message":[],"_revision_number":21},{"id":"8c9ef5c9303870af0f027fb6b4a04be0558e8b03","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-01-29 13:48:11.000000000","message":"Patch Set 21:\n\nGeorgy, we probably won\u0027t support CentOS 7 in git master (aka Ussuri). I think it\u0027s fair to expect the last release supporting it to be Train.","accounts_in_message":[],"_revision_number":21},{"id":"993fc9f7c5d7caaff9830ca918d76c89e2042ea2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-30 16:59:35.000000000","message":"Patch Set 21:\n\nthe main issue right now is that there could be a potential bug in libvirt for CentOS 8.\nI\u0027ve opened a bug in Bugzilla for that:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1795980","accounts_in_message":[],"_revision_number":21},{"id":"39ac04ed7fed54746758a57588e65c0d8f8638f6","author":{"_account_id":31556,"name":"Georgy Karataev","email":"georgkarat@gmail.com","username":"gkarat"},"date":"2020-02-02 09:22:11.000000000","message":"Patch Set 21:\n\n(3 comments)\n\nSorry, haven\u0027t properly read the patch desc. I\u0027ll try to take a look at centos8 bugs and probably propose some fix","accounts_in_message":[],"_revision_number":21},{"id":"0b38ddc826a6f04360511b4532cbf07ba50d821e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-05 16:48:33.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"b7dabac5684be51c2240dcf658b804a0e8318f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 17:12:24.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50cc3d86b85d4fc283dd9ba7e988adaf : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d05fd9d11b4d4c01812acb678e0af0e7 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fafcb60b642e49948cc6a2d56a9196f9 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dda643b0597b4d8897f0952d7df89237 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8511c0c7f1f494abeccf472d5de01c8 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70fccd6aeadf45b4aecfc7952d39ca50 : SUCCESS in 9m 41s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/454462aab24540308aaf50cce6d9e9c9 : SUCCESS in 19m 37s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/4b038fcaac8d4a5687be5a5fb9d7d0b6 : FAILURE in 19m 58s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b090ab22a9b486d8e54677ca1bab6a5 : FAILURE in 7m 55s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/331480aaea2b4ceea3f42c48bc21f7cf : FAILURE in 7m 43s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a20887515a204d908a265549704b00ac : FAILURE in 10m 47s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/ce3d3249ecfa4f7796f85e5ed75aeb1c : FAILURE in 8m 14s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/f71912a95e044d17a5162ea037c4c82e : FAILURE in 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"140c2a9868646b85be1a4fee9079df00040875ea","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-05 17:41:59.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"78f4ccafcbd390436e64ccb49de1a251929fe884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 18:04:53.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecf352ee44e147008de680aaf9f805ed : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9155d3cc2e3d44ffb9a62479ff105679 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e9dbfba267d495db0d78a37cf301e50 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21ed403066654098b27576f244d78e09 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfc387a547094013ba28e6f1f6363bb9 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037e421c5374469b841b2d00374b9273 : SUCCESS in 8m 06s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b40c009342db4da38bce40c6b64c2dc4 : SUCCESS in 16m 40s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/f90a5c27190f43c9a3ab67486b7050be : FAILURE in 17m 44s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/715c45e2e6b549f7bf1ca27189c4d356 : FAILURE in 7m 20s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9fbbe51d51449debf0c3409c86b16be : FAILURE in 6m 43s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ce37bd988344d698af342d1acd0b5e0 : FAILURE in 10m 45s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/2ca676d46d3b4532850a687de9653299 : FAILURE in 8m 53s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/9d577b25a60e4fda9277e88bf5fc6447 : FAILURE in 14m 58s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"34bc3ea3d9b0c213255c3da21ba821a254fe957f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-06 08:21:55.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"ae5a09f83f79536a1ae810ac1b0e39a658d3910e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 08:39:10.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b684ed16a50b42efad088642c8b9e8fd : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59801c6ceea440398646bdd1bcc9c342 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04e0c2380bc3417b8ad881bc59cb5ac5 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80037bc403e8461b95ff568f48bea5d0 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8042a2ec18bc42288543c3232d873943 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3dbf092363346d089631a06f33d4e9d : SUCCESS in 6m 49s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ecc91ca12c748e097e9ba0f2198bc96 : FAILURE in 11m 41s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/2246cd0be6b344ecad5f1a2fd6fb10d6 : FAILURE in 15m 48s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/589bcbfcb1a8434fa00ea3ceead34ba9 : FAILURE in 6m 33s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93310a24e48d40f6af00e3eb87de898e : FAILURE in 5m 48s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4ddec218fd547a784cc9b53231136c3 : FAILURE in 8m 50s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/833edefe8455433187c93b8735602913 : FAILURE in 7m 24s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/11b70bf03dee44bc8597e79b4c6e464b : FAILURE in 13m 40s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"01b864078bac51bd78ef149a90de4549f41759b8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-06 11:33:04.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"a467318482cc0d5544d7a06283c89a34939a81ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 12:08:21.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eea746b41ff482580a037fd0f854e65 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6073684b34144a6a8977b7f564ebb7f7 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a1770c814444219902de5b6897961f2 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25e188b1c2814038b19b2003f0245538 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad90bf9ac2834f319b41c6b910c93b41 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac00c65fa754def9c45a625d35c8e8e : SUCCESS in 7m 19s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/775566d79e4e4cdf9d07c8f4624cb8cb : SUCCESS in 17m 20s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/d4c4be3914af4bbfaf74c2986bb0d2ae : FAILURE in 20m 39s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4383443d7214c4e9b45e2a7da0c0e77 : FAILURE in 7m 10s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f798f38a55654231a47cf45105085f20 : FAILURE in 7m 11s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/809599df499d4e3ea3bb6525874ddd63 : FAILURE in 10m 20s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/04a7fbb926b24696a02ce00ffd87baf0 : FAILURE in 9m 33s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/d6d852fd93fd408489f2b5fd6d9da068 : FAILURE in 14m 38s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"367530941afe05219016bcf57fa095fad03e4ce7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-06 14:01:27.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"d5a931bc26a37475355a1d4ff0dfe08b3ca69c33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 14:28:43.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d320e606dcf843f199f5ca272a762953 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/467880b8fc094b52a7f96ba0e30fbe2d : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5625b6c3f7da45ba9148bcda0b93be48 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9987c86ce23b42e0b0151e4ea3f1323d : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb7a82b33ca8443ca13eea6bdf1c1b39 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1001fe431e6f48ed9274555c531ff7d3 : SUCCESS in 8m 57s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e9b5d2bf40d4dfdb088d4eaf604ef5c : SUCCESS in 18m 39s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/4945c4b125324b5a8c06f69098d119cc : FAILURE in 22m 24s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e569842ad66a4f2c95509232a265e638 : FAILURE in 7m 12s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e120d2bd41764e138c5ad84fbfcd25d2 : FAILURE in 6m 57s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2be87778172437a81f88362fc011bcc : FAILURE in 11m 45s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/2d9b012cb08849dfb550c2f87248eafa : FAILURE in 8m 10s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/898f8b8f5f1948fe90163ce5a9edb527 : FAILURE in 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"4cc5df0a10d933b61c3ea9bdad5f70e7287576ee","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-11 14:41:32.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"d26c827920f7a2d019bb5181575b48258ddd81ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:09:04.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb678464fcb4d99b2ee7c2de49f18c0 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/233d74b514094a868aa92985ba576c47 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5dfb48a865584ad59879c45242eb26c0 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89720a5b577b450396c8809dc6e24cbc : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c32cddad9c7546cab5decb70a2c312c0 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48c23c2375834d2788bcc85d26778fa4 : SUCCESS in 6m 08s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9306d8eb84b4ca2ab5505776eba2798 : SUCCESS in 17m 08s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/5dc66c19b76c427dbf3e857333e79fb1 : FAILURE in 22m 44s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1704b9c975247c5b63247f9218f7ae7 : FAILURE in 5m 18s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85dac34d9d7d41c8826cecebcd52199d : FAILURE in 6m 35s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ba6c49f1a1a4241899115190a6d0a96 : FAILURE in 14m 07s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/200e9ec66c52431d9a95a49097dccea9 : FAILURE in 10m 14s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/eb25976bcf6045eb82759ad58c683174 : FAILURE in 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"30be71f0fcc0519cecac69972b8bf3b43f7a566f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-04 17:33:39.000000000","message":"Patch Set 27:\n\nthis is blocked until the bug in libvirt is fixed\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1804224","accounts_in_message":[],"_revision_number":27},{"id":"ee8f9321890ec1e6e66514100123869e791ada3e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-05 09:57:27.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"12bc133f1dc7adbf88d774fc811a2ca69846c1bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 10:13:44.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85168fd7b69e47fd9716e6786d0eabaa : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22c90d28f55549f3a71bbdb8757f0ba5 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdbc25e13e0b42309ea3ac2c58038618 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac6662114b8146f2bd60a084d154404a : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7603bbfbd1014852b4daab0c885253a1 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6600e234e55641918c2b66e8bf57cfac : SUCCESS in 6m 35s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce664995c3e7467ca53077ece0c3348f : FAILURE in 10m 57s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be368144f1e140edb4fea7ef9c25090c : FAILURE in 7m 09s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ceb28551ac4476ca612700821b92845 : FAILURE in 6m 32s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b1461d835774498a593d9a09caf2ed1 : FAILURE in 10m 56s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/775e1346f4f64710948a5e25b4d0d35c : FAILURE in 8m 33s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/b158c243e3af497db74d9579aae1a8e7 : FAILURE in 15m 00s (non-voting)\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/d7c4b4a033374d918edfa30701493c6a : FAILURE in 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"079314f2c08ce0020a20486179a94921085480a0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 11:23:53.000000000","message":"Patch Set 28:\n\nStrange that ubuntu bionic fails:\n\nTASK [ironic-enroll-dynamic : Dynamic enrollment] ******************************\n\n2020-03-05 10:08:40.349418 | ubuntu-bionic | The full traceback is:\n2020-03-05 10:08:40.349468 | ubuntu-bionic | Traceback (most recent call last):\n2020-03-05 10:08:40.349513 | ubuntu-bionic |   File \"/home/zuul/.ansible/tmp/ansible-tmp-1583402919.0079417-240485771866116/AnsiballZ_os_ironic.py\", line 114, in \u003cmodule\u003e\n2020-03-05 10:08:40.349541 | ubuntu-bionic |     _ansiballz_main()\n2020-03-05 10:08:40.349566 | ubuntu-bionic |   File \"/home/zuul/.ansible/tmp/ansible-tmp-1583402919.0079417-240485771866116/AnsiballZ_os_ironic.py\", line 106, in _ansiballz_main\n2020-03-05 10:08:40.349590 | ubuntu-bionic |     invoke_module(zipped_mod, temp_path, ANSIBALLZ_PARAMS)\n2020-03-05 10:08:40.349613 | ubuntu-bionic |   File \"/home/zuul/.ansible/tmp/ansible-tmp-1583402919.0079417-240485771866116/AnsiballZ_os_ironic.py\", line 49, in invoke_module\n2020-03-05 10:08:40.349637 | ubuntu-bionic |     imp.load_module(\u0027__main__\u0027, mod, module, MOD_DESC)\n2020-03-05 10:08:40.349668 | ubuntu-bionic |   File \"/usr/lib/python3.6/imp.py\", line 235, in load_module\n2020-03-05 10:08:40.349691 | ubuntu-bionic |     return load_source(name, filename, file)\n2020-03-05 10:08:40.349714 | ubuntu-bionic |   File \"/usr/lib/python3.6/imp.py\", line 170, in load_source\n2020-03-05 10:08:40.349736 | ubuntu-bionic |     module \u003d _exec(spec, sys.modules[name])\n2020-03-05 10:08:40.349758 | ubuntu-bionic |   File \"\u003cfrozen importlib._bootstrap\u003e\", line 618, in _exec\n2020-03-05 10:08:40.349779 | ubuntu-bionic |   File \"\u003cfrozen importlib._bootstrap_external\u003e\", line 678, in exec_module\n2020-03-05 10:08:40.349802 | ubuntu-bionic |   File \"\u003cfrozen importlib._bootstrap\u003e\", line 219, in _call_with_frames_removed\n2020-03-05 10:08:40.349824 | ubuntu-bionic |   File \"/tmp/ansible_os_ironic_payload_udlcdkc9/__main__.py\", line 355, in \u003cmodule\u003e\n2020-03-05 10:08:40.349847 | ubuntu-bionic |   File \"/tmp/ansible_os_ironic_payload_udlcdkc9/__main__.py\", line 247, in main\n2020-03-05 10:08:40.349884 | ubuntu-bionic |   File \"/tmp/ansible_os_ironic_payload_udlcdkc9/ansible_os_ironic_payload.zip/ansible/module_utils/openstack.py\", line 126, in openstack_cloud_from_module\n2020-03-05 10:08:40.349909 | ubuntu-bionic |   File \"/usr/lib/python3.6/distutils/version.py\", line 40, in __init__\n2020-03-05 10:08:40.349932 | ubuntu-bionic |     self.parse(vstring)\n2020-03-05 10:08:40.349954 | ubuntu-bionic |   File \"/usr/lib/python3.6/distutils/version.py\", line 135, in parse\n2020-03-05 10:08:40.349976 | ubuntu-bionic |     match \u003d self.version_re.match(vstring)\n2020-03-05 10:08:40.349999 | ubuntu-bionic | TypeError: expected string or bytes-like object","accounts_in_message":[],"_revision_number":28},{"id":"01e27ebd293324b4d745cb185b685c6338277e15","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 11:40:13.000000000","message":"Patch Set 28:\n\n\u003e Strange that ubuntu bionic fails:\n \u003e \n \u003e TASK [ironic-enroll-dynamic : Dynamic enrollment] ******************************\n \u003e \n \u003e 2020-03-05 10:08:40.349418 | ubuntu-bionic | The full traceback is:\n \u003e 2020-03-05 10:08:40.349468 | ubuntu-bionic | Traceback (most recent\n \u003e call last):\n \u003e 2020-03-05 10:08:40.349513 | ubuntu-bionic |   File\n \u003e \"/home/zuul/.ansible/tmp/ansible-tmp-1583402919.0079417-240485771866116/AnsiballZ_os_ironic.py\",\n \u003e line 114, in \u003cmodule\u003e\n \u003e 2020-03-05 10:08:40.349541 | ubuntu-bionic |     _ansiballz_main()\n \u003e 2020-03-05 10:08:40.349566 | ubuntu-bionic |   File\n \u003e \"/home/zuul/.ansible/tmp/ansible-tmp-1583402919.0079417-240485771866116/AnsiballZ_os_ironic.py\",\n \u003e line 106, in _ansiballz_main\n \u003e 2020-03-05 10:08:40.349590 | ubuntu-bionic |     invoke_module(zipped_mod,\n \u003e temp_path, ANSIBALLZ_PARAMS)\n \u003e 2020-03-05 10:08:40.349613 | ubuntu-bionic |   File\n \u003e \"/home/zuul/.ansible/tmp/ansible-tmp-1583402919.0079417-240485771866116/AnsiballZ_os_ironic.py\",\n \u003e line 49, in invoke_module\n \u003e 2020-03-05 10:08:40.349637 | ubuntu-bionic |     imp.load_module(\u0027__main__\u0027,\n \u003e mod, module, MOD_DESC)\n \u003e 2020-03-05 10:08:40.349668 | ubuntu-bionic |   File\n \u003e \"/usr/lib/python3.6/imp.py\", line 235, in load_module\n \u003e 2020-03-05 10:08:40.349691 | ubuntu-bionic |     return\n \u003e load_source(name, filename, file)\n \u003e 2020-03-05 10:08:40.349714 | ubuntu-bionic |   File\n \u003e \"/usr/lib/python3.6/imp.py\", line 170, in load_source\n \u003e 2020-03-05 10:08:40.349736 | ubuntu-bionic |     module \u003d\n \u003e _exec(spec, sys.modules[name])\n \u003e 2020-03-05 10:08:40.349758 | ubuntu-bionic |   File \"\u003cfrozen\n \u003e importlib._bootstrap\u003e\", line 618, in _exec\n \u003e 2020-03-05 10:08:40.349779 | ubuntu-bionic |   File \"\u003cfrozen\n \u003e importlib._bootstrap_external\u003e\", line 678, in exec_module\n \u003e 2020-03-05 10:08:40.349802 | ubuntu-bionic |   File \"\u003cfrozen\n \u003e importlib._bootstrap\u003e\", line 219, in _call_with_frames_removed\n \u003e 2020-03-05 10:08:40.349824 | ubuntu-bionic |   File\n \u003e \"/tmp/ansible_os_ironic_payload_udlcdkc9/__main__.py\", line 355, in\n \u003e \u003cmodule\u003e\n \u003e 2020-03-05 10:08:40.349847 | ubuntu-bionic |   File\n \u003e \"/tmp/ansible_os_ironic_payload_udlcdkc9/__main__.py\", line 247, in\n \u003e main\n \u003e 2020-03-05 10:08:40.349884 | ubuntu-bionic |   File\n \u003e \"/tmp/ansible_os_ironic_payload_udlcdkc9/ansible_os_ironic_payload.zip/ansible/module_utils/openstack.py\",\n \u003e line 126, in openstack_cloud_from_module\n \u003e 2020-03-05 10:08:40.349909 | ubuntu-bionic |   File\n \u003e \"/usr/lib/python3.6/distutils/version.py\", line 40, in __init__\n \u003e 2020-03-05 10:08:40.349932 | ubuntu-bionic |     self.parse(vstring)\n \u003e 2020-03-05 10:08:40.349954 | ubuntu-bionic |   File\n \u003e \"/usr/lib/python3.6/distutils/version.py\", line 135, in parse\n \u003e 2020-03-05 10:08:40.349976 | ubuntu-bionic |     match \u003d\n \u003e self.version_re.match(vstring)\n \u003e 2020-03-05 10:08:40.349999 | ubuntu-bionic | TypeError: expected\n \u003e string or bytes-like object\n\nI\u0027ve hit the same issue elsewhere in kolla, probably not us.","accounts_in_message":[],"_revision_number":28},{"id":"7ecd76dcc6204fc9726c17fd62e91d2e1f6eab9d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 10:30:52.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"da8f39dc6365bf990d21addabf65fb525900618b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 10:34:06.000000000","message":"Patch Set 28: Code-Review+1\n\n(1 comment)\n\nProbably needs a release note to say we\u0027re dropping Py2/CentOS 7 and adding Py3/CentOS 8.","accounts_in_message":[],"_revision_number":28},{"id":"782da7044d2467f94a5c8d231b5086e501c02a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 10:56:40.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/828472a54ec84c8e93879d66e4ccb249 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da928a70382a4d8f886125bb19b17aa4 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7985ac86cb441ab85108a31efb50037 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7128f5d2eeef48b19b32124919e623ff : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/821b8364f9c14d2bb2641492f162dc58 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15c1e0bc2388476c9c0001f8c71f160c : SUCCESS in 6m 17s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f64b84a00f64562b8fac0c3d9e8c4ef : SUCCESS in 16m 52s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a37ebd7d77f4232b7f2dbd0a4f236b9 : FAILURE in 5m 55s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e011a8088b474d43b20ba6dbd2485893 : FAILURE in 6m 21s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6def3f3aeb3c41138293cfa07b71f9f8 : FAILURE in 9m 19s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/88c73bea8570431baa3e9a3528afe192 : FAILURE in 8m 00s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/4f4c83d9e0f640e6833b8375070dcdfd : FAILURE in 14m 06s (non-voting)\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/187815eccdf4437498c33783724885fc : FAILURE in 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"a488f509f901d5d1a951f8683deb6e68d4579810","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 13:12:13.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"21b9547d506d4b6234da51155c58f20d31893032","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 13:13:10.000000000","message":"Patch Set 29: Code-Review+2\n\nAdded a release note and split out the testing parts into https://review.opendev.org/#/c/711416. Thanks Riccardo!","accounts_in_message":[],"_revision_number":29},{"id":"96e66460138f285bd202db268a9186f3831d31d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 13:36:03.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/447ba068be194fad835ae4bf0322dc5d : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd6156de20c84b788f0da02d71fea064 : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9eed509f20d940f7b319a9f807819964 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24398481db1d426b966585b92800e3dc : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e960d7ebd7d4129a85dfeb40f413b6b : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569b427613e44336973b03b9d5079921 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb73bf5af21748159d254958f419ef00 : SUCCESS in 5m 41s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/761d6745f42049e494b5276897354f96 : SUCCESS in 18m 47s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ceecd26c8b742cf89bad0b2eed34c36 : FAILURE in 6m 35s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4faf7fe6f77c4c268b62f215bbdd7f0f : FAILURE in 7m 21s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b76d545de37483a9c36a1f832c1dbbe : FAILURE in 10m 38s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/08ea945e22fe456590795c97ffe06a68 : FAILURE in 5m 27s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/cd9e999e985544319f063ac4fb75a020 : FAILURE in 6m 02s (non-voting)\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/0636aad0fee4429ebea36b9a798f4a88 : FAILURE in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"aaa98001be025d776739cac2d9de9f7375307214","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-06 13:38:30.000000000","message":"Patch Set 29:\n\n\u003e Added a release note and split out the testing parts into\n \u003e https://review.opendev.org/#/c/711416. Thanks Riccardo!\n\nwe shouldn\u0027t need the release note, we added one already here:\nhttps://review.opendev.org/#/c/703758/","accounts_in_message":[],"_revision_number":29},{"id":"af982d56c7e53c94685def674573ec3692c629dd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 13:58:27.000000000","message":"Patch Set 30: Published edit on patch set 29.","accounts_in_message":[],"_revision_number":30},{"id":"075545389b6343dd0e10eb0c793593c4227923e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 13:58:41.000000000","message":"Patch Set 30: Code-Review+2\n\nRemoved reno","accounts_in_message":[],"_revision_number":30},{"id":"458e9ac3ad0aad51c9c591128dfa73ddc1541348","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 14:17:22.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f47d39892e104baf961d410fc123b930 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58dbd36a6a36436e98de99ccdc7b05af : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfa93941815a450480b31c1ce55b69c1 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1af0dff62aa04678b544d55aac4cd494 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe436b24c0344e2988978eff87f4a324 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80857837b37d4776ae662487535361a3 : SUCCESS in 6m 31s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ad4bd45dcc946c39f45e29077ec6087 : SUCCESS in 17m 07s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/898fb6e17eb54e609fb031cb66672ec9 : FAILURE in 7m 10s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaf9cab953bd4020928bc7e527ed9c1b : FAILURE in 6m 30s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52065d81ecce4019beb5dcb221adbfbf : FAILURE in 10m 19s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/e99e5add9f344430a8313b8362699a26 : FAILURE in 6m 10s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/79aabe4a18c4405191eed66f5f6e45b2 : FAILURE in 6m 13s (non-voting)\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/e7c98ba2906c4c1e89a3b1a23011be1c : FAILURE in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"7b759a7e5fb6e67e97523c3c50a7eeefc27b8bab","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-03-09 14:36:06.000000000","message":"Patch Set 30: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":30},{"id":"3d3a0eda6c896b0ae2462f40186bbc18e0702cd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 14:36:18.000000000","message":"Patch Set 30: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":30},{"id":"35580c5ecde945c91db4273cd9fd883fc955ae29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:04:17.000000000","message":"Patch Set 30: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db507d376b604f6aac1720250ba923e8 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcb10c5f931644d9832912e6df76e79a : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ad0129c2adf406db48b0cd8dd452fbe : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4002cbe5236f453887dddc258ff36d5f : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2666b4179834218a14f4d2f51abc054 : SUCCESS in 8m 18s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60219d4d3e684bbfbd14e8ae74d92403 : SUCCESS in 19m 55s","accounts_in_message":[],"_revision_number":30},{"id":"0861982e0c80b9b292499736df888248d53e8319","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:04:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":30},{"id":"02043826055f32652f8af81f40d220a8dc920744","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:06:48.000000000","message":"Patch Set 30:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3a43571c4d422c917d3448483fa273 : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":30}],"current_revision_number":30,"current_revision":"0c3baf04dd11a0017695150e3b7d53c63b4ebc94","revisions":{"2539ce6abbad40f64ccf245eac969b00db38fcf2":{"kind":"REWORK","_number":1,"created":"2020-01-23 09:47:13.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/1"}}},"commit":{"parents":[{"commit":"2b743864c2a4afba16c925505d0ba2096c6187d9","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/2b743864c2a4afba16c925505d0ba2096c6187d9"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/2539ce6abbad40f64ccf245eac969b00db38fcf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/2539ce6abbad40f64ccf245eac969b00db38fcf2"}]},"branch":"refs/heads/master"},"0aad2070467d34698547c21d87bc14adc1925f0b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-23 10:39:32.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/2"}}},"commit":{"parents":[{"commit":"f4c626f303a159ad13b6f4297f18551b4067dc6d","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f4c626f303a159ad13b6f4297f18551b4067dc6d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 10:39:32.000000000","tz":0},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0aad2070467d34698547c21d87bc14adc1925f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0aad2070467d34698547c21d87bc14adc1925f0b"}]},"branch":"refs/heads/master"},"3ce17b5e00c170bd52d6f4098e95b9287f2d07d6":{"kind":"REWORK","_number":3,"created":"2020-01-23 10:52:09.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/3"}}},"commit":{"parents":[{"commit":"f4c626f303a159ad13b6f4297f18551b4067dc6d","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f4c626f303a159ad13b6f4297f18551b4067dc6d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 10:49:54.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/3ce17b5e00c170bd52d6f4098e95b9287f2d07d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/3ce17b5e00c170bd52d6f4098e95b9287f2d07d6"}]},"branch":"refs/heads/master"},"28c734f5c2c9ebba1b65af16e730ec50a4d52914":{"kind":"REWORK","_number":4,"created":"2020-01-23 14:42:43.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/4"}}},"commit":{"parents":[{"commit":"f4c626f303a159ad13b6f4297f18551b4067dc6d","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f4c626f303a159ad13b6f4297f18551b4067dc6d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 14:42:33.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/28c734f5c2c9ebba1b65af16e730ec50a4d52914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/28c734f5c2c9ebba1b65af16e730ec50a4d52914"}]},"branch":"refs/heads/master"},"99bcdc2a5098772e07bf62fe8a811999f0d8f24f":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-01-23 14:43:14.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/5"}}},"commit":{"parents":[{"commit":"f4c626f303a159ad13b6f4297f18551b4067dc6d","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f4c626f303a159ad13b6f4297f18551b4067dc6d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 14:42:49.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/99bcdc2a5098772e07bf62fe8a811999f0d8f24f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/99bcdc2a5098772e07bf62fe8a811999f0d8f24f"}]},"branch":"refs/heads/master"},"a8a4e2b45a03a69214a293ab110ad484d1b96a5a":{"kind":"REWORK","_number":6,"created":"2020-01-23 17:08:21.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/6"}}},"commit":{"parents":[{"commit":"f4c626f303a159ad13b6f4297f18551b4067dc6d","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f4c626f303a159ad13b6f4297f18551b4067dc6d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 17:07:20.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a8a4e2b45a03a69214a293ab110ad484d1b96a5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a8a4e2b45a03a69214a293ab110ad484d1b96a5a"}]},"branch":"refs/heads/master"},"87138c6e06b3844d885b7cbad46b33a90beaee73":{"kind":"REWORK","_number":7,"created":"2020-01-24 09:54:08.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/7"}}},"commit":{"parents":[{"commit":"f4c626f303a159ad13b6f4297f18551b4067dc6d","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f4c626f303a159ad13b6f4297f18551b4067dc6d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-24 09:53:31.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/87138c6e06b3844d885b7cbad46b33a90beaee73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/87138c6e06b3844d885b7cbad46b33a90beaee73"}]},"branch":"refs/heads/master"},"5513bce42d086dd9f941eb09b3403929d5754088":{"kind":"REWORK","_number":8,"created":"2020-01-24 13:57:28.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/8"}}},"commit":{"parents":[{"commit":"f4c626f303a159ad13b6f4297f18551b4067dc6d","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f4c626f303a159ad13b6f4297f18551b4067dc6d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-24 13:53:33.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/5513bce42d086dd9f941eb09b3403929d5754088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/5513bce42d086dd9f941eb09b3403929d5754088"}]},"branch":"refs/heads/master"},"f75b19e7efe9cecaf9eeb902072bc0dfea51d015":{"kind":"REWORK","_number":9,"created":"2020-01-24 14:43:11.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/9"}}},"commit":{"parents":[{"commit":"f4c626f303a159ad13b6f4297f18551b4067dc6d","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f4c626f303a159ad13b6f4297f18551b4067dc6d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-24 14:42:23.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to q35\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f75b19e7efe9cecaf9eeb902072bc0dfea51d015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f75b19e7efe9cecaf9eeb902072bc0dfea51d015"}]},"branch":"refs/heads/master"},"a6a1ab41bb1c3b5a47a2381ee13588b2e8d5ec21":{"kind":"REWORK","_number":10,"created":"2020-01-24 17:48:49.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/10"}}},"commit":{"parents":[{"commit":"f4c626f303a159ad13b6f4297f18551b4067dc6d","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f4c626f303a159ad13b6f4297f18551b4067dc6d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-24 17:44:53.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to q35\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a6a1ab41bb1c3b5a47a2381ee13588b2e8d5ec21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a6a1ab41bb1c3b5a47a2381ee13588b2e8d5ec21"}]},"branch":"refs/heads/master"},"01321954c4a3b07d8a4364e695ec739d881b9f8e":{"kind":"REWORK","_number":11,"created":"2020-01-27 11:18:14.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/11"}}},"commit":{"parents":[{"commit":"0fb46b734b8d170666b0804662dd73247066261c","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0fb46b734b8d170666b0804662dd73247066261c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-27 11:17:58.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to q35\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/01321954c4a3b07d8a4364e695ec739d881b9f8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/01321954c4a3b07d8a4364e695ec739d881b9f8e"}]},"branch":"refs/heads/master"},"85c93540b62fc311bbc7c1764f91810ae68f86d4":{"kind":"REWORK","_number":12,"created":"2020-01-27 13:43:53.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/12"}}},"commit":{"parents":[{"commit":"0fb46b734b8d170666b0804662dd73247066261c","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0fb46b734b8d170666b0804662dd73247066261c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-27 13:43:41.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to q35\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/85c93540b62fc311bbc7c1764f91810ae68f86d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/85c93540b62fc311bbc7c1764f91810ae68f86d4"}]},"branch":"refs/heads/master"},"00ff8f121997ef9464ab76f877bcd4bcb50834b8":{"kind":"REWORK","_number":13,"created":"2020-01-27 14:14:52.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/13"}}},"commit":{"parents":[{"commit":"0fb46b734b8d170666b0804662dd73247066261c","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0fb46b734b8d170666b0804662dd73247066261c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-27 14:14:42.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to q35\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/00ff8f121997ef9464ab76f877bcd4bcb50834b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/00ff8f121997ef9464ab76f877bcd4bcb50834b8"}]},"branch":"refs/heads/master"},"adfa4fe407e2853b1d223baf96fc16958c961d25":{"kind":"REWORK","_number":14,"created":"2020-01-27 16:03:12.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/14"}}},"commit":{"parents":[{"commit":"0fb46b734b8d170666b0804662dd73247066261c","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0fb46b734b8d170666b0804662dd73247066261c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-27 16:03:04.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to q35\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/adfa4fe407e2853b1d223baf96fc16958c961d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/adfa4fe407e2853b1d223baf96fc16958c961d25"}]},"branch":"refs/heads/master"},"6decd15106f533d10e5bded69f13a0b3336e13b8":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-01-27 16:03:52.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/15"}}},"commit":{"parents":[{"commit":"fbdfa6360e9ed1ea5d0b7733f54d8761eea98884","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbdfa6360e9ed1ea5d0b7733f54d8761eea98884"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-27 16:03:43.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to q35\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/6decd15106f533d10e5bded69f13a0b3336e13b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/6decd15106f533d10e5bded69f13a0b3336e13b8"}]},"branch":"refs/heads/master"},"c49eff411e10cc53914ea9703dcb8146d0ae31d3":{"kind":"REWORK","_number":16,"created":"2020-01-27 17:51:46.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/16"}}},"commit":{"parents":[{"commit":"fbdfa6360e9ed1ea5d0b7733f54d8761eea98884","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbdfa6360e9ed1ea5d0b7733f54d8761eea98884"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-27 17:51:32.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to q35\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c49eff411e10cc53914ea9703dcb8146d0ae31d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c49eff411e10cc53914ea9703dcb8146d0ae31d3"}]},"branch":"refs/heads/master"},"97539b95f7465ffea2720eb74facb8baf523f4c0":{"kind":"REWORK","_number":17,"created":"2020-01-27 18:20:40.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/17"}}},"commit":{"parents":[{"commit":"fbdfa6360e9ed1ea5d0b7733f54d8761eea98884","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbdfa6360e9ed1ea5d0b7733f54d8761eea98884"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-27 18:20:24.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to q35\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/97539b95f7465ffea2720eb74facb8baf523f4c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/97539b95f7465ffea2720eb74facb8baf523f4c0"}]},"branch":"refs/heads/master"},"3d904dfbb95b68bd2cedeafc8c4daa524c772cf5":{"kind":"REWORK","_number":18,"created":"2020-01-27 20:43:42.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/18"}}},"commit":{"parents":[{"commit":"fbdfa6360e9ed1ea5d0b7733f54d8761eea98884","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbdfa6360e9ed1ea5d0b7733f54d8761eea98884"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-27 20:43:24.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to q35\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/3d904dfbb95b68bd2cedeafc8c4daa524c772cf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/3d904dfbb95b68bd2cedeafc8c4daa524c772cf5"}]},"branch":"refs/heads/master"},"e0066e379b15ccaf5852982002f24f8e6784cae3":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2020-01-27 20:47:34.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/19"}}},"commit":{"parents":[{"commit":"fbdfa6360e9ed1ea5d0b7733f54d8761eea98884","subject":"Removing Centos 7 in favor of Centos 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbdfa6360e9ed1ea5d0b7733f54d8761eea98884"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-27 20:46:27.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to pc\n- installing package trousers to add tss user and group\nneeded by libvirt\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e0066e379b15ccaf5852982002f24f8e6784cae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e0066e379b15ccaf5852982002f24f8e6784cae3"}]},"branch":"refs/heads/master"},"be7202945c84611864ee2f3b4e8b4cdaadbe1473":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2020-01-28 10:26:48.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/20"}}},"commit":{"parents":[{"commit":"fdeefa810f645047fe340f3220b72a916ef67966","subject":"Merge \"Collect output of pip freeze in the CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fdeefa810f645047fe340f3220b72a916ef67966"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-28 10:26:46.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to pc\n- installing package trousers to add tss user and group\nneeded by libvirt\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/be7202945c84611864ee2f3b4e8b4cdaadbe1473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/be7202945c84611864ee2f3b4e8b4cdaadbe1473"}]},"branch":"refs/heads/master"},"bce760f0e892ac3a7697ce1a6d3437921783765c":{"kind":"REWORK","_number":21,"created":"2020-01-28 10:42:38.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/21"}}},"commit":{"parents":[{"commit":"fdeefa810f645047fe340f3220b72a916ef67966","subject":"Merge \"Collect output of pip freeze in the CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fdeefa810f645047fe340f3220b72a916ef67966"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-28 10:42:30.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to pc\n- installing package trousers to add tss user and group\nneeded by libvirt\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/bce760f0e892ac3a7697ce1a6d3437921783765c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/bce760f0e892ac3a7697ce1a6d3437921783765c"}]},"branch":"refs/heads/master"},"237f33f70d818e8884b5c73e03e926ae579705ab":{"kind":"REWORK","_number":22,"created":"2020-02-05 16:48:33.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/22"}}},"commit":{"parents":[{"commit":"aace973db9949aee1a8b0882b57f6f0e9c96cd11","subject":"Guard the status code checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/aace973db9949aee1a8b0882b57f6f0e9c96cd11"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-02-05 16:48:23.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to pc\n- installing package trousers to add tss user and group\nneeded by libvirt\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/237f33f70d818e8884b5c73e03e926ae579705ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/237f33f70d818e8884b5c73e03e926ae579705ab"}]},"branch":"refs/heads/master"},"70a2e6afb25496cce7179e7d751fe8c165995011":{"kind":"REWORK","_number":23,"created":"2020-02-05 17:41:59.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/23"}}},"commit":{"parents":[{"commit":"aace973db9949aee1a8b0882b57f6f0e9c96cd11","subject":"Guard the status code checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/aace973db9949aee1a8b0882b57f6f0e9c96cd11"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-02-05 17:41:50.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to pc\n- installing package trousers to add tss user and group\nneeded by libvirt\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/70a2e6afb25496cce7179e7d751fe8c165995011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/70a2e6afb25496cce7179e7d751fe8c165995011"}]},"branch":"refs/heads/master"},"26246a05549eed031b9a88b2adffb24227593c22":{"kind":"REWORK","_number":24,"created":"2020-02-06 08:21:55.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/24"}}},"commit":{"parents":[{"commit":"aace973db9949aee1a8b0882b57f6f0e9c96cd11","subject":"Guard the status code checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/aace973db9949aee1a8b0882b57f6f0e9c96cd11"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-02-06 08:21:42.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to pc\n- installing package trousers to add tss user and group\nneeded by libvirt\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/26246a05549eed031b9a88b2adffb24227593c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/26246a05549eed031b9a88b2adffb24227593c22"}]},"branch":"refs/heads/master"},"5364e8948d19db561df498034eee7ac54dd1e50f":{"kind":"REWORK","_number":25,"created":"2020-02-06 11:33:04.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/25"}}},"commit":{"parents":[{"commit":"aace973db9949aee1a8b0882b57f6f0e9c96cd11","subject":"Guard the status code checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/aace973db9949aee1a8b0882b57f6f0e9c96cd11"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-02-06 11:32:57.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to pc\n- installing package trousers to add tss user and group\nneeded by libvirt\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/5364e8948d19db561df498034eee7ac54dd1e50f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/5364e8948d19db561df498034eee7ac54dd1e50f"}]},"branch":"refs/heads/master"},"d745f87ee350f78380e553463287762f55e58455":{"kind":"REWORK","_number":26,"created":"2020-02-06 14:01:27.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/26"}}},"commit":{"parents":[{"commit":"aace973db9949aee1a8b0882b57f6f0e9c96cd11","subject":"Guard the status code checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/aace973db9949aee1a8b0882b57f6f0e9c96cd11"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-02-06 14:01:13.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to pc\n- installing package trousers to add tss user and group\nneeded by libvirt\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d745f87ee350f78380e553463287762f55e58455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d745f87ee350f78380e553463287762f55e58455"}]},"branch":"refs/heads/master"},"c3326addc1aa6dd453c4eac48d6578c107d5272a":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2020-02-11 14:41:32.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/27"}}},"commit":{"parents":[{"commit":"74f8005978b2d61067c85c811cf3a2ab3b90baa9","subject":"Merge \"Sync pip_install files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/74f8005978b2d61067c85c811cf3a2ab3b90baa9"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-02-11 13:31:26.000000000","tz":60},"subject":"Make CentOS 8 tinyipa job voting again","message":"Make CentOS 8 tinyipa job voting again\n\nFixing things in scripts and playbooks to make CentOS 8 tinyipa\njob vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to pc\n- installing package trousers to add tss user and group\nneeded by libvirt\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c3326addc1aa6dd453c4eac48d6578c107d5272a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c3326addc1aa6dd453c4eac48d6578c107d5272a"}]},"branch":"refs/heads/master"},"c31b3873d66b113a2a48eea8bbb67613fd444123":{"kind":"REWORK","_number":28,"created":"2020-03-05 09:57:27.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/703954/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/28"}}},"commit":{"parents":[{"commit":"74f8005978b2d61067c85c811cf3a2ab3b90baa9","subject":"Merge \"Sync pip_install files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/74f8005978b2d61067c85c811cf3a2ab3b90baa9"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-03-05 09:55:37.000000000","tz":60},"subject":"Preapare to switch Centos 8 job voting again","message":"Preapare to switch Centos 8 job voting again\n\nFixing things in scripts and playbooks in preparation of\nmaking CentOS 8 tinyipa job vote again.\n\n- python packages are not defined anymore\n- kvm_stats util has been moved to kernel-tools, the package\nqemu-kvm-tools doesn\u0027t exist anymore\n- change yum to dnf when needed\n- qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n/usr/libexec/qemu-kvm instead of /bin/qemu-system-x86_64\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n- qemu machine type pc-1.0 is outdated, change it to pc\n- installing package trousers to add tss user and group\nneeded by libvirt\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c31b3873d66b113a2a48eea8bbb67613fd444123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c31b3873d66b113a2a48eea8bbb67613fd444123"}]},"branch":"refs/heads/master"},"17c3ef21ece3953b54fa2521084eefa0b1861976":{"kind":"REWORK","_number":29,"created":"2020-03-06 13:12:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/54/703954/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/29"}}},"commit":{"parents":[{"commit":"74f8005978b2d61067c85c811cf3a2ab3b90baa9","subject":"Merge \"Sync pip_install files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/74f8005978b2d61067c85c811cf3a2ab3b90baa9"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 13:08:16.000000000","tz":0},"subject":"Prepare to switch CentOS 8 job voting again","message":"Prepare to switch CentOS 8 job voting again\n\nFixing things in scripts and playbooks in preparation of\nmaking CentOS 8 tinyipa job vote again.\n\n- python packages are not defined anymore\n- change yum to dnf when needed\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/17c3ef21ece3953b54fa2521084eefa0b1861976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/17c3ef21ece3953b54fa2521084eefa0b1861976"}]},"branch":"refs/heads/master"},"0c3baf04dd11a0017695150e3b7d53c63b4ebc94":{"kind":"REWORK","_number":30,"created":"2020-03-06 13:58:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/54/703954/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/703954/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/703954/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/703954/30"}}},"commit":{"parents":[{"commit":"74f8005978b2d61067c85c811cf3a2ab3b90baa9","subject":"Merge \"Sync pip_install files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/74f8005978b2d61067c85c811cf3a2ab3b90baa9"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-23 09:46:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 13:58:25.000000000","tz":0},"subject":"Prepare to switch CentOS 8 job voting again","message":"Prepare to switch CentOS 8 job voting again\n\nFixing things in scripts and playbooks in preparation of\nmaking CentOS 8 tinyipa job vote again.\n\n- python packages are not defined anymore\n- change yum to dnf when needed\n- change python2 packages to equal python3 when possible\n- ipxe binary has changed to ipxe-x86_64.efi\n\nChange-Id: I89989b6d2e35f0291ba1c1aed8b3317cfe2d691d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0c3baf04dd11a0017695150e3b7d53c63b4ebc94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0c3baf04dd11a0017695150e3b7d53c63b4ebc94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
