)]}'
{"id":"openstack%2Fbifrost~711815","triplet_id":"openstack%2Fbifrost~master~I940f522f87a42e20ee594be897b1f6853cef13e4","project":"openstack/bifrost","branch":"master","topic":"no-test-as-filter","hashtags":[],"change_id":"I940f522f87a42e20ee594be897b1f6853cef13e4","subject":"Do not use tests as filters","status":"MERGED","created":"2020-03-07 14:15:08.000000000","updated":"2020-03-09 15:38:40.000000000","submitted":"2020-03-09 15:36:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711815-1583768213708-9c168db8","meta_rev_id":"d69689762655125ffd1cfa40775173a917121c26","_number":711815,"virtual_id_number":711815,"owner":{"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 15:36:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-03-09 15:15:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-09 15:12:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2020-03-09 15:15:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 03:02:02.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-03-09 08:10:20.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-03-09 15:12:07.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-03-09 15:15:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-09 15:36:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b1ec2485c23751b37a173c8c17be38311e39b03","author":{"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},"date":"2020-03-07 14:15:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b24fe5e48447287543baf24540a9321bc7fda44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 14:33:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e985564e0c2c4146a795c312bb952993 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bf8e3644d874786bf875484150e7bc9 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b0a742db2174bb7a6e87b5e3d56a6b5 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff479dec97f24524be28f6db97eb84c4 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eeaef9e7559843d993c3b274870ae355 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a796e3fac54f828c8f1f3c3aa3a463 : SUCCESS in 7m 44s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88963661ae464a5ab0fec85e25fb49b7 : SUCCESS in 17m 36s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bacd1b3bdef47ce81524795e83c6ffb : FAILURE in 6m 19s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d4dcdccc1f544e7b3068e80787d8088 : FAILURE in 6m 12s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/673b6a5222244a998a1e98c528b23101 : FAILURE in 10m 05s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/e8911166398c4dbe815f266092dc3b92 : FAILURE in 4m 24s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/9fe399fbe24741cea60a735f0107a343 : FAILURE in 4m 17s (non-voting)\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/ddb082e825ef48f3b164fade08cbc42e : FAILURE in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3627739566d6f4a9363c331fc1cfc6b2ea368bb","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-03-09 03:02:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f1500af3ffe30b861d282ea5875f8a81fe61117","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-09 08:10:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44aa7522b99a186a2101fa7e106fb3cac92c2c5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 08:10:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05e329f266ae531d2d7ec43dcf01529ce9f52757","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 08:21:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6ada90be8a6446997ca06b5d5450790 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e88ba299edd341b28a47e5e65b988c69 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f37b67932284360b4eb5218a2dc691a : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd218dc7c75d4929bda84c5e7c3eb44f : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/765bc896095e4080a7722ce2f4a6f686 : SUCCESS in 6m 10s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f2d2dff3de94349b6f4a64f2a0636a7 : FAILURE in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1932e2e6f31bc93de45bae50a6219dd27f8e6193","author":{"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},"date":"2020-03-09 08:27:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93557c13c1c38214e8e3440665bf9ca108960e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 08:35:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4383d06616fa4ea8bbffb7d71e00eb0f : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7164fb4edcd84a39b1c794f9d17732cc : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/060405df5085482ea397c5ecdd512342 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8531eb95195472bae85d83db201cf42 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9822a2dd9d8349cd910764e02285d608 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a1a2696d3874b9584d4aa6709f0f987 : SUCCESS in 7m 00s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfb1a868c63a41029bdf41bfa2355f84 : FAILURE in 5m 41s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5eb2767d278343b2804d5411f76a0822 : FAILURE in 5m 40s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb4c5e630bf545fba013f9dcc0e4d82b : FAILURE in 5m 56s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dacdb6bd6f8e401884ff87f52a60e025 : FAILURE in 5m 39s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/c093d6fec9534107a03a9f79484a44af : FAILURE in 4m 26s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/2c18ca8a216b4cb7b9ba4020b326739a : FAILURE in 4m 21s (non-voting)\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/3b397ffd44fe4d538e368bf93e7109f8 : FAILURE in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c890e1411c5bb2185d4fb4fad5b840ff8b7ca0c5","author":{"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},"date":"2020-03-09 09:08:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ed5cc1092e87cfd5d207d4bfc5799af2d3a14ed3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-09 09:17:08.000000000","message":"Patch Set 1: Code-Review+2\n\nHit this recently, was going to push a patch but now don\u0027t need to. Thanks :)","accounts_in_message":[],"_revision_number":1},{"id":"37b123bed148b89656550e12880de2a53363533d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 09:19:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e2c27cd9359454a9e1fb7c4625a0026 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94193f7d8d3c41e9bc98ed1c7389ad0c : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61a65590553a463598b92b72eb83b24f : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fecbf3a7963450ba8a350a11a931567 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/214dcf9b8d774d998a5cb753f4e0cc45 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59910276d51a40ba84699adbd0e8819f : SUCCESS in 5m 44s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fc2fa36df804d96a2da7762429b8a85 : FAILURE in 5m 12s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb80e0109eaf41f79b8bb39374d56c0f : FAILURE in 7m 59s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c040cc8e19a445ebabb664f905c2f5d9 : FAILURE in 5m 37s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c54935717f254756bc81ad0709f88421 : FAILURE in 5m 36s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/5ebc7343c03b442f977ff1e0d62ba7df : FAILURE in 4m 09s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/3678f539a52f4ff0872afdff74dd1a13 : FAILURE in 3m 49s (non-voting)\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/95130108f4f147cfa99966f24819d8e0 : FAILURE in 3m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9134e9dc3d5ce64a98dceef4e19af5b06bb322c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-09 09:25:40.000000000","message":"Patch Set 1:\n\nLooks like it\u0027s failing on virtualbmc. There was a new release last week which could be the problem.","accounts_in_message":[],"_revision_number":1},{"id":"17ab62f1ffc2e6aae1cfa6bee0556e403784bd40","author":{"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},"date":"2020-03-09 09:52:24.000000000","message":"Patch Set 1:\n\n\u003e Looks like it\u0027s failing on virtualbmc. There was a new release last\n \u003e week which could be the problem.\n\nCheck job worked fine when I uploaded this change on Saturday and when I compare the logs of that run against this one, I see it complains about few missing files.\n\nLog of successful run on Saturday: https://bfb5eff6c274dcb9d46d-fe98a86d691c6300a22f04f47f232829.ssl.cf2.rackcdn.com/711815/1/check/bifrost-integration-tinyipa-ubuntu-bionic/8896366/job-output.txt\nLog of this run: https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_1fc/711815/1/check/bifrost-integration-tinyipa-ubuntu-bionic/1fc2fa3/job-output.txt\n\nHere is the snippet where it fails but I don\u0027t see any change in Bifrost repo since Friday so I am not sure what\u0027s wrong...\n\n\n2020-03-09 09:18:04.132576 | ubuntu-bionic | + cp \u0027/httpboot/ipxe.*\u0027 /home/zuul/logs/pxe/\n2020-03-09 09:18:04.134157 | ubuntu-bionic | cp: cannot stat \u0027/httpboot/ipxe.*\u0027: No such file or directory\n2020-03-09 09:18:04.134430 | ubuntu-bionic | + cp -aL /httpboot/pxelinux.cfg/ /home/zuul/logs/pxe/\n2020-03-09 09:18:04.135822 | ubuntu-bionic | cp: cannot stat \u0027/httpboot/pxelinux.cfg/\u0027: No such file or directory\n2020-03-09 09:18:04.136075 | ubuntu-bionic | + source /home/zuul/openrc bifrost\n2020-03-09 09:18:04.136198 | ubuntu-bionic | /home/zuul/src/opendev.org/openstack/bifrost/scripts/collect-test-info.sh: line 78: /home/zuul/openrc: No such file or directory\n2020-03-09 09:18:04.136671 | ubuntu-bionic | ++ openstack baremetal node list -c Name -f value\n2020-03-09 09:18:04.136841 | ubuntu-bionic | /home/zuul/src/opendev.org/openstack/bifrost/scripts/collect-test-info.sh: line 79: openstack: command not found","accounts_in_message":[],"_revision_number":1},{"id":"f6a078a6b75f9171b30e99b64fa1206707fb7509","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-09 10:02:12.000000000","message":"Patch Set 1:\n\n\u003e \u003e Looks like it\u0027s failing on virtualbmc. There was a new release\n \u003e last\n \u003e \u003e week which could be the problem.\n \u003e \n \u003e Check job worked fine when I uploaded this change on Saturday and\n \u003e when I compare the logs of that run against this one, I see it\n \u003e complains about few missing files.\n \u003e \n \u003e Log of successful run on Saturday: https://bfb5eff6c274dcb9d46d-fe98a86d691c6300a22f04f47f232829.ssl.cf2.rackcdn.com/711815/1/check/bifrost-integration-tinyipa-ubuntu-bionic/8896366/job-output.txt\n \u003e Log of this run: https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_1fc/711815/1/check/bifrost-integration-tinyipa-ubuntu-bionic/1fc2fa3/job-output.txt\n \u003e \n \u003e Here is the snippet where it fails but I don\u0027t see any change in\n \u003e Bifrost repo since Friday so I am not sure what\u0027s wrong...\n \u003e \n \u003e \n \u003e 2020-03-09 09:18:04.132576 | ubuntu-bionic | + cp \u0027/httpboot/ipxe.*\u0027\n \u003e /home/zuul/logs/pxe/\n \u003e 2020-03-09 09:18:04.134157 | ubuntu-bionic | cp: cannot stat\n \u003e \u0027/httpboot/ipxe.*\u0027: No such file or directory\n \u003e 2020-03-09 09:18:04.134430 | ubuntu-bionic | + cp -aL\n \u003e /httpboot/pxelinux.cfg/ /home/zuul/logs/pxe/\n \u003e 2020-03-09 09:18:04.135822 | ubuntu-bionic | cp: cannot stat\n \u003e \u0027/httpboot/pxelinux.cfg/\u0027: No such file or directory\n \u003e 2020-03-09 09:18:04.136075 | ubuntu-bionic | + source\n \u003e /home/zuul/openrc bifrost\n \u003e 2020-03-09 09:18:04.136198 | ubuntu-bionic | /home/zuul/src/opendev.org/openstack/bifrost/scripts/collect-test-info.sh:\n \u003e line 78: /home/zuul/openrc: No such file or directory\n \u003e 2020-03-09 09:18:04.136671 | ubuntu-bionic | ++ openstack baremetal\n \u003e node list -c Name -f value\n \u003e 2020-03-09 09:18:04.136841 | ubuntu-bionic | /home/zuul/src/opendev.org/openstack/bifrost/scripts/collect-test-info.sh:\n \u003e line 79: openstack: command not found\n\nIf you scroll up from there you can see the error about virtualbmc. That error is in the log gathering step after the main job.","accounts_in_message":[],"_revision_number":1},{"id":"346389271355d7f1f95d6e4422fcff3323d05dc7","author":{"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},"date":"2020-03-09 11:24:05.000000000","message":"Patch Set 1:\n\nnow I see it. Thanks for pointing that.\n\nI suppose the change below is an attempt to solve the vbmc issue\n\nhttps://review.opendev.org/#/c/711904/\n\nI\u0027ll wait for it then.","accounts_in_message":[],"_revision_number":1},{"id":"d653c171b47488cda98d0c1e88e8d0e2fbc1315b","author":{"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},"date":"2020-03-09 14:33:55.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b5bf968f8d9e3b58ed2d51dac0f2cfbeef3c3549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:04:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02c06c773e240f2bc41cd6c50b5a723 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46cb3c8b84d443639b9dfb0d3ce8a03d : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/548ef491326845e185e7a7d659a34d53 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba5ff6654a50459abe4c761857be67a4 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42ebe45a334d43388e9c84568e8f6924 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb8ac69ce2d497b8191ee3a0a650679 : SUCCESS in 6m 10s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69f88c995513436cb8e0fe32ddc116dd : SUCCESS in 16m 53s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdc45e836c5e484b842eae9168f3df1d : FAILURE in 5m 44s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23c05d52207a44d1809f3a7337dfbc93 : SUCCESS in 29m 51s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/703a61ee419e411d8ada08f9cee1c597 : FAILURE in 10m 25s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/8681c997225a4e88bc8bdc0ffb80a3d1 : FAILURE in 4m 25s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/e7bb772c953345d281bf8b19ae507383 : FAILURE in 3m 59s (non-voting)\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/e4e3a8a025ef4618bb8308fa1ee73b2a : FAILURE in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"41959c995db75d5766736c2c8c36dddfa5438bda","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-09 15:12:07.000000000","message":"Patch Set 2: Code-Review+2\n\nUgh, I don\u0027t like it :( But thank you for updating!","accounts_in_message":[],"_revision_number":2},{"id":"7fce96a8653822db48b0e6f09fee926206c09c8a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-09 15:15:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"11491aaddb889c537183b8e0e906a148e1e31988","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:15:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c06160c76e128ecd2062d2b0000b11eeb028e003","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:36:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d7f16177f7f08469a7f944c08ca3c40bd03e629a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:36:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13af7a2ef0be4ae284bf5f3a9d26beb1 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7202fd3e51b4a588433ca4c03074e5d : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14e1b734459d405ba49618e7451e6366 : SUCCESS in 6m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2be920b6d2904457a9bca8ad56f4916f : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08724dcb64c244d5a732a48496bcc184 : SUCCESS in 7m 19s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/338a5a9cfaf14358b0947e7414578a4f : SUCCESS in 19m 34s","accounts_in_message":[],"_revision_number":2},{"id":"7227c6592e94b1eb5022e3f6bb98cf563ff67442","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-09 15:38:20.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit ead314ed68c94992e9d3ff41dedb1a4735b9cf76","accounts_in_message":[],"_revision_number":2},{"id":"81fe4565b70d59b9172fd5ffb4eac3a940f1849a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:38:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3dd71a075847f798043b70b8c48387 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b072b6e3abc49474e943daad2fe8df6ab22ef5d7","revisions":{"4bd1bf6428ca885d45e041a015dcf2f77215d919":{"kind":"REWORK","_number":1,"created":"2020-03-07 14:15:08.000000000","uploader":{"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},"ref":"refs/changes/15/711815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/15/711815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/15/711815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/15/711815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/15/711815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/15/711815/1"}}},"commit":{"parents":[{"commit":"44dc37a6c27dc1d94753319a4ed7ec3946238f5a","subject":"Merge \"Fix \"unbound variable\" error in scripts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/44dc37a6c27dc1d94753319a4ed7ec3946238f5a"}]}],"author":{"name":"Fatih Degirmenci","email":"fdegir@gmail.com","date":"2020-03-07 14:12:43.000000000","tz":60},"committer":{"name":"Fatih Degirmenci","email":"fdegir@gmail.com","date":"2020-03-07 14:15:06.000000000","tz":60},"subject":"Do not use tests as filters","message":"Do not use tests as filters\n\nUsing tests as filters is deprecated and will be removed in Ansible\nversion 2.9 so this change changes \"result|search\" to \"result is search\"\nas recommended in the warning message.\n\nChange-Id: I940f522f87a42e20ee594be897b1f6853cef13e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4bd1bf6428ca885d45e041a015dcf2f77215d919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4bd1bf6428ca885d45e041a015dcf2f77215d919"}]},"branch":"refs/heads/master"},"b072b6e3abc49474e943daad2fe8df6ab22ef5d7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-09 14:33:55.000000000","uploader":{"_account_id":15362,"name":"fdegir","display_name":"Fatih Degirmenci","email":"fdegir@gmail.com","username":"fdegir"},"ref":"refs/changes/15/711815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/15/711815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/15/711815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/15/711815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/15/711815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/15/711815/2"}}},"commit":{"parents":[{"commit":"44dc37a6c27dc1d94753319a4ed7ec3946238f5a","subject":"Merge \"Fix \"unbound variable\" error in scripts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/44dc37a6c27dc1d94753319a4ed7ec3946238f5a"}]}],"author":{"name":"Fatih Degirmenci","email":"fdegir@gmail.com","date":"2020-03-07 14:12:43.000000000","tz":60},"committer":{"name":"Fatih Degirmenci","email":"fdegir@gmail.com","date":"2020-03-09 14:33:33.000000000","tz":60},"subject":"Do not use tests as filters","message":"Do not use tests as filters\n\nUsing tests as filters is deprecated and will be removed in Ansible\nversion 2.9 so this change changes \"result|search\" to \"result is search\"\nas recommended in the warning message.\n\nDepends-On: https://review.opendev.org/#/c/711904/\n\nChange-Id: I940f522f87a42e20ee594be897b1f6853cef13e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b072b6e3abc49474e943daad2fe8df6ab22ef5d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b072b6e3abc49474e943daad2fe8df6ab22ef5d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
