)]}'
{"id":"openstack%2Fbifrost~713443","triplet_id":"openstack%2Fbifrost~stable%2Ftrain~Ifd67adaa98e920f42c80d239349ae61d7d610f66","project":"openstack/bifrost","branch":"stable/train","hashtags":[],"change_id":"Ifd67adaa98e920f42c80d239349ae61d7d610f66","subject":"Ensure DIB is installed from source","status":"MERGED","created":"2020-03-17 13:31:23.000000000","updated":"2020-03-19 11:53:31.000000000","submitted":"2020-03-19 11:50:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713443-1584618632179-baf8bc82","meta_rev_id":"a4fc091b4b837794c995e7ca00458f61268b1b7a","_number":713443,"virtual_id_number":713443,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 11:50:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-03-18 17:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-19 11:17:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-19 11:17:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-18 17:28:16.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-03-19 11:17:31.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-03-19 11:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1602ae5cae373a1761298e9985a32252430b4fe6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-17 13:31:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cd8a0ac0cb3d149d1c122294595baa8bc794f21d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 14:30:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/703b8990e24841e98de1f9827cba2832 : SUCCESS in 8m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92035de975624f96aa171148ebb7649d : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e17fb1b4341477984120d8a9d74f8f5 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c2197c05bbc46e2a7b9ab5e7a0828d3 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe453ad3e684df294622db106bdf29d : SUCCESS in 8m 01s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3ef3afbe971a46139a778c36beb2a70a : SUCCESS in 22m 30s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/439bd129d3004d7a8dd70a710c199728 : SUCCESS in 27m 47s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/6da3f6a5cbfb426dacd78b5c62092ef9 : SUCCESS in 21m 51s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/07bd4e9e26bb4b1c9094014fe4179923 : SUCCESS in 24m 57s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/f7132b10993147fb8e567eb74f47de72 : FAILURE in 28m 48s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/e2e8211f75c141e0bb481e3e626e9154 : FAILURE in 33m 24s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/ab51029d6c5b4adfbe1a32ef6b8d3ab6 : SUCCESS in 44m 45s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a860c607381e46b4855b0ee3b41c3256 : SUCCESS in 37m 51s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/09d479a61e8d40478acc4053b1f42010 : SUCCESS in 40m 23s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/935c692f2b8b4aa8bd17e38dea4575b7 : FAILURE in 50m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6266668d73f844b49205ca57fe2ec77b : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"44b6921144a0c3986b7cd18b19abd398df063783","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-18 17:28:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"048b62e5210416d0f3584ae5df06bd9105afade3","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-19 11:17:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"efa72375330bc5627242508a073910655f3c7093","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:17:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfff5dea54cdbe0efbf020af44cc328688ba687c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:50:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94c9ee8b2aab40488a04b38b7e0635bc : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4da606aee4844733a328f364f7c35112 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b5b306609974261bdfb92028a813854 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02dcca98ff07403e8d91621fd6aa25bd : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae99e53eb084e3daf7ffe46d0846d2f : SUCCESS in 7m 26s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a038be03ab56439f93abdfdf39907190 : SUCCESS in 20m 57s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d337060551e2497ca872e510aec454fc : SUCCESS in 26m 15s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/88bde4e211e44db8bdcfb73c0db1dc61 : SUCCESS in 22m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84ee451c0d0a43aea9fd3ef4fae152d9 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3993aad18866d59346d4a13f50f0d9dd995d4c28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:50:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4fc091b4b837794c995e7ca00458f61268b1b7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:53:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/901fd6d36b0140cfa58f342c6ce0f7a1 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ed7d236dcf89364a257cab1a7b46ab776ed5f98","revisions":{"1ed7d236dcf89364a257cab1a7b46ab776ed5f98":{"kind":"REWORK","_number":1,"created":"2020-03-17 13:31:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/713443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/43/713443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/713443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/713443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/713443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/43/713443/1"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 16:19:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-17 13:31:23.000000000","tz":0},"subject":"Ensure DIB is installed from source","message":"Ensure DIB is installed from source\n\nBifrost clones the DIB repo and installs it. However, after running\ninstall.yaml pip freeze shows diskimage-builder 2.33 is installed.\n\nThis is caused by the recent addition of IPA builder, which has its own\ndependency on diskimage-builder.\n\nThis change fixes the issue by swapping the order of installation of IPA\nbuilder and DIB, to ensure DIB is installed last.\n\nChange-Id: Ifd67adaa98e920f42c80d239349ae61d7d610f66\nStory: 2007401\nTask: 38999\n(cherry picked from commit d878f65eea7493acb45e475a00ccd1d7ef1de50a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/1ed7d236dcf89364a257cab1a7b46ab776ed5f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/1ed7d236dcf89364a257cab1a7b46ab776ed5f98"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
