)]}'
{"id":"openstack%2Fbifrost~713508","triplet_id":"openstack%2Fbifrost~stable%2Ftrain~I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5","project":"openstack/bifrost","branch":"stable/train","topic":"centos8-train","hashtags":[],"change_id":"I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5","subject":"CentOS 8 support (Train)","status":"MERGED","created":"2020-03-17 18:23:03.000000000","updated":"2020-03-23 14:43:03.000000000","submitted":"2020-03-21 13:25:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713508-1584797118767-dfe5b5ea","meta_rev_id":"a06e06f657b50de1ca62bb9c698afd4255bd5c63","_number":713508,"virtual_id_number":713508,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-19 18:03:13.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-21 13:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-19 17:18:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-21 12:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-21 12:40:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 17:18:59.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-03-21 12:40:26.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-03-21 13:25:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fad48e9b3f13e91eccf6d497330509ce66c2a630","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-17 18:23:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06334842b5e92eee4a9163df90a8bd0847da216b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 18:23:18.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/bifrost on branch stable/train.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][1][\u0027bifrost-integration-tinyipa-centos-7\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      # templates:\n      #   - openstack-python-jobs\n      #   - openstack-python3-train-jobs\n      #   - check-requirements\n      #   - publish-openstack-docs-pti\n      #   - release-notes-jobs-python3\n      check:\n        jobs:\n          #- bifrost-integration-tinyipa-ubuntu-xenial\n          - bifrost-integration-tinyipa-centos-8\n          #- bifrost-integration-tinyipa-keystone-ubuntu-xenial\n          #- bifrost-integration-tinyipa-opensuse-15\n          #- Non-voting jobs\n          #- bifrost-integration-dhcp-ubuntu-xenial:\n          #-   voting: false\n          #- bifrost-integration-dhcp-opensuse-15:\n          #-   voting: false\n          #- bifrost-integration-dhcp-centos-7:\n          #-   voting: false\n          #- bifrost-integration-dhcp-centos-8:\n          #-   voting: false\n          #- bifrost-integration-dibipa-debian-centos-7:\n          #-   voting: false\n          #- bifrost-integration-dibipa-debian-centos-8:\n          #-   voting: false\n          #- bifrost-integration-dibipa-debian-ubuntu-xenial:\n          #-   voting: false\n          #- bifrost-integration-dibipa-debian-opensuse-15:\n          #-   voting: false\n          - bifrost-integration-tinyipa-centos-7:\n          -   voting: false\n          #- openstack-tox-lower-constraints\n      gate:\n        jobs:\n          - bifrost-integration-tinyipa-ubuntu-xenial\n          - bifrost-integration-tinyipa-centos-8\n          - bifrost-integration-tinyipa-keystone-ubuntu-xenial\n          - bifrost-integration-tinyipa-opensuse-15\n          - openstack-tox-lower-constraints\n\n  in \"openstack/bifrost/zuul.d/project.yaml@stable/train\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"5b2ae424fa4a43e3763aa5c819515650b1d006d7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-17 18:31:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"530ddfddbf1af3e62eb01786fc411fb600e9885d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 18:42:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/a2a10d6ad824447f8f33ce50e1800134 : FAILURE in 5m 34s\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/29c9fd3b4b944753aac2d6ad08e59937 : FAILURE in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"780d2553f51d3ed7e61f32be089163a590687599","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-17 18:51:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"344d31f4f99fb30f929739a0cc2c06296d6ca82b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 19:14:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/d47c63c716ec4283bfff8000c367b80d : SUCCESS in 19m 57s\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/4fa965d972574a8683b0affba14b2ffe : FAILURE in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4ce279dbc013b562f40c76c5547301e306c6531","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-17 19:27:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3811c5e0d495626e9063c77bd2e148f32924c844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:03:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/74d14a3699a64f98997c3ab2f4d2b5b2 : SUCCESS in 29m 15s\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/d006e71f9e474f40be1f249d537e9335 : FAILURE in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"09ad64314c56ff5c55b551c0d7d5ebff14831e9f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 10:09:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"78b38b9db7abca762d7f5be3f05316b2704f859d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 10:43:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/316e46f1597041a8818ef9b7bbc65870 : SUCCESS in 15m 53s\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/4f2a06b3419947788d78836fcbfb5877 : SUCCESS in 27m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a338ac159d2a928271e3f422ebdc87eb36a4f73c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 10:50:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d419f3b51d250576709a3b7dc00be1c208794d9b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 11:29:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"07869240cc6ff8b5f100d40237b17e5bc74ca4c4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 11:50:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"de5fa2539f17ce521884436d468f7ecf04dc68fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 12:37:14.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e365035759bc487292704571cf055974 : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da01cb6672754c9bb9ef51317435b88b : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4c15117eb884136bcc50b86fb794320 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/321fbcd13c96405ca1cda1590c6ea295 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96721bda475749ae8fe00b4becbb81f9 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7cd4ab02b32440c90aae0d993534550 : SUCCESS in 5m 23s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0d2f81b4eb1849d7a72c8cd8ebeb51fb : SUCCESS in 20m 22s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/0130934878f54320a1ce57d5a210ef1d : SUCCESS in 14m 25s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1fb07082696d4ccdbeaa2b3ee356211e : FAILURE in 9m 01s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/846e7f574a6e4b098fcdfb0cbad21b11 : FAILURE in 5m 21s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9b40e0f7c6a24447a7898ab19dae8bba : SUCCESS in 19m 27s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/deca69c266724e11be012576288a77cc : FAILURE in 4m 49s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/45f82a2d62444813a386f453051bb759 : FAILURE in 29m 29s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/1119fb85d1834133886240410f0c0ef7 : FAILURE in 8m 00s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/549d303525184b6bb5c69f3fc9796c30 : SUCCESS in 41m 17s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/fe8a063f29ac4cbbb80845bc58828795 : FAILURE in 14m 51s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/91421e068d0f4e349c66a6a8376cc317 : SUCCESS in 35m 15s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/37abc2960c8e45a6bc7aa46c9ac8607d : FAILURE in 3m 58s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/7cb38268ab0d4fd1afddc6990d19a40a : SUCCESS in 26m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb4ae93f48a649abbe7fffd8924e0482 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":8},{"id":"81d9286642e5933f0c0cb5b4f1555d040599193b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 12:44:15.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"93dbae8ff93b55832c373bcda6ea3db59d8918d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 13:07:39.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7634fa8ca7c4ac512e2e592fd9fb324bf732d603","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 13:10:03.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"5629f911c185804d2be1c02c579f7817b451bf7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 13:58:10.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/087cb0e24f8c433a82575edd2e1ad7d8 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a3f4e0628aa424cb04054da858d4b06 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe35c95986ce4a96b69f5719d2cbd5d5 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/613f2008fe7f4b7f91b98d5c6008d850 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a9a2b090c545de9f7ffe0c11f94d6b : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20d109d12cc44abfa934a37b1767e9fc : SUCCESS in 7m 48s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0279c769d1f04c4995f39f784d6b2ca4 : SUCCESS in 22m 31s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/b3ee267637784485ba246365e99b5885 : SUCCESS in 16m 02s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b21d4876b864474db6ab018e67e4b872 : SUCCESS in 26m 52s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/eab1ee979460448d8d58222c76ac39fd : SUCCESS in 24m 32s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/2418965aa5d845d4aa00f446714e9f45 : SUCCESS in 17m 38s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/4ba905b75dba449eb3c9cd610a0b1ec6 : FAILURE in 30m 12s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/76567ed03aeb4301bbbdfa61814694e0 : FAILURE in 30m 05s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/20ef62de29994244bdae41dba10bc7e5 : FAILURE in 18m 30s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/1f6c99f9799b4fa0bbbb76aaf8643a5a : SUCCESS in 38m 59s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/f88b9542db1c4306be788ae2c5e0ab38 : FAILURE in 18m 09s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e53636257d194299939d808546391ee3 : SUCCESS in 38m 10s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/14fbfbe7bc1541a6822f3c09f27efae3 : SUCCESS in 37m 54s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/a6ad6eb7d440484abbebc46aba4f6602 : SUCCESS in 23m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83fcf43da8574fe9ab5127dfb5f053ea : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":11},{"id":"b8a878a14c5daff06f8db6224fc0e24199632ff6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 16:46:08.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"4ac44d8665311f69b62803a13db9551724a40463","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 17:22:18.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ea17e82e6a17331f005e9d7587ec44a7dd1ca930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 18:22:25.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/485a860b98d243039b165e6fcd035c12 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89a2008cf70242329eb8b3a28420992a : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd19c3385b89470a9bbeceb6dbf28ff1 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ad888e0a5bf439aaf267ec5d123e09b : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd5d2e9a495e4324b455c5c69406e52c : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9f0454eef084b6ebd4c9fc487ad07b2 : SUCCESS in 4m 43s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5fe382cb3601445eb50610f7cafe5298 : SUCCESS in 32m 24s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/9cc2d8f4653e4fb7a6428be99b567ee4 : SUCCESS in 13m 10s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d01cc6d3cccf42c884108689b4f96694 : SUCCESS in 25m 13s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/81eded7a686b4459a8b49839f669cb92 : SUCCESS in 21m 01s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/42fc6c4f1c3b4ce5a9424a826e5bbe50 : FAILURE in 27m 23s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/827b5cd20119457c9ee7abd87a6439fb : FAILURE in 28m 42s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/1500fbc22bd84d16a814266c4fefd227 : FAILURE in 30m 50s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/16a716456c974d10b98f2a90d9ff69a2 : FAILURE in 15m 48s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/c718d8e765e0431ea3cb83420a615908 : SUCCESS in 44m 20s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f728c3b08a734922bc77cd191ab36a52 : SUCCESS in 36m 27s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/9c59e4ad64574172a6c0f23af7900a77 : SUCCESS in 37m 14s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/c937468d5aee4fb28169c1894ae09fa7 : SUCCESS in 21m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b48a8eb4cd8476893c0e8b8a761e9de : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":13},{"id":"28ee215f048a06a6a191dc1d566be4384a175ac2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 19:46:58.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"6df61993200c4db8d7cf16b7d7204f085e998152","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 20:32:36.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81ca696c210f4b7a85878aba17cd87e4 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab9d7333158b4783857647825e96e07d : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6a08920d25b4d4eb943879d41aee402 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55ee2a54ed1d43efb0b2089e0c73858c : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc6cef529f7c4e1995fd66ef7c3884c9 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c73a74a54d24e4b97d4a0071004b072 : SUCCESS in 4m 44s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9763f198899d440b83aae5151d601a1d : SUCCESS in 23m 35s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/951bf753d67046a4890b11c135002522 : SUCCESS in 16m 44s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/4ed9b44654964be299eebf8b574c8785 : SUCCESS in 26m 27s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/04767a8be3034e6ebfea309ce50f6cc9 : SUCCESS in 21m 07s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5c079cbd3a8042ee885fc53c846b069a : FAILURE in 27m 44s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/5379cef708f04ef995c99642e38dbef4 : FAILURE in 27m 23s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/8c880793c3c74b87add2bf7ddb20efec : FAILURE in 30m 33s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/768c3368419843a088cd93bead75ee95 : FAILURE in 15m 20s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/21088da9ee67466bb26418d09eeed2a0 : SUCCESS in 40m 45s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/21cbf54e0d044f02a0aa75dfeeb870e0 : SUCCESS in 37m 16s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/1b41c2bbbe3f4694a1fe350225ac88b1 : SUCCESS in 35m 27s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/aff40056b88b4752b0a76a2392abef96 : SUCCESS in 26m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9401ee61fed84d4381abef124a1d4684 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":14},{"id":"77a75a867d359c6d9927c5f5f318b5d9cf27fd62","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 11:00:40.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a882cfa68f06b4900b2612f1f3254837a4739e3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:07:27.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/467d093773da487993a8c8f73584fbab : FAILURE in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/721b6e668a394666b4d8c3204a17f2f5 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e6b604cf63c469e848b87281b5c7f5c : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2947415866344b3db40d1e6ed1837bfb : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06fe272093c94899919c7c5eb121a5eb : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b9368d891e145d5abab04351425b5d2 : SUCCESS in 5m 11s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c5940717958a458a821cec6836956288 : SUCCESS in 19m 04s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/62dd24b200594d27a6c8aac258fa4aec : SUCCESS in 16m 37s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/21b02f57f4da4e82b8e7fd55d191de5e : SUCCESS in 26m 15s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/fa47921ea2134d6e94221656c1e534b6 : SUCCESS in 20m 46s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d79728581c054d78a0bfd8899f4ccf01 : SUCCESS in 22m 08s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/c23f750beb9f4420879a3d8b00c9344d : SUCCESS in 25m 21s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/871b4ca53cc84813b7d90e558307e33c : SUCCESS in 23m 45s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/2d0bf1b1a1f34ff8a3d204d8dfc3332f : FAILURE in 15m 52s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/4607381834c240ddae14fc29e6dcab60 : SUCCESS in 37m 27s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f539cddb81564ddab1a80ac3bc3ac405 : SUCCESS in 35m 22s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/f6097ac888974fe9b3536d0bb52f441f : SUCCESS in 43m 35s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/c37d7cf1753244a49fc3e9bd2722a892 : SUCCESS in 26m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2f75dd605154923aa7a405e94816580 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":15},{"id":"4876d46f59bacbfef89ca803f248f86388b58e5f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 12:47:15.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"1e767cca83e39f0c9cbf4216669ad68606ca098b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 13:33:05.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aaf16e2b4274912a1fd94440e087e37 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff0f55d0afac464e9b0c3d1ab8358397 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8368173b14384e00b1cd3d4a31929c39 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d296c0a82b5c43cfadbdd7c20572ad37 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13576a3504514f40b7622de275d8b0dc : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/062a3c691a0b428a99b5c73c8186b58c : SUCCESS in 5m 09s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/40b2686c348f43f285a3cbcaa34378c9 : SUCCESS in 21m 34s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/00c4fa072d444c1a8d06f37b31c794ce : SUCCESS in 27m 54s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/fc5c4395cacf413cabe70e2115c6771d : SUCCESS in 23m 37s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/bb6fc238f6ca49488e56394ce333b95a : SUCCESS in 21m 45s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/041e5e8075764d74b8e0c7d8a51fab85 : SUCCESS in 22m 26s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/4f270dac8b46428dab2bbb425e37f277 : SUCCESS in 23m 52s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/297be14de03f4354b58b627f15e819d7 : SUCCESS in 33m 11s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/65ba5e269e454542a760e06a7cd6f0fa : SUCCESS in 31m 05s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/aa20f9169ccc4dadaa64d125d5bbb98b : SUCCESS in 39m 49s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/8c82538ab68f41c78b3e34c56a6e3471 : SUCCESS in 36m 31s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/3e5a1b0cf35f4138a5c571ffc1443c94 : SUCCESS in 37m 22s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/77d422b7d0b14ce284114ed81b6b95bd : SUCCESS in 26m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c828651f769b40f29ea7643244af28b7 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":16},{"id":"fcf4aca23490f4d6389a77f90627a659d06b4a29","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 13:58:50.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"a936ad03284995005313c0b30b2ff5fa00a9313b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-19 14:09:21.000000000","message":"Patch Set 17: Code-Review+2\n\nlooks great! thanks!","accounts_in_message":[],"_revision_number":17},{"id":"f22a4edac14120efff6fbf69994c63e3abd253bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:47:46.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b4c61ff9d049b695ca9fadbe1e219a : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16e9f4b837ab4a97afc9dceb0135225b : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9931c9c79b1a4fefa3a2c4480bebe31c : SUCCESS in 7m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/109159b6d1dd4eff91d375be1f506974 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4220233dd6e49f8862ce366f5adb87c : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63a7bf31a6354a5080f110e25ec61b6d : SUCCESS in 5m 41s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1f5a769d44474ccaaa8a551aff1a9334 : SUCCESS in 20m 55s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/29abed20dae7406bb109ff2f02005b52 : SUCCESS in 25m 31s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/548aba6cb24d43a1938210f45e6ceb92 : SUCCESS in 27m 30s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/8e72e72860934be9aaa2370c6aee9714 : SUCCESS in 21m 45s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1af1ccdf7f364e9e8fc817e70c459387 : SUCCESS in 24m 09s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/8cbb7813449348b88a224722bbdedd5a : SUCCESS in 27m 32s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/a24d6724d10f4aab8dbf3c7ee32c03c8 : SUCCESS in 33m 09s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/dce11c2f64f643678b8e073551e0833d : SUCCESS in 30m 56s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/f38240965017483fbbd5eedcce701ee2 : SUCCESS in 38m 03s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/285ed1e3bc414930a9ef17f95edebeae : SUCCESS in 38m 46s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/1cca22211fdf4d0180d25a94c63639a1 : SUCCESS in 34m 48s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/8471b067ca1947d3a3c734f19d93a206 : SUCCESS in 24m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66b4613e6af44a7d82cabda90e06d1f3 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":17},{"id":"0714c3a67c793c16ebec7c12dcc04afd49415bc8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 17:10:27.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"0b59bdf478b49c837fb64b823695c13911ee9fea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 17:11:13.000000000","message":"Patch Set 18:\n\nAdded iptables to package list. For some reason not required on master but missing in kolla image.","accounts_in_message":[],"_revision_number":18},{"id":"5ba58ed41895b6f2bfb6a13be06ef908596b9a3f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-19 17:18:59.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"83ad78ccceb11acfdc68f7236f902d52f297e78b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 18:03:13.000000000","message":"Patch Set 18:\n\nKolla job passing now :)","accounts_in_message":[],"_revision_number":18},{"id":"531f7461b1f20d055ba053490fc30de1b7d930d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:10:09.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a6e5af099844659c888452c1a35afe : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d746dacbec7a40b988df0697e49dca84 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cf337cefa7e4f689d9e05a15e819fd4 : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b58018783cfa4610a74d05c7ac880463 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe24d42c3ed49d9befad48f75637751 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a6c641b046d421d8159d394e3b6bf69 : SUCCESS in 7m 17s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/bbceeefc6c954b7384f6426b2b7e8c3d : SUCCESS in 24m 02s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/81c516cc8d6a45b7a7a9d9692dce066a : SUCCESS in 27m 31s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/cfe29dd0244d480ab2a7efb574fa211d : SUCCESS in 25m 25s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/1baec23902264a01a58f55082b9ef62a : SUCCESS in 24m 22s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/2aa7620aa5ea4ac39a2d19bda17ab398 : FAILURE in 29m 09s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/a3ab162cedef4f188c35c9ad2e5aae65 : SUCCESS in 22m 55s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/014037db00394b37817db697884dbae9 : FAILURE in 30m 38s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/18d62a11cfd9448cb4fa4ed376ee9c15 : SUCCESS in 31m 22s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/84af460cdfbd4f2fb4132d8cbc4ce752 : SUCCESS in 35m 39s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3cf2e81056f34fe1a9113dd35cdb7fe2 : SUCCESS in 36m 15s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/cd58dfcae68f4481876675ca75b6ba5d : SUCCESS in 40m 21s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/2360ffa407e94fbdab72a3afe41f013c : SUCCESS in 27m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4f27744b4ba44ae8838517419c989f8 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":18},{"id":"5c1dee54787c072fd870a1cf5ed4ad729df38e43","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-21 12:40:26.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"26b87f426602c79ef715732415ddb4b2764e8dbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 12:40:44.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"36cbcf4c3d06909a527f1e6a07da2d1bcc3e221e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 13:25:18.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0979b67832504dce94dcb0dab6c6a052 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d011d85967b64c58a2b7d4c4e25ac16f : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e8aa63b214345e0be2d043ce8d8f0a0 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4aef6c6baa634842a813edf49a12ec3e : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1887c9a6bd4f87aa65b616abfb557e : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e01521b44c24c93a9d6d7988bd448c1 : SUCCESS in 3m 57s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ef106dbd376a42e7a608a84a46eef932 : SUCCESS in 20m 03s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/faf7b05096094022b63c56dd443b6fb5 : SUCCESS in 22m 27s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/2c676c8d43c24fe6baf91e244b258908 : SUCCESS in 23m 45s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/f0feab5fd5194636b1e08b7397d1f9f0 : SUCCESS in 22m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfd447d2edbb455cb1f89ebc7bbadbe0 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":18},{"id":"84aeccd4958b3aaa9798af0255bfc7a9f0a5147f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 13:25:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18},{"id":"3464eb21fb33c40318d6e1660bc7c585aee8f40d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 13:27:17.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7bebc72644b4522add2a942681631d9 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57097a10c03f46ea9adefc0f23f5ea26 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":18},{"id":"e84e44afb635c9ae78dde4d45cc06bcad0ad4906","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-23 10:02:42.000000000","message":"Patch Set 18:\n\n(2 comments)\n\nThanks for reviews :)","accounts_in_message":[],"_revision_number":18},{"id":"64c962e301aa55e66f23621d072ab866539abb3d","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-23 12:00:55.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"33155ec0372408c590eba4d2ef725f6674f1b840","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-23 14:05:29.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"aee74941fefdb88405fd2cee7be278e2e72f1760","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-23 14:39:09.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"c4412abc3bf8bc114a30f3a8142b4c676e6de1b8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-23 14:40:43.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"05ade799c7322d1c667f5709f13661b0c5de2ec0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-23 14:43:03.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"d1ddbdf37b2c43dd86bb5aa40bf1b5dd247b33f8","revisions":{"b2b647a2e75daeda11f8a4596da7fb0520382c7f":{"kind":"REWORK","_number":1,"created":"2020-03-17 18:23:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/1"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-17 18:22:50.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b2b647a2e75daeda11f8a4596da7fb0520382c7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b2b647a2e75daeda11f8a4596da7fb0520382c7f"}]},"branch":"refs/heads/stable/train"},"df2263965a48029ecc0829594e7b8fafdc6df818":{"kind":"REWORK","_number":2,"created":"2020-03-17 18:31:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/2"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-17 18:31:26.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/df2263965a48029ecc0829594e7b8fafdc6df818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/df2263965a48029ecc0829594e7b8fafdc6df818"}]},"branch":"refs/heads/stable/train"},"bd4ce50feed4cf173c55a1d2501b03d55d4d493c":{"kind":"REWORK","_number":3,"created":"2020-03-17 18:51:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/3"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-17 18:51:34.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/bd4ce50feed4cf173c55a1d2501b03d55d4d493c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/bd4ce50feed4cf173c55a1d2501b03d55d4d493c"}]},"branch":"refs/heads/stable/train"},"d668412d5907a8c848ac2b69f069b4dedb6c99be":{"kind":"REWORK","_number":4,"created":"2020-03-17 19:27:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/4"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-17 19:27:39.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d668412d5907a8c848ac2b69f069b4dedb6c99be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d668412d5907a8c848ac2b69f069b4dedb6c99be"}]},"branch":"refs/heads/stable/train"},"de1de47232bc009faeb969190ab84eb12354e6ed":{"kind":"REWORK","_number":5,"created":"2020-03-18 10:09:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/5"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-18 10:09:39.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/de1de47232bc009faeb969190ab84eb12354e6ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/de1de47232bc009faeb969190ab84eb12354e6ed"}]},"branch":"refs/heads/stable/train"},"452e46d00fd1a610d7478e34feeb85721295eb62":{"kind":"REWORK","_number":6,"created":"2020-03-18 10:50:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/6"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-18 10:50:39.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/452e46d00fd1a610d7478e34feeb85721295eb62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/452e46d00fd1a610d7478e34feeb85721295eb62"}]},"branch":"refs/heads/stable/train"},"9507ac36b0bb03e11d924c60e2ce5f571896a00b":{"kind":"REWORK","_number":7,"created":"2020-03-18 11:29:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/7"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-18 11:29:32.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9507ac36b0bb03e11d924c60e2ce5f571896a00b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9507ac36b0bb03e11d924c60e2ce5f571896a00b"}]},"branch":"refs/heads/stable/train"},"c6ea08109aa320361ab02c38f081b2984b284afa":{"kind":"REWORK","_number":8,"created":"2020-03-18 11:50:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/8"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-18 11:50:43.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c6ea08109aa320361ab02c38f081b2984b284afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c6ea08109aa320361ab02c38f081b2984b284afa"}]},"branch":"refs/heads/stable/train"},"cc8c8e470e56d3e5adf0bf76dd66719e670bf90d":{"kind":"REWORK","_number":9,"created":"2020-03-18 12:44:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/9"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-18 12:40:46.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/cc8c8e470e56d3e5adf0bf76dd66719e670bf90d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/cc8c8e470e56d3e5adf0bf76dd66719e670bf90d"}]},"branch":"refs/heads/stable/train"},"e3781eaf43c1a69664cdcb0b2243421b9a8c220f":{"kind":"REWORK","_number":10,"created":"2020-03-18 13:07:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/10"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-18 13:07:29.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e3781eaf43c1a69664cdcb0b2243421b9a8c220f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e3781eaf43c1a69664cdcb0b2243421b9a8c220f"}]},"branch":"refs/heads/stable/train"},"99c7a179deacca2aaae2b71299559053add781cb":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-03-18 13:10:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/11"}}},"commit":{"parents":[{"commit":"9c5790f0bf4632974e449fc9117366d0142a1652","subject":"Run vbmcd as a Systemd service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c5790f0bf4632974e449fc9117366d0142a1652"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-18 13:09:53.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/99c7a179deacca2aaae2b71299559053add781cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/99c7a179deacca2aaae2b71299559053add781cb"}]},"branch":"refs/heads/stable/train"},"49ea409fffb36e8aa74a88f146df89f1f633fae6":{"kind":"REWORK","_number":12,"created":"2020-03-18 16:46:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/12"}}},"commit":{"parents":[{"commit":"9c5790f0bf4632974e449fc9117366d0142a1652","subject":"Run vbmcd as a Systemd service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c5790f0bf4632974e449fc9117366d0142a1652"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-18 16:45:59.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/49ea409fffb36e8aa74a88f146df89f1f633fae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/49ea409fffb36e8aa74a88f146df89f1f633fae6"}]},"branch":"refs/heads/stable/train"},"ba55da3291a499dc1dd366f3e2b530f35a812e14":{"kind":"REWORK","_number":13,"created":"2020-03-18 17:22:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/13"}}},"commit":{"parents":[{"commit":"9c5790f0bf4632974e449fc9117366d0142a1652","subject":"Run vbmcd as a Systemd service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c5790f0bf4632974e449fc9117366d0142a1652"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-18 17:21:58.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/ba55da3291a499dc1dd366f3e2b530f35a812e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/ba55da3291a499dc1dd366f3e2b530f35a812e14"}]},"branch":"refs/heads/stable/train"},"54c78e52b9da02959a1580a6a84f0293d1049ee7":{"kind":"REWORK","_number":14,"created":"2020-03-18 19:46:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/14"}}},"commit":{"parents":[{"commit":"9c5790f0bf4632974e449fc9117366d0142a1652","subject":"Run vbmcd as a Systemd service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c5790f0bf4632974e449fc9117366d0142a1652"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-18 19:46:44.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/54c78e52b9da02959a1580a6a84f0293d1049ee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/54c78e52b9da02959a1580a6a84f0293d1049ee7"}]},"branch":"refs/heads/stable/train"},"e4e84a905eba969129a4e018635f12fe6534842a":{"kind":"REWORK","_number":15,"created":"2020-03-19 11:00:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/15"}}},"commit":{"parents":[{"commit":"9c5790f0bf4632974e449fc9117366d0142a1652","subject":"Run vbmcd as a Systemd service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c5790f0bf4632974e449fc9117366d0142a1652"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-19 10:53:23.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e4e84a905eba969129a4e018635f12fe6534842a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e4e84a905eba969129a4e018635f12fe6534842a"}]},"branch":"refs/heads/stable/train"},"16faaee4fe6fcc566a1043f5fd7555fcc45b3f25":{"kind":"REWORK","_number":16,"created":"2020-03-19 12:47:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/16"}}},"commit":{"parents":[{"commit":"9c5790f0bf4632974e449fc9117366d0142a1652","subject":"Run vbmcd as a Systemd service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c5790f0bf4632974e449fc9117366d0142a1652"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-19 12:47:10.000000000","tz":0},"subject":"WIP: CentOS 8 support (Train)","message":"WIP: CentOS 8 support (Train)\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/16faaee4fe6fcc566a1043f5fd7555fcc45b3f25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/16faaee4fe6fcc566a1043f5fd7555fcc45b3f25"}]},"branch":"refs/heads/stable/train"},"9860dce5ffcf81bfc848cfe2406678469df341f7":{"kind":"REWORK","_number":17,"created":"2020-03-19 13:58:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/17"}}},"commit":{"parents":[{"commit":"9c5790f0bf4632974e449fc9117366d0142a1652","subject":"Run vbmcd as a Systemd service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c5790f0bf4632974e449fc9117366d0142a1652"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-19 13:56:11.000000000","tz":0},"subject":"CentOS 8 support (Train)","message":"CentOS 8 support (Train)\n\nThis is a selective backport of the CentOS 8 support from the master\nbranch. It is a combination of parts of various commits, with\nmodifications to continue to support CentOS 7 and Python 2.\n\n* Make bindep.txt CentOS version specific\n* kvm_stats util has been moved to kernel-tools, the package\n  qemu-kvm-tools doesn\u0027t exist anymore\n* qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n  /usr/libexec/qemu-kvm instead of\n  /bin/qemu-system-x86_64\n* installing package trousers to add tss user and group needed by\n  libvirt\n* Stop setting CPU mode - allow libvirt to select. host-model was\n  failing on libvirt 4.5.0 on CentOS 8 due to\n  https://bugzilla.redhat.com/show_bug.cgi?id\u003d1804224\n* Stop using unversioned Python executable - use ansible_python\n* Modify required packages for CentOS 8\n* Use Ansible 2.8 in CI\n* Switch from yum to dnf where necessary\n* Use a virtualenv for CentOS 8 CI\n* Add two CentOS 8 jobs\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9860dce5ffcf81bfc848cfe2406678469df341f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9860dce5ffcf81bfc848cfe2406678469df341f7"}]},"branch":"refs/heads/stable/train"},"d1ddbdf37b2c43dd86bb5aa40bf1b5dd247b33f8":{"kind":"REWORK","_number":18,"created":"2020-03-19 17:10:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/713508/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/08/713508/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/08/713508/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/08/713508/18"}}},"commit":{"parents":[{"commit":"9c5790f0bf4632974e449fc9117366d0142a1652","subject":"Run vbmcd as a Systemd service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c5790f0bf4632974e449fc9117366d0142a1652"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:02:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-19 17:10:24.000000000","tz":0},"subject":"CentOS 8 support (Train)","message":"CentOS 8 support (Train)\n\nThis is a selective backport of the CentOS 8 support from the master\nbranch. It is a combination of parts of various commits, with\nmodifications to continue to support CentOS 7 and Python 2.\n\n* Make bindep.txt CentOS version specific\n* kvm_stats util has been moved to kernel-tools, the package\n  qemu-kvm-tools doesn\u0027t exist anymore\n* qemu-system-x86 doesn\u0027t exist anymore, therefore we run\n  /usr/libexec/qemu-kvm instead of\n  /bin/qemu-system-x86_64\n* installing package trousers to add tss user and group needed by\n  libvirt\n* Stop setting CPU mode - allow libvirt to select. host-model was\n  failing on libvirt 4.5.0 on CentOS 8 due to\n  https://bugzilla.redhat.com/show_bug.cgi?id\u003d1804224\n* Stop using unversioned Python executable - use ansible_python\n* Modify required packages for CentOS 8\n* Use Ansible 2.8 in CI\n* Switch from yum to dnf where necessary\n* Use a virtualenv for CentOS 8 CI\n* Add two CentOS 8 jobs\n\nChange-Id: I9101a7d9c378b7fec9e6a94ff6e86e7d604680d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d1ddbdf37b2c43dd86bb5aa40bf1b5dd247b33f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d1ddbdf37b2c43dd86bb5aa40bf1b5dd247b33f8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
