)]}'
{"id":"openstack%2Fbifrost~713648","triplet_id":"openstack%2Fbifrost~stable%2Ftrain~I66e8e5f5a9c16ac90e132074c826d508458ae4d1","project":"openstack/bifrost","branch":"stable/train","hashtags":[],"change_id":"I66e8e5f5a9c16ac90e132074c826d508458ae4d1","subject":"Run vbmcd as a Systemd service","status":"MERGED","created":"2020-03-18 13:10:03.000000000","updated":"2020-03-18 14:30:27.000000000","submitted":"2020-03-18 14:28:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713648-1584541725920-cb471cbf","meta_rev_id":"b1ce3835d692dc693f9356263b6318410961e60a","_number":713648,"virtual_id_number":713648,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-18 14:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-03-18 13:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-18 13:11:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2020-03-18 13:12:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-18 13:11:13.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-03-18 13:12:57.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-03-18 14:28:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62108ca38924758bef5c01a27a23d9a3138b78dd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 13:10:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1842e39e00dd9ccc0aafdcb373d2cbd4833cdf2b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-18 13:11:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27c055293b8cb95d16ff88b65f06e542555d6b5b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-18 13:12:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7dc352aec1eacc34bb45219e69139e111f2adfda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 13:58:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3c47b2a501495e91278ac564cc2fde : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9d8875d1d054f94ad1e94bdacf2d9fc : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf9c7884a18544dc8a4d4f92440b8efc : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a6fbef4be5a4e4ca2582c652eec61c2 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f38a96de184aedb8451704be3cab9e : SUCCESS in 8m 36s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/08514ae472b742cc8c36c04ab1032a48 : SUCCESS in 27m 01s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f50cad1af5a146aa84dc3bac3d9a5e58 : SUCCESS in 22m 02s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/204c526fa9ec4201ae3c6e38e6c78fdd : SUCCESS in 22m 09s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/bdafff41aeb640b4b3057b162f64e3aa : FAILURE in 30m 09s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/2b4f5c47ae55482e8c5409e949d98e82 : FAILURE in 29m 35s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/a10de60769e240e8ab6c55dfbf3598d0 : FAILURE in 30m 25s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/893f1bf06a4841369f914db7cfdd2be8 : SUCCESS in 41m 04s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/fb24189086094ead9c57d109140e1148 : SUCCESS in 40m 09s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/2cd8495982804c5db1e91614621ce14e : SUCCESS in 37m 17s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/0e06b591a0a6412abc38aa6a06742d52 : SUCCESS in 31m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5ae06944f5e4231ba44ad19e03d5492 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5a94247cb9feddf95c0b461c1dc2028486ea9a27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 13:58:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"859cef783f1af2a29ef855451ac1fe53e3232bda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 14:28:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/271a5f299f814e219577d7e3981588e8 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3f6fae55f514d268b43f61eec6704c8 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d09216ab09964555ad672e037dda867b : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5645c9a82e7449f3b72816673eba7589 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3db6052aa54f2cb20d5d770440f5aa : SUCCESS in 7m 23s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/86290b83ec104ec08f3b66c62a0f7eef : SUCCESS in 23m 24s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/160f500f8e81442883dc235ad8933f5e : SUCCESS in 25m 13s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/f9307c31008b4fa7b3258f4e0e5f6ed0 : SUCCESS in 21m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72d45df7fc7c48d59ad75ab53d4b8485 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b6bc6b34da50d5f7f9967e28144926e33f156dd7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 14:28:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b1ce3835d692dc693f9356263b6318410961e60a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 14:30:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d162fe5a7414f3d9c9c80b08a9b15ff : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c5790f0bf4632974e449fc9117366d0142a1652","revisions":{"9c5790f0bf4632974e449fc9117366d0142a1652":{"kind":"REWORK","_number":1,"created":"2020-03-18 13:10:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/48/713648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/48/713648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/48/713648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/48/713648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/48/713648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/48/713648/1"}}},"commit":{"parents":[{"commit":"4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4eb5ba296d145ef3eee69fe9c0dc1713b4d41bc1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-09 09:26:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-18 13:09:53.000000000","tz":0},"subject":"Run vbmcd as a Systemd service","message":"Run vbmcd as a Systemd service\n\nVirtualbmc 2.0 dropped the daemon autostart feature. This means we need\nto explicitly start it. Running it as a Systemd unit, code adapted from\nTenks.\n\nIn order to allow Systemd to execute the daemon on systems with SELinux\nenabled, it must be installed to the system rather than a virtualenv.\n\nChange-Id: I66e8e5f5a9c16ac90e132074c826d508458ae4d1\n(cherry picked from commit 6513e5a6dc92dff06e7d5caeadf14db434b3648f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c5790f0bf4632974e449fc9117366d0142a1652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c5790f0bf4632974e449fc9117366d0142a1652"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
