)]}'
{"id":"openstack%2Fbifrost~733125","triplet_id":"openstack%2Fbifrost~stable%2Ftrain~Ibada78686ffb4db1d8dae05acf980081d5265fc6","project":"openstack/bifrost","branch":"stable/train","hashtags":[],"change_id":"Ibada78686ffb4db1d8dae05acf980081d5265fc6","subject":"CI: reduce memory and disk for testing nodes","status":"MERGED","created":"2020-06-03 12:29:04.000000000","updated":"2020-06-04 11:17:13.000000000","submitted":"2020-06-04 11:02:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733125-1591268538783-ef5efc7f","meta_rev_id":"267b8145dc88289bc63611bb1435704a3fb9c1bd","_number":733125,"virtual_id_number":733125,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 11:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"date":"2020-06-04 09:11:34.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-03 12:30:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-04 10:29:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 10:29:49.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-06-04 11:02:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"329e1ad45a306ddb38671fb1569828b6a37c4533","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-03 12:29:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"567c3c47d554b363e8ce1b32f47bacb864abff37","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-06-03 12:30:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9fe9e962908eeaeb659205641645ed26d79d0b79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:54:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad136839a1da4fe3a2d25c500584ec45 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/826c91b616284c4b864f121b6efd8811 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/567645fdf7ee4be7b1932a53e4ec0bb4 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b24ce64541c94587a8f96cf32adecf96 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f184deaa3fa74ca0b2bcf35963784849 : SUCCESS in 6m 31s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1fdb6b261b794906882d3af57af3ecdb : FAILURE in 6m 18s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/c2d5d92d0f314483aa4e88cb0a42670f : SUCCESS in 24m 35s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d9f80e50f16643ffb9fd0e8eea9e5e8f : FAILURE in 12m 11s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/2eea593b15ef42e3b46f58ff07292b18 : FAILURE in 9m 32s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/afc6f95577fe46c9b097e435b07093b9 : FAILURE in 8m 50s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/14daafabdce849c79df9967d2e995bb8 : FAILURE in 11m 02s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/e55991495ea64cc8bfb64f7bcabef0a0 : FAILURE in 9m 10s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/85aacf9c13864c77b380add096980624 : SUCCESS in 23m 18s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/02cd6e94e0c44d7aaf39add80a5c7427 : FAILURE in 10m 43s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/032d01985ee94a9fb62a59adaffd27df : FAILURE in 8m 59s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/88afc2e428ff4f3a910bca1ff0a57c0d : FAILURE in 8m 50s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/797564d30ba448408f4076a3ee386932 : FAILURE in 11m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65ab2dd43dd24798b93c6362ed4b5404 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"76784c7679aed344bf0640572888437a090ac9a5","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-04 09:11:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"707190b67402a32cfadca14dd136bc007e9eaaa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:00:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/333affdf312c46c4b57f89a8061e73ec : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8600fd28bc4a44f6b5815be3f5bacfb5 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1b88b4878754c3288c2195dc5258883 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/164a5ad1c0be4f05b470caa28f9e3d36 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a456708bafb140b08f97fbb6349f9d07 : SUCCESS in 7m 01s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c21865fd49dc494391ba70921923f693 : SUCCESS in 19m 34s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/d786cd09d6894835b806498e32191191 : SUCCESS in 25m 32s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5f6a76d3cccf4f60b1693824f064e545 : SUCCESS in 24m 34s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/c50f65fd23454ef2b0f82dfd3ad218cc : SUCCESS in 19m 16s\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d012b78d82aa4697bfc731e8249cfd0a : SUCCESS in 19m 58s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/93fbedd8d715483a9b8819bd5fce957a : SUCCESS in 25m 24s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/554a1e3e2114491f93fbe992697cc6b2 : SUCCESS in 30m 41s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/4caf00b7691d4f44abf0b54c62858334 : SUCCESS in 24m 53s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/8415e68b828f4ec08d1f6af4f5617c15 : SUCCESS in 41m 48s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c7a42d1a5a3a4fd58ee2f0a578fde504 : SUCCESS in 41m 58s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/5a2b0cf24b68429abe6ce3d92cdfe28a : SUCCESS in 31m 50s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/581d8b6c8d254b258a58507bab28eb57 : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c80a9a8e369e42ccbb0ad3b01d0c25a0 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e84b084d8f5f7119e2bdbca317e440f5897d91ee","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-06-04 10:29:49.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21f0d3d5d50b0cf3230df5caf0710d5b90faed30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:30:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a9fb09d6afb39664abde6a52bb8c912e7d454f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:02:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92de57ddf59c413396740c9f28962549 : SUCCESS in 6m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6854308b97dd404c9405e6d7737809c9 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a79c6600c9f4355baab19277510e765 : SUCCESS in 7m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e640e9f1ece4536b06ea7483b1944ed : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae6c6e27a734baba5588cd140a6dd33 : SUCCESS in 7m 51s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/659a1c5aac784406838bdd2881bafbf0 : SUCCESS in 22m 56s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/502b250dd8844512a7cdc23bda8a5fc9 : SUCCESS in 26m 30s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9fd8e220994045f1b8e9e1cedb349ce2 : SUCCESS in 26m 23s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/fdc2bf0b45424a34a695cbb6e4c9cedd : SUCCESS in 22m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3fdf90415274a089ed4caf9aabad6d4 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d71906fa5b345d915ea722af8d5f09c50303ce36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:02:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7a77761ecc784d177439262b27ec3998860179fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:04:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ca29fff73143d1b47e7783c2187ec0 : SUCCESS in 2m 08s","accounts_in_message":[],"_revision_number":1},{"id":"267b8145dc88289bc63611bb1435704a3fb9c1bd","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-04 11:17:13.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 00fbdfbe20b7100d0f131eeba379d0c9202ae97b","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"975b2b31032471386853977975422d244fe1e8e3","revisions":{"975b2b31032471386853977975422d244fe1e8e3":{"kind":"REWORK","_number":1,"created":"2020-06-03 12:29:04.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/25/733125/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/25/733125/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/25/733125/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/25/733125/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/25/733125/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/25/733125/1"}}},"commit":{"parents":[{"commit":"c6984da67a4b68315aac40ff9127172b9b51c34c","subject":"Merge \"CI: move the libvirt storage pool to /opt\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c6984da67a4b68315aac40ff9127172b9b51c34c"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-06-03 09:46:42.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-06-03 12:28:27.000000000","tz":120},"subject":"CI: reduce memory and disk for testing nodes","message":"CI: reduce memory and disk for testing nodes\n\nThere is not reason to use 10 or 12G of disk for Cirros.\n\nAlso change the RAM to 512M, although we\u0027ll have to revert it back\non master and ussuri once we switch to DIB images by default.\n\nConflicts:\n\tscripts/test-bifrost.sh\n\nChange-Id: Ibada78686ffb4db1d8dae05acf980081d5265fc6\n(cherry picked from commit 85e1b8207885d64daf986158dba8768ae49b257e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/975b2b31032471386853977975422d244fe1e8e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/975b2b31032471386853977975422d244fe1e8e3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
