)]}'
{"id":"openstack%2Fbifrost~733505","triplet_id":"openstack%2Fbifrost~stable%2Frocky~I378510457650e1502fb4c30b13601b222b289825","project":"openstack/bifrost","branch":"stable/rocky","topic":"use_dib_tag","hashtags":[],"change_id":"I378510457650e1502fb4c30b13601b222b289825","subject":"Use tag for dib","status":"MERGED","created":"2020-06-04 09:25:39.000000000","updated":"2020-06-04 12:16:43.000000000","submitted":"2020-06-04 12:12:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733505-1591272733117-71d89d23","meta_rev_id":"8f952438c73b8e6f51b656da5399eb068fb40b12","_number":733505,"virtual_id_number":733505,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 12:12:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-04 09:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-04 11:53:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-04 11:53:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 09:37:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-04 11:53:59.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-06-04 12:12:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"511ce07ddac53466dc5d92b8a2b2f6b196106ce9","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-06-04 09:25:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4edeab5dd624f2b2e12e7df4f2f6a49355d643a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 09:37:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3e8fe65d4610ed12e42ee2dfad69eb154886d08f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:54:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93b1e3c2fbca4d80b744fe8717c7c20c : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66b28b3281f748c59676e3f3d5a3eade : SUCCESS in 6m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5f9248309a78446fabf1e0e24a3fb3dd : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b8fc9c42d6047bfa3a6da0ea7067fe9 : SUCCESS in 4m 16s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/7ce012e0a2e24284bbb1195b03bb2fa5 : SUCCESS in 18m 14s\n- bifrost-integration-tinyipa-keystone-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ea1151c2242040ddb57726a715d8048d : FAILURE in 10m 23s (non-voting)\n- bifrost-integration-dhcp-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/569b58e26a5547f6a619aa7b35bca9f7 : SUCCESS in 17m 42s (non-voting)\n- bifrost-integration-dhcp-centos-7 https://zuul.opendev.org/t/openstack/build/b2516d3f0345437fba20fbe5281764fa : FAILURE in 9m 09s (non-voting)\n- bifrost-integration-dibipa-debian-centos-7 https://zuul.opendev.org/t/openstack/build/5dbcfa5edf7140a88e6bf9cc760af0fe : FAILURE in 7m 10s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/4e1f6e4cd2e44feea9322f1e1721a573 : FAILURE in 9m 57s (non-voting)\n- bifrost-integration-tinyipa-centos-7 https://zuul.opendev.org/t/openstack/build/3e8abb305f72476781e8e89282c7bf2e : FAILURE in 7m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a54d5296fca4146a54699b7be5baf68 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"bf4ffedf6b8cecf3743f20efab546c44579ab1c3","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-04 11:53:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0c6860e105217897d1050ae126ccbb17926e4c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:55:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ceaa0465cbbd9b55744e154ca3498496e5c9775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:12:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae9906d0dd9f420e8b32e73097063fcc : SUCCESS in 6m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/55169da854934d9faf9b02992b271a54 : SUCCESS in 6m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/72ba84fd0654487f81c0a53f8d1fb1b3 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e647573b947e4755b03088a0dc3d86ba : SUCCESS in 4m 18s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ef931d5ea21e4448a3955c5ffa51767f : SUCCESS in 16m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aeb353e9520e4d838accecd85877112d : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8ed656f7e1bedb4a5d2800a8d1ef2a96dd7e8144","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:12:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9110c4e470c5e18f4865205ab9447de5c16cd315","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:16:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18300166f78448b5857068da279d9c49 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de386d2234ee169f18e7f3d19472d1fedc687967","revisions":{"de386d2234ee169f18e7f3d19472d1fedc687967":{"kind":"REWORK","_number":1,"created":"2020-06-04 09:25:39.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/05/733505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/05/733505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/733505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/733505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/733505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/05/733505/1"}}},"commit":{"parents":[{"commit":"43e514fb0120b8e8844607539167b66d988d902c","subject":"Remove ip kernel command line argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/43e514fb0120b8e8844607539167b66d988d902c"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-06-03 13:29:05.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-06-04 09:23:25.000000000","tz":120},"subject":"Use tag for dib","message":"Use tag for dib\n\nStable/train and older branches have py27 support, since diskimage-builder\nremoved the support we can\u0027t relly on the master branch. Let\u0027s use the\ntag 2.38.0 since still provides py27 support.\n\nConflicts:\n        playbooks/roles/bifrost-prep-for-install/defaults/main.yml\n\nChange-Id: I378510457650e1502fb4c30b13601b222b289825\n(cherry picked from commit 9626f7a557c159a620d8ea2905b1995c22807cea)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/de386d2234ee169f18e7f3d19472d1fedc687967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/de386d2234ee169f18e7f3d19472d1fedc687967"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
