)]}'
{"id":"openstack%2Fbifrost~740224","triplet_id":"openstack%2Fbifrost~master~I588fe1cb625683efa6c1e4f21b6c84a098230954","project":"openstack/bifrost","branch":"master","topic":"git-url-root","hashtags":[],"change_id":"I588fe1cb625683efa6c1e4f21b6c84a098230954","subject":"Add git_url_root to override root for all repositories at once","status":"MERGED","created":"2020-07-09 10:24:04.000000000","updated":"2020-07-20 08:24:38.000000000","submitted":"2020-07-20 08:23:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740224-1595233380055-bfbdc47d","meta_rev_id":"4f89aef75fa929346bf21abee478d0ccc3b519ab","_number":740224,"virtual_id_number":740224,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-20 08:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-07-17 04:16:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-20 07:51:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-20 07:51:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-17 04:16:10.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-07-20 07:51:43.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-07-20 08:22:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"078cb3e8889573cb42a07df77e9ee56784ba91b0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-09 10:24:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a6f2615b435015b4ab53d3f7c28991976c8b958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 11:10:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0655783814f2495fbfd39b2c2a6965a9 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00b192d105f549a288a580985b8501fb : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/093ef271061d47ee8baa805b52e1950c : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc61368acfe64f458aa56938b80517e1 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00feb23020014cfabee18a2b9063ada1 : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e543331345344f41925adaff17ccf0a6 : SUCCESS in 5m 18s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29da2b7fc2414ffabd290234582015cb : SUCCESS in 18m 50s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/8a1dd0a630ea4bae84e15692a98b4198 : SUCCESS in 19m 07s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ed98299c52d4d61b3007dca3ac034c4 : SUCCESS in 20m 12s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/bcba19bd1fad4b6998843e1533ee3107 : SUCCESS in 16m 50s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/130a8c8913f549408c7075646db2828e : SUCCESS in 16m 11s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/34284d58390d4055bafd29bf8825c5d4 : SUCCESS in 13m 02s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/c16e01a2669d43f98103b272054b9163 : SUCCESS in 21m 57s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/b19e0cea9c344b36af763fd75dfbfcdd : SUCCESS in 11m 07s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aed179b6d3f2460093c23193c1bc3062 : SUCCESS in 17m 44s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/6db568ef0ea64038b70e8f6e370d50ea : SUCCESS in 24m 16s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1f7496ba352427ebfa15d5cda9b4b9d : SUCCESS in 33m 11s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/0fc38e4839764601970b6beefe7c8c12 : SUCCESS in 36m 43s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/092b0731f3d340e39d852218cb731de4 : SUCCESS in 19m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"851296c1487a966d9dae6180295c4409f498c2bd","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-17 04:16:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"53f6c4327a00672bc3a91774ceaca5e022759f35","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-07-20 07:51:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d6d1a626155b6f2d08f4cf7c2f2fc25bd5c78ba7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 07:51:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b56465366e245f12ebde478e343308820e6ea17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 08:22:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae64fbe8e4f42cd9d9956238b4f1bc5 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/834ae9d3cbcc47648231ba4eafce66d8 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62a12d93840746c39c4af1e016c271ba : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4d161e0aec341488c5a3ae34a0743c8 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8876c893f5da487cab4cd56ba73d35b9 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ac7f28d6f054e14894e5f8a08d25198 : SUCCESS in 4m 20s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/979b52d7fdf4460a9f228da4ca04f2e7 : SUCCESS in 15m 58s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/0343dfc5ccd0430187ac76430bbaba19 : SUCCESS in 18m 28s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8adfd9fca0164e3fb03d291d8171f405 : SUCCESS in 18m 09s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/dc97902fcaa84ba18aee1502da64e9e9 : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0009b199e8d2049460cf138ac11dfcfdc4132733","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 08:23:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e0e1a76825ba6696dc2240917e6326fef1b4586f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 08:24:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c90b8655bb4be1b152e20ec34c6c06 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69c4eadb4c4e4889b89da0a0362b4fc8 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e9249c5cb815ec8089ab13571c270dadc62aaa6","revisions":{"9e9249c5cb815ec8089ab13571c270dadc62aaa6":{"kind":"REWORK","_number":1,"created":"2020-07-09 10:24:04.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/24/740224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/24/740224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/24/740224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/24/740224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/24/740224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/24/740224/1"}}},"commit":{"parents":[{"commit":"ab69cca0cea475a4caa0351af3ad65e2ee226368","subject":"Get rid of ci_testing variable in roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/ab69cca0cea475a4caa0351af3ad65e2ee226368"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-09 10:23:03.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-09 10:23:03.000000000","tz":120},"subject":"Add git_url_root to override root for all repositories at once","message":"Add git_url_root to override root for all repositories at once\n\nChange-Id: I588fe1cb625683efa6c1e4f21b6c84a098230954\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9e9249c5cb815ec8089ab13571c270dadc62aaa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9e9249c5cb815ec8089ab13571c270dadc62aaa6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
