)]}'
{"id":"openstack%2Fbifrost~740246","triplet_id":"openstack%2Fbifrost~master~I99a3a202125f5ca805030a2363729aa81c8f5b4e","project":"openstack/bifrost","branch":"master","topic":"bifrost-cli-extra","hashtags":[],"change_id":"I99a3a202125f5ca805030a2363729aa81c8f5b4e","subject":"Support extra-vars in bifrost-cli","status":"MERGED","created":"2020-07-09 11:22:40.000000000","updated":"2020-07-21 21:57:51.000000000","submitted":"2020-07-21 21:56:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740246-1595368591787-78954b85","meta_rev_id":"6d693524eea918d4b0890ce710e492609b76837d","_number":740246,"virtual_id_number":740246,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-21 21:56:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-07-21 14:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-21 21:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-21 21:19:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-21 14:04:05.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-07-21 21:19:25.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-07-21 21:56:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4bec0c767f2f49ebbfef68976fcfb91356732753","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-09 11:22:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97273f797fe8bf42cef1e5a2375ad341d51756ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 11:40:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c34a2ffa07404b0ebf7e6cc757d4a63b : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e87a559d38a840b19f51b9c2bac0d74d : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c233f5013b3d40a185ecda659356cf4c : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03e5f16d39514be59ef18c22f606ac28 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7cb26fa1d094887ab426ef91b426221 : SUCCESS in 6m 32s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4885fba0a0fd4030933b4da3411f7b87 : FAILURE in 6m 03s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/b7239044850248909eef12818e2c54b1 : FAILURE in 7m 30s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80bd175ec4a04173a9baaeae501ed432 : FAILURE in 5m 14s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/b124345e60a24b7baa00dbee7d28737a : FAILURE in 7m 11s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fc0dedccaf148e0b5ad635a89c8aa2e : FAILURE in 5m 33s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/1424aeac3b0448e492ada01ca6a33968 : FAILURE in 5m 54s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/8afd6c2dd0bc469f83ea235dc4efebe8 : FAILURE in 6m 59s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/b4a034b7707448beb4fcc0dfb0ff8bc1 : FAILURE in 4m 15s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/845ef6cd4e1740cfaefed136cdb98b84 : FAILURE in 4m 48s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/93b749ef583742efa3686280708b5289 : FAILURE in 6m 07s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21929b2a22c2493982f5571ef75c6bcd : FAILURE in 6m 41s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/8cc80708f0f44917989a62e93729c467 : FAILURE in 7m 24s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5482ee4b90604fe5b604d6031cb21707 : FAILURE in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78551f8249061c0eeacf3c30fd192decdde0412a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-09 12:28:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4fc509ccdb0e3b9a23fdb7a5699b201e85c306be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 12:50:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b97998ccb60434e828ef6cf98c3941b : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e414c52ad1f746e5bfa8612bb2d24e43 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a61698c421ac4bd6905972b19300b133 : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38196fc836404db8a94bdb8ad967e5af : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c97ef01b91204a41a3dfbcfa70b9d42e : SUCCESS in 8m 24s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01a458fcf5e74090975dfd49bc65a791 : FAILURE in 5m 54s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/5703d4d8eb8843ad934a41cc3248da05 : FAILURE in 6m 59s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5d0d311fc2f4886849475d9b0370dfa : FAILURE in 5m 51s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/96d5d64dae7b4fa5ab869f680c49c013 : FAILURE in 8m 48s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c65aaadfaf184789aed44bdf373c9b2c : FAILURE in 4m 20s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/2d3ff5403d90478384639376f1fc0ebc : FAILURE in 4m 32s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/e0a2154754254d989a176a261de8b5f0 : FAILURE in 7m 27s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/c1673b7069954780b1d4d68e97eb379a : FAILURE in 5m 08s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b48fd0e74a214043af37b976a8b9260b : FAILURE in 10m 47s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/85329b14defa49ab9cec3c35ab9e02e3 : FAILURE in 5m 56s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c90612002b24473a6626df45b17f6e4 : FAILURE in 4m 58s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/28935e58998c46fc8380d67c96eef8a7 : FAILURE in 6m 56s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37732ffe1172475a89bbf0b0d8913bc4 : FAILURE in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d3fce71a96e7bb7e7b7dd38a835cc50837f1c673","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-20 09:55:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"10e0da4d4e302c3c3fb169bdf02c262ebc2604d1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-20 10:07:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"60c4b8f76ef60ed031b1dddf5be066cf9a8647b0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-20 10:09:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0b8aaa242bfdbbced7b0fa9087e0dbd17119250a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 10:55:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d29f02b020cf4d3fb9e7250e0a771dec : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72f37f03dac5408e9a8eb373d7329ecd : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a218f1a36d374b28af1ddd3c69b32a78 : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7316193462754e75916a0315d827fb59 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2451a576dc6473892b7c262fb901855 : SUCCESS in 6m 04s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95c8f8909038427f9e3876f471ad5b2c : SUCCESS in 19m 18s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/ad058031166145579e21728f685e0e0f : SUCCESS in 20m 41s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32b23ef233d240bfb63b2baad625bb0f : SUCCESS in 20m 33s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/b4ddcf676df24a049cf5f5e123d7d21e : SUCCESS in 26m 44s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9149f873494a41289d8f0a41541dc1ff : SUCCESS in 18m 02s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/f2f9347194614d8ea418a2da39bf818c : SUCCESS in 21m 07s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/bdf2987f1b8c4eaaba9ca93e9034ec90 : SUCCESS in 21m 49s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/5bcfd3b3d87240d480f6241fda90d25f : SUCCESS in 17m 25s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd1218c86fe9468488996e9aa5d818ce : SUCCESS in 22m 58s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/d9b81775defb4419aa4b9554fb2be5f9 : SUCCESS in 24m 22s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d365f5ac4d4648508665c2e0f1c97ecf : SUCCESS in 39m 58s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/9998c594e7f249bd8c8b472df61d9b92 : SUCCESS in 41m 49s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d7047df463845459c63026fda856945 : SUCCESS in 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cfd849e59d3d4477dbf641b1dde7f831a91d5d15","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-20 11:50:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"10ebd6c6886bcdb4c120f32510e6bb92dd650911","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 12:49:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72375623b93748aba989abd56a27ad7f : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e099fc8936b430eb81ef1b0d4d245a1 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/836812ddbd934d9795b16cf849ad7d9f : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca81988f24f64ffcbc2962d23a5622eb : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b8709b670b459dbc7ebddff92656f6 : SUCCESS in 7m 41s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddbc85d22b794a15a52dcca6ed2aeca3 : SUCCESS in 19m 01s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/2aa224fcacd2474f8c10fc58b8b1a584 : FAILURE in 35m 51s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eecb98350fd24a3998545cf037252c68 : SUCCESS in 21m 24s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/3516ad135384422bae9fa2f540b70d6a : FAILURE in 39m 10s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bd8102e16bd4cbb868636a3d1826225 : SUCCESS in 19m 35s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/fc61cb395beb4a648aec99d6b07b664b : SUCCESS in 19m 08s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/9158b261d15e4ff7a143500c69f818b7 : SUCCESS in 23m 01s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/3478832ac93c4fe394128713e0832fdc : SUCCESS in 17m 57s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2e3bedd97fe4c1983114d2e00c541ad : FAILURE in 56m 49s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/3896dc68f9d54804a69a981a333bd475 : SUCCESS in 20m 56s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/967a4675ae144e25a6ace3345b3c5929 : SUCCESS in 35m 33s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/5205d2ac8bfc40079489dc2983adadab : SUCCESS in 38m 47s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fe539c14bc24648ba5f8ae7a2f0dbd5 : SUCCESS in 26m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6b8ae34e93f54fd27919e02f2f030595f5869562","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-20 12:57:35.000000000","message":"Patch Set 6:\n\nrecheck timeouts","accounts_in_message":[],"_revision_number":6},{"id":"28b101da317b2aae354729b0d42b6841ad489f62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 13:43:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe18886f014a41f4b63e84c841ddfc53 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cdabe1cb54a4fbf82a3b20c95d1321d : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91e6a28576394df0bc479c23e8eb8560 : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6266b67a2eb94b7c818070cc8eb74e73 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc1edf52ded4c5d89d62ac03d0beaf9 : SUCCESS in 6m 43s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ab952b29e954ac5a51f5c155be9785c : SUCCESS in 17m 32s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/7567025e8ea34dc5ae58a2df98238b36 : SUCCESS in 22m 43s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93fac459e2464dd6bb628cd9fbe98c06 : SUCCESS in 20m 49s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/5abc1b978c25473bab8e5e5941485c8f : SUCCESS in 25m 29s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71e4bdf13e2a45a1bf724e10cda007cf : SUCCESS in 17m 18s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/0388d8c23f454c3eae6d1f46e4353610 : SUCCESS in 18m 48s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/fa136f3d8967433ab0b029cb20acfb3c : SUCCESS in 20m 12s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/5ee0f767c4fb46b6a7fe2f0d4c82348c : SUCCESS in 20m 50s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38d66171a94a4a69b57b3c2b69992489 : SUCCESS in 31m 25s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/f048dd1dc72c4321a24546864c0a67ec : SUCCESS in 25m 15s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a2ae408c0c24e1481c72e1d87a5a323 : SUCCESS in 36m 35s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/dc7821d36e4745c59161bea247b455b7 : SUCCESS in 44m 17s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8e82ac384a949d19b6e03428c4b0637 : SUCCESS in 18m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6bbc47eebde00ea3f13f3152b0a8343940ba1db5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-21 12:39:47.000000000","message":"Patch Set 6: Code-Review+2\n\nagreed on IRC to follow up with a release note for this","accounts_in_message":[],"_revision_number":6},{"id":"eb0cb732e694401f7c0e3a305b15c6d3f5677f3f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-21 12:50:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"34784b2959cd608c0c9967c00b59dc698c4833e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 13:35:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5683693faf741d49a29212580328aca : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d03175c01174a368d191e2535666d10 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cad0cf625e74e7aab93a61502ac7e4d : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9e5d525c01b4ec5b4a7762a8eca1ce2 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9348172949b04833aedc287e5d5ecdb9 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21501ed9e6134bb98683768c28484f70 : SUCCESS in 4m 47s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0811f289cbe48199f6499287fabc6f5 : SUCCESS in 19m 14s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/fc1c492432b2420d8cf06187e22d773a : SUCCESS in 29m 49s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fc6d37141744f378b6a5d835be3f56e : SUCCESS in 21m 45s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/09a5c792293c4e37907872a1a757a02a : SUCCESS in 26m 47s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c557b1247fd4681ad5acace7b132a5b : SUCCESS in 19m 25s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/d16ad3ba9e5f495bbe51674f8bf34d81 : SUCCESS in 22m 53s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/a0f977a7e9a043f0b3e3faed502bd791 : SUCCESS in 25m 17s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/c29aad055d8a4a4c8c27ebb6f9755680 : SUCCESS in 20m 48s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26e73727210c4e9f83150087d9e23d6a : SUCCESS in 21m 59s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/8ede4e24a8de43a4ac4d8d9875643e0c : SUCCESS in 26m 58s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0b7a43ba70444bdad7a6b76ee75635d : SUCCESS in 40m 17s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/619f9969d25b45219d833615d50cdaf5 : SUCCESS in 38m 53s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38457f10bc7d42dc92ac91cfe98904f2 : SUCCESS in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8ac4ae910f5c4ced25faefe107bd4ee4c184125f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-21 14:04:05.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"c3cabf18a9f36b392b60a67abf6513b467ca4e57","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-07-21 21:19:25.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"a249a773d23768ff94a65294aaa9b7f7bbcd0b58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 21:19:38.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b2595446b15ebc4922c1db9303c29935dba2d1f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 21:56:31.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ad2cd2fb044b36a6bb7b3ffa2f9343 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c63685dfabc148a18d9713db3c6d04d7 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/581f78709ac6461cb8e4421c42dbef45 : SUCCESS in 4m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07093f378d624b3ea7fcdec4695f2aa1 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b18b687fdcd94abfa7de97b0a34ded86 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b122fdc91014f1bb38d494c46dd97b8 : SUCCESS in 4m 13s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2114fb091f74b8caa5f87cec793208c : SUCCESS in 17m 40s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/bf0752de3abf429b8d58708f93678546 : SUCCESS in 23m 17s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b53931e889b4fa5a4c6d09883213db9 : SUCCESS in 21m 00s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/1c584c83d5394d40bbe97a497940eb8a : SUCCESS in 31m 04s","accounts_in_message":[],"_revision_number":7},{"id":"8dd1527630cadd225c98c459540b3552450dd43f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 21:56:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"5e4ecde432af33abf4b90f6d3880971877ce7f99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 21:57:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e02b674330b84a7b95c2be65108c9676 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bd0095be02541ada2c91c864cd924ef : SUCCESS in 58s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b67115d48c7d84ce7508a9cf33433286baedea91","revisions":{"b7069185a255074529b7a52c66bb7a1d840498aa":{"kind":"REWORK","_number":1,"created":"2020-07-09 11:22:40.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/46/740246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/46/740246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/46/740246/1"}}},"commit":{"parents":[{"commit":"9e9249c5cb815ec8089ab13571c270dadc62aaa6","subject":"Add git_url_root to override root for all repositories at once","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9e9249c5cb815ec8089ab13571c270dadc62aaa6"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-09 11:21:56.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-09 11:21:56.000000000","tz":120},"subject":"Support extra-vars in bifrost-cli","message":"Support extra-vars in bifrost-cli\n\nUse it to provide git_url_root in the CI\n\nChange-Id: I99a3a202125f5ca805030a2363729aa81c8f5b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b7069185a255074529b7a52c66bb7a1d840498aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b7069185a255074529b7a52c66bb7a1d840498aa"}]},"branch":"refs/heads/master"},"a3e17f634828e3820926432264f3fa6d6e65fbeb":{"kind":"REWORK","_number":2,"created":"2020-07-09 12:28:55.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/46/740246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/46/740246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/46/740246/2"}}},"commit":{"parents":[{"commit":"9e9249c5cb815ec8089ab13571c270dadc62aaa6","subject":"Add git_url_root to override root for all repositories at once","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9e9249c5cb815ec8089ab13571c270dadc62aaa6"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-09 11:21:56.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-09 12:28:37.000000000","tz":120},"subject":"Support extra-vars in bifrost-cli","message":"Support extra-vars in bifrost-cli\n\nUse it to provide git_url_root in the CI\n\nChange-Id: I99a3a202125f5ca805030a2363729aa81c8f5b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a3e17f634828e3820926432264f3fa6d6e65fbeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a3e17f634828e3820926432264f3fa6d6e65fbeb"}]},"branch":"refs/heads/master"},"a16d576467b96f9f103ea562163b1db586c6802d":{"kind":"REWORK","_number":3,"created":"2020-07-20 09:55:17.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/46/740246/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/46/740246/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/46/740246/3"}}},"commit":{"parents":[{"commit":"c77eb016b9557d9013da64af9c2f114b20e346b2","subject":"Merge \"Make skip_install a normal boolean variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c77eb016b9557d9013da64af9c2f114b20e346b2"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-09 11:21:56.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-20 09:54:46.000000000","tz":120},"subject":"Support extra-vars in bifrost-cli","message":"Support extra-vars in bifrost-cli\n\nUse it to provide git_url_root in the CI\n\nChange-Id: I99a3a202125f5ca805030a2363729aa81c8f5b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a16d576467b96f9f103ea562163b1db586c6802d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a16d576467b96f9f103ea562163b1db586c6802d"}]},"branch":"refs/heads/master"},"94309fc566978a660599b1a91f00e28dee0ad2da":{"kind":"REWORK","_number":4,"created":"2020-07-20 10:07:33.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/46/740246/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/46/740246/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/46/740246/4"}}},"commit":{"parents":[{"commit":"c77eb016b9557d9013da64af9c2f114b20e346b2","subject":"Merge \"Make skip_install a normal boolean variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c77eb016b9557d9013da64af9c2f114b20e346b2"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-09 11:21:56.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-20 10:07:04.000000000","tz":120},"subject":"Support extra-vars in bifrost-cli","message":"Support extra-vars in bifrost-cli\n\nUse it to provide git_url_root in the CI\n\nChange-Id: I99a3a202125f5ca805030a2363729aa81c8f5b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/94309fc566978a660599b1a91f00e28dee0ad2da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/94309fc566978a660599b1a91f00e28dee0ad2da"}]},"branch":"refs/heads/master"},"93ddfaa7ad2ea2b523697d18e130bb29116223e9":{"kind":"REWORK","_number":5,"created":"2020-07-20 10:09:51.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/46/740246/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/46/740246/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/46/740246/5"}}},"commit":{"parents":[{"commit":"be7f82e6c6f4197873bc36ba9752d6a17062615a","subject":"Merge \"Enable metadata cleaning by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/be7f82e6c6f4197873bc36ba9752d6a17062615a"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-09 11:21:56.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-20 10:09:30.000000000","tz":120},"subject":"Support extra-vars in bifrost-cli","message":"Support extra-vars in bifrost-cli\n\nUse it to provide git_url_root in the CI\n\nChange-Id: I99a3a202125f5ca805030a2363729aa81c8f5b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/93ddfaa7ad2ea2b523697d18e130bb29116223e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/93ddfaa7ad2ea2b523697d18e130bb29116223e9"}]},"branch":"refs/heads/master"},"0312b915e6d0cacda589912aee6a2891d8894f27":{"kind":"REWORK","_number":6,"created":"2020-07-20 11:50:56.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/46/740246/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/46/740246/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/46/740246/6"}}},"commit":{"parents":[{"commit":"be7f82e6c6f4197873bc36ba9752d6a17062615a","subject":"Merge \"Enable metadata cleaning by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/be7f82e6c6f4197873bc36ba9752d6a17062615a"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-09 11:21:56.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-20 11:50:26.000000000","tz":120},"subject":"Support extra-vars in bifrost-cli","message":"Support extra-vars in bifrost-cli\n\nUse it to provide git_url_root in the CI\n\nChange-Id: I99a3a202125f5ca805030a2363729aa81c8f5b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0312b915e6d0cacda589912aee6a2891d8894f27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0312b915e6d0cacda589912aee6a2891d8894f27"}]},"branch":"refs/heads/master"},"b67115d48c7d84ce7508a9cf33433286baedea91":{"kind":"REWORK","_number":7,"created":"2020-07-21 12:50:22.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/46/740246/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/46/740246/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/46/740246/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/46/740246/7"}}},"commit":{"parents":[{"commit":"be7f82e6c6f4197873bc36ba9752d6a17062615a","subject":"Merge \"Enable metadata cleaning by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/be7f82e6c6f4197873bc36ba9752d6a17062615a"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-09 11:21:56.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-21 12:50:02.000000000","tz":120},"subject":"Support extra-vars in bifrost-cli","message":"Support extra-vars in bifrost-cli\n\nUse it to provide git_url_root in the CI\n\nChange-Id: I99a3a202125f5ca805030a2363729aa81c8f5b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b67115d48c7d84ce7508a9cf33433286baedea91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b67115d48c7d84ce7508a9cf33433286baedea91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
