)]}'
{"id":"openstack%2Fbifrost~748054","triplet_id":"openstack%2Fbifrost~stable%2Fussuri~If617a305d4efc09335f675f1ec68e07cf81970c6","project":"openstack/bifrost","branch":"stable/ussuri","topic":"chmod-fix","hashtags":[],"change_id":"If617a305d4efc09335f675f1ec68e07cf81970c6","subject":"Explicitly set permissions on /httpboot contents","status":"MERGED","created":"2020-08-26 00:18:33.000000000","updated":"2020-08-26 10:37:35.000000000","submitted":"2020-08-26 10:34:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748054-1598438061749-1545f216","meta_rev_id":"b0d3b1b6db3cb1feecc1f8c90136798a98ec1d07","_number":748054,"virtual_id_number":748054,"owner":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 10:34:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-08-26 03:01:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-26 10:04:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-26 10:04:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 03:01:46.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-08-26 10:04:32.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-08-26 10:34:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd52508bfddf5d7d5d62fb3d944c3a8b41ea15cb","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2020-08-26 00:18:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"849d7795d434a46761ff604dd7fa8416bd599952","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2020-08-26 00:27:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"64c08727afb9b36237fe4074af3b5a54de6d3a2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 01:01:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfdc2e7549a0453197961474f77de602 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e50d2b5b0583453d8a1e69f823cd7f4e : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8787ae298c514eaeb43dde58da491ef4 : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/878429a5fa7b467b877aa10ee750c5ff : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e40e7ffb574123837abab017f5c591 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b1d3160a9de4010a0ea5930fefefa64 : SUCCESS in 3m 58s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bee1c411be8f44dea3d584e45cafc24a : SUCCESS in 15m 28s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/e50ff4587cb34e0485159807e69caa1c : SUCCESS in 13m 54s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b91b1e557ea54d55bdb85bc1eae622f6 : SUCCESS in 19m 13s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/6c9891a7c4684513ae605d30f604428a : SUCCESS in 17m 18s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7392bb4d2a9e419a91d1a371d8294df7 : SUCCESS in 15m 58s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/79c3fe2fdd834f84ab2750810b897728 : SUCCESS in 17m 38s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/811e60046f7b4e4b849fc9dc3a0000c1 : SUCCESS in 20m 48s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/c6145ba1782442fa92eb8295dbbef161 : SUCCESS in 23m 46s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1f90b8868a944f8a8da7886555f314f : FAILURE in 18m 58s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdcb2f6c737b4279b05f3750790641a5 : FAILURE in 29m 09s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/82bbd90a58db4ec0be04533459e93ea9 : FAILURE in 21m 09s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/ab7b1f21bc074d869e118a4a5d29b91c : FAILURE in 30m 22s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/2a502a26f48a40c880d040c4dde970c9 : FAILURE in 18m 03s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/9defc8384ea641fd83db7833806522d3 : FAILURE in 12m 01s (non-voting)\n- bifrost-integration-dhcp-debian-buster https://zuul.opendev.org/t/openstack/build/a8bf29beb2784633be13a6992c5ae092 : FAILURE in 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1070ff5b00b5ae01666bcb910e3f25d860a8ee14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 01:01:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfdc2e7549a0453197961474f77de602 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e50d2b5b0583453d8a1e69f823cd7f4e : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8787ae298c514eaeb43dde58da491ef4 : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/878429a5fa7b467b877aa10ee750c5ff : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e40e7ffb574123837abab017f5c591 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b1d3160a9de4010a0ea5930fefefa64 : SUCCESS in 3m 58s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bee1c411be8f44dea3d584e45cafc24a : SUCCESS in 15m 28s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/e50ff4587cb34e0485159807e69caa1c : SUCCESS in 13m 54s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b91b1e557ea54d55bdb85bc1eae622f6 : SUCCESS in 19m 13s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/6c9891a7c4684513ae605d30f604428a : SUCCESS in 17m 18s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7392bb4d2a9e419a91d1a371d8294df7 : SUCCESS in 15m 58s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/79c3fe2fdd834f84ab2750810b897728 : SUCCESS in 17m 38s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/811e60046f7b4e4b849fc9dc3a0000c1 : SUCCESS in 20m 48s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/c6145ba1782442fa92eb8295dbbef161 : SUCCESS in 23m 46s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1f90b8868a944f8a8da7886555f314f : FAILURE in 18m 58s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdcb2f6c737b4279b05f3750790641a5 : FAILURE in 29m 09s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/82bbd90a58db4ec0be04533459e93ea9 : FAILURE in 21m 09s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/ab7b1f21bc074d869e118a4a5d29b91c : FAILURE in 30m 22s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/2a502a26f48a40c880d040c4dde970c9 : FAILURE in 18m 03s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/9defc8384ea641fd83db7833806522d3 : FAILURE in 12m 01s (non-voting)\n- bifrost-integration-dhcp-debian-buster https://zuul.opendev.org/t/openstack/build/a8bf29beb2784633be13a6992c5ae092 : FAILURE in 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d87265e481b485d4a8921e753d6ad2bf6755847","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-08-26 03:01:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"871e0a1cd8cba0a260f8b6fe52be5eaa3d32605a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-26 10:04:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0edd06c94d3e671f6596b4326cb270a3c3948164","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:04:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9932a21d6327f7913ed454fa78f86016126067aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:34:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c664aecc1984d9ea107855ef6dcfc3d : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e7ca8af28964c9a99c0b21ccd496ae2 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/194da5e96beb468a99e6c6775aa254eb : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75f4362676e544f89f369886bc07eeed : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0452cc72537a403cb0b84aecc5f2141d : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbf158b89ba24c96b9dbf81a44f93e28 : SUCCESS in 3m 54s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1df4ffabefee46838de0967d16f4f10c : SUCCESS in 16m 40s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/72fe0ca787cb4f4c9e9e54de39d00760 : SUCCESS in 21m 01s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1d40c51f0cf497cabacd6629069f00c : SUCCESS in 21m 21s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/8eb2ef744af046958d5a23483859628e : SUCCESS in 23m 32s","accounts_in_message":[],"_revision_number":2},{"id":"368b35c26c038fa164435c8d0d3a739cdba5e3ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:34:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d61332db6cf9abdc59a8addfc48e0ab1b0ecb338","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:37:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d497b6ba7b41b3a79240c22f2085aa : SUCCESS in 2m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dd1e18a96704140972ff30d4aaaa60f : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5db92e3d705305648721e7470b09722c5f5d691f","revisions":{"290d9ba0922ef4e32ebfe540dcede5bad0324f0f":{"kind":"REWORK","_number":1,"created":"2020-08-26 00:18:33.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/54/748054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/748054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/748054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/748054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/748054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/748054/1"}}},"commit":{"parents":[{"commit":"334809957e7db31dcc93ca2c4676f2f051717751","subject":"Explicitly enable DHCP services on baremetal CentOS/RHEL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/334809957e7db31dcc93ca2c4676f2f051717751"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2020-08-25 04:10:31.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2020-08-26 00:15:49.000000000","tz":600},"subject":"Explicitly set permissions on /httpboot contents","message":"Explicitly set permissions on /httpboot contents\n\nAnsible versions 2.9.12 and 2.8.14 change default behavior\nin regards to applying umask to ansible-created files. Due to\nthis, newly created files may have overly restrictive permissions,\ncausing issues in use cases where files need to be world-readable\nsuch as contents of /httpboot folder in Ironic. This patch adds\nexplicit setting of permissions to ensure Ironic network boot\ncontinues to work correctly.\n\nChange-Id: If617a305d4efc09335f675f1ec68e07cf81970c6\n(cherry picked from commit 1f37c41413)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/290d9ba0922ef4e32ebfe540dcede5bad0324f0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/290d9ba0922ef4e32ebfe540dcede5bad0324f0f"}]},"branch":"refs/heads/stable/ussuri"},"5db92e3d705305648721e7470b09722c5f5d691f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-26 00:27:17.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/54/748054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/748054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/748054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/748054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/748054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/748054/2"}}},"commit":{"parents":[{"commit":"334809957e7db31dcc93ca2c4676f2f051717751","subject":"Explicitly enable DHCP services on baremetal CentOS/RHEL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/334809957e7db31dcc93ca2c4676f2f051717751"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2020-08-25 04:10:31.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2020-08-26 00:26:42.000000000","tz":600},"subject":"Explicitly set permissions on /httpboot contents","message":"Explicitly set permissions on /httpboot contents\n\nAnsible versions 2.9.12 and 2.8.14 change default behavior\nin regards to applying umask to ansible-created files. Due to\nthis, newly created files may have overly restrictive permissions,\ncausing issues in use cases where files need to be world-readable\nsuch as contents of /httpboot folder in Ironic. This patch adds\nexplicit setting of permissions to ensure Ironic network boot\ncontinues to work correctly.\n\nChange-Id: If617a305d4efc09335f675f1ec68e07cf81970c6\n(cherry picked from commit 1f37c414130d58e64ba6d9d2fc9cc9f7e690453f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/5db92e3d705305648721e7470b09722c5f5d691f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/5db92e3d705305648721e7470b09722c5f5d691f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
