)]}'
{"id":"openstack%2Fbifrost~748373","triplet_id":"openstack%2Fbifrost~stable%2Fussuri~If89f1506ddd345df217739bf5dbb15d18b44c096","project":"openstack/bifrost","branch":"stable/ussuri","topic":"tftp-context-stable/ussuri","hashtags":[],"change_id":"If89f1506ddd345df217739bf5dbb15d18b44c096","subject":"Add correct SELinux context for /tftpboot and fix map-file ACL","status":"MERGED","created":"2020-08-27 08:34:08.000000000","updated":"2020-08-28 13:15:21.000000000","submitted":"2020-08-28 13:13:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748373-1598620388395-9749b533","meta_rev_id":"9a7e4bd68ea05479266a794574a3a544e31126d8","_number":748373,"virtual_id_number":748373,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 13:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-27 15:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2020-08-28 12:38:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 15:46:39.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-08-28 12:38:34.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-08-28 13:13:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bfc4b55ec9aecc44ed5e38c1b6d6fc973a4787b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-27 08:34:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"76ac2f1096013af2ac42df08171351c2243ee342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 10:02:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae851833bbff4174a2f31d90a3317e13 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0aab126d09a14eaa9d9548024adb41e2 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36abfa9e17ca496eb8ee368209160a70 : SUCCESS in 7m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bd98bbbe23c4d42b710e02cd0c8fdfe : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/693b0db6fa20479bb9f4afd4ca0e095a : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bdd2f77e777458fb55fba807c210717 : SUCCESS in 5m 55s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b4a2496d986496b82302634b8de3fac : SUCCESS in 17m 06s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/7f3b04f37b0842b4be99c307a8b4c712 : SUCCESS in 19m 42s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff5e10e472f84351ae21d3e1950a2a83 : SUCCESS in 20m 24s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/95d37d4ca8bb4666af6665a98e9983d4 : SUCCESS in 24m 35s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10197629f3fa4e60a3a394b784bd645c : SUCCESS in 13m 10s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/67d1855582f64ffa8410c3a653f90712 : SUCCESS in 16m 51s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/8390c45cad4245f681c6d2225b001644 : SUCCESS in 20m 21s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/aabfc85a947546d3ab80466b24f0ebc4 : SUCCESS in 20m 51s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b31addd7fc21451598db8f65271c3010 : FAILURE in 26m 06s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27d6ff21df8c44fa9c56f6feb189a806 : FAILURE in 27m 14s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/739726119f8244b1b2bb62445b8aa1d5 : FAILURE in 21m 25s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/c93009a20d0842e390dd94460cce6bdc : FAILURE in 33m 08s (non-voting)\n- bifrost-integration-dhcp-opensuse-15 https://zuul.opendev.org/t/openstack/build/fa8577cd2dd649a990931f7ff57be91f : FAILURE in 19m 34s (non-voting)\n- bifrost-integration-dibipa-debian-opensuse-15 https://zuul.opendev.org/t/openstack/build/211629af295145bc8a5ed25868568d3d : FAILURE in 12m 33s (non-voting)\n- bifrost-integration-dhcp-debian-buster https://zuul.opendev.org/t/openstack/build/69f245a5158c4ad79578c6e121a46bc1 : FAILURE in 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e05ce9d3e995c5d29a337d80219de90d3bc58289","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-08-27 15:46:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc098beadacafa0011e3c6382685c9be645bb186","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-28 12:38:34.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e7afa2c80bf3fc228628dad9e17a4ebd210f536","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 12:38:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c151ca853806f9973029a813303b7978edb41c20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 13:13:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aa4663dd8614d3c88c641141cfe7c2a : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e521330689594835b4206593d0ff468e : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38e34cc4b55848ebbf27c93e9004fd44 : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21e13684d4b14e63b153b6744b204dcd : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7696fbe7ae104261a1a7e59c936e33b7 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea75f72b5e4145669ff62e3cafedea88 : SUCCESS in 5m 30s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b3a983760864b7b8348e1d2a70cb324 : SUCCESS in 18m 01s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/e6bd7dfffd83423fa0b7877c5d882d08 : SUCCESS in 21m 35s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a60511f4ecc4c9fb1da9de83dcfbd36 : SUCCESS in 24m 08s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/3b7de36901d842fcb9f62ebac78fcea5 : SUCCESS in 29m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ee4be5c407586484c1da2742ea62ffdb69d88cd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 13:13:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a6627d2ea0d0f00b77818cd204e35598fa61544e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 13:15:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1608db41f44743ecb7f22accaafd5e89 : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e3fa83494a94977ac02389f865a4a5d : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49de1f951c2f4c115a767687460ae012779df252","revisions":{"49de1f951c2f4c115a767687460ae012779df252":{"kind":"REWORK","_number":1,"created":"2020-08-27 08:34:08.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/73/748373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/73/748373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/73/748373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/73/748373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/73/748373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/73/748373/1"}}},"commit":{"parents":[{"commit":"5db92e3d705305648721e7470b09722c5f5d691f","subject":"Explicitly set permissions on /httpboot contents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/5db92e3d705305648721e7470b09722c5f5d691f"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-26 15:30:27.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-27 08:34:08.000000000","tz":0},"subject":"Add correct SELinux context for /tftpboot and fix map-file ACL","message":"Add correct SELinux context for /tftpboot and fix map-file ACL\n\nChange-Id: If89f1506ddd345df217739bf5dbb15d18b44c096\n(cherry picked from commit 3aacfd32260d8b916e358e6aec298861d1b480b5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/49de1f951c2f4c115a767687460ae012779df252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/49de1f951c2f4c115a767687460ae012779df252"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
