)]}'
{"id":"openstack%2Fbifrost~748597","triplet_id":"openstack%2Fbifrost~master~I3293a554ad4af76b47c73fe820e1809f561aefef","project":"openstack/bifrost","branch":"master","topic":"no-empty-check","hashtags":[],"change_id":"I3293a554ad4af76b47c73fe820e1809f561aefef","subject":"ansible-lint: fix error 602","status":"MERGED","created":"2020-08-28 10:05:35.000000000","updated":"2020-09-01 09:15:41.000000000","submitted":"2020-09-01 09:12:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748597-1598951536402-f4e8b138","meta_rev_id":"5ad60eee6a8fc4532db7ccd425c9312a6f631004","_number":748597,"virtual_id_number":748597,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 09:12:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2020-09-01 07:31:53.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-01 08:23:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-01 08:23:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 07:31:53.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-09-01 08:23:54.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-09-01 09:12:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67b357026037a0f331f81ee642f934a9eff23a6b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-28 10:05:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"055170274ded4a7dbdbbbb5cdddea698e5159c0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 10:20:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962939fc9c4048fd803cd6597d0e3ff7 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04de52cf9cbb4f17ac7da297c14a61ad : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0647853de4046d39450da0d2faa8c28 : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afce99a0b4944d9c922438c82d265843 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cbe19b6a133462c9e6c0a01752010e2 : SUCCESS in 6m 44s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/b3afde64101d4b838afeac6bcf8410e3 : SUCCESS in 6m 05s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6802747bf6f49f8b0bf6a6b7852c03a : FAILURE in 6m 02s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/7ca686eec3d24926b4abccd2f209f202 : FAILURE in 9m 30s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5e854c30dba4533b93719b5d21f074b : FAILURE in 5m 11s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/582ebd0d053d42fa9e409ce4624f633f : FAILURE in 6m 03s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd4c6c7c1c604cb39972dda8bc2857bb : FAILURE in 5m 27s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/565488fc02d748f5b94f8786a4467307 : FAILURE in 8m 36s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/e7c57fcf25e44b89b11aaf7393abe13f : FAILURE in 8m 23s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/7860b594b28447bc8edc502fc051243b : FAILURE in 6m 41s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adc69fe62f6c4420a65a26307f869d8a : FAILURE in 5m 16s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/cc72026abf5f44158142fb08acb0e22d : FAILURE in 5m 57s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/020f655a2f9148fb9285e15b7b7fb035 : FAILURE in 5m 04s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/29ed20ba26f04444aa91a12e3d352303 : FAILURE in 6m 56s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff52aa3ab77f47bd8482275f0dd5579f : FAILURE in 5m 18s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/71d20ae152f34dd6843700d411fc08ee : FAILURE in 7m 26s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/b4f91ed299254d8e9ae166f565f641f1 : FAILURE in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02c337e9f1150104dc899cd6e07422bbfaf8f68c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-28 11:39:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b86e545b12939ae57c0ba78557a971c1c7761d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 12:22:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4d0695f9cdd436f9432ca7b7e2d35fc : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8b71c1fb1554f21b2ab807ff70a321a : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/def5219679934679bb1dc8b733728943 : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/517f8ee77b764fccab7bd7a6e924a57f : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cedcd7fcc754c818aa35ccfb06b75fb : SUCCESS in 5m 43s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/2033445308944b16928d1ac18f1311f8 : SUCCESS in 5m 19s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a50c42d150054db0b2f108ccbab4b368 : FAILURE in 8m 02s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/1eeb446f4b6b48399cde33e6c70c6697 : FAILURE in 11m 26s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aed9ce9cf08d4c05a6069dfc2b1252d9 : FAILURE in 10m 52s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/50749387ac404142ab6d954095070fb6 : FAILURE in 15m 48s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcf5d4dbb0f94daf86292ecd24fc08ce : FAILURE in 9m 45s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/89e8552ee47849c8ad2f4771f23c5e9f : FAILURE in 10m 38s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/f71c009f8dcd419580b88851158ad8b9 : FAILURE in 11m 43s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/a7bd04b0ecbc4333975c6aa70a6b3f85 : FAILURE in 14m 37s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7dac7e7f7b244d55bd334443a0a0e383 : FAILURE in 9m 16s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/3a3f4d0d70aa4371bbbb4e75960e46c9 : FAILURE in 9m 00s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d587f4272fd4b508731b4f53cc23680 : FAILURE in 27m 35s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/ea08169d94534fcfb98b9aeb5bbe184f : FAILURE in 25m 37s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc601339f1114f2b8fe4a5a2e177af6c : FAILURE in 9m 57s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/7fbaa79a0b854aa0a041b6186fdc5f2d : FAILURE in 11m 45s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/4c413e85d35246048b73f6fb11862f2a : FAILURE in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9456b114f21e2032eaba59970b1539ce26976f18","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-28 12:27:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1194ec4017156b9dad5f8fd21d4734855ae886f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 13:14:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52a5843eae846a5ab5d55a292d0d964 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fd0fc292e3b4a2481c20e4fd55a33cf : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68317d0f92b24439b605a87839b8e785 : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59d3b3e3b33a4313aba7ce42efd25502 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a140b398edf478ea4672ad0a97bd698 : SUCCESS in 6m 51s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/d39b0d761bac4d9cbdb4049fe1086fe0 : SUCCESS in 6m 26s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2808924f3f74ebc80f8e592edfcf13b : FAILURE in 9m 44s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/22be0f3d133f4c7090ab1babd72d3425 : FAILURE in 11m 18s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77f2f4ff2efe40ee8989a683c5e376ec : FAILURE in 11m 41s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/c2d419802feb4b8f8636b1aa11f8c487 : FAILURE in 15m 26s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e980b1f2d2fc4b57ab98c53df470609d : FAILURE in 8m 07s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/c60b167245094dafabc1cb0d40556923 : FAILURE in 11m 31s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/e9e8b7997ed34a6b97b92b62cff83dbe : FAILURE in 13m 56s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/011265267bff436a86155c1da2c8a013 : FAILURE in 9m 51s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33038f7129ef44b990a97d74ef928692 : FAILURE in 10m 12s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/594f8c596cc64f029f4666b5750776ca : FAILURE in 11m 41s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5763e9b3d914a36970829d8edab7a0a : FAILURE in 23m 32s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/21dddd2ed4ac4563b75d156eeaecd366 : FAILURE in 26m 06s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2f8ff3f6c494e6a969c2b143d3e197f : FAILURE in 12m 54s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/019688f8c7e44ec194078a3b5069e243 : FAILURE in 10m 51s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/497d289f67c541a5af293725c9508c70 : FAILURE in 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"988565a89cf62e1a3eb58f77deff9938e8f219ba","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-28 13:23:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3ab2d8eed58704321006fe0bdb688a3e5d7f0473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 13:56:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2fcc846a734418787cba418dfc4a835 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02d20fba353c4133840d6fb2d5abf48e : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ff2e83d50924786a33e1b3049456b5d : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc78bb7e209345d1b1c7a2717e279912 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/defad909e6694802852480b59564ea15 : SUCCESS in 6m 52s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/9e655e2fd6ce4e63a00a8b0ccafa6f53 : SUCCESS in 5m 01s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a08f5336dc2944159e178b19f0023bcc : FAILURE in 10m 19s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/1bf128bcee1d42158a9399287cd76db2 : FAILURE in 14m 58s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7ab722c87de4b9eacd4d3b355be2c20 : FAILURE in 13m 03s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/3041839512314e16906b9be78bfcc6d3 : FAILURE in 16m 29s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/559664a42c6c48c49afc88be0e1c0b23 : FAILURE in 11m 58s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/0c69b8e240b3411d97428b6a4af49867 : FAILURE in 12m 05s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/2d72401971094b3588fe976934f327e0 : FAILURE in 14m 51s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/ba506a211df0472a8c73d2686d56eab0 : FAILURE in 10m 03s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0693e1e64444eafaefd24eaada18292 : FAILURE in 10m 34s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/d4888000ee4741b2b831472f8e9855ff : FAILURE in 14m 12s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b0a1b22d772425db1c3b63d7e8fa8b4 : FAILURE in 16m 33s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/fe4ff094ebcf4d648add3d9598bc80eb : FAILURE in 25m 53s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2990fec00d9d418782e2490dbae07bae : FAILURE in 10m 44s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/6b6744c6c8f54f1ea38e90b12ad520e5 : FAILURE in 10m 46s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/60edde9d34794cb7b990acdfa1af1f1e : FAILURE in 13m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"076810226453b8233e6992472dc85ab0322400e4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-31 08:39:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d08026c5049d9f76f500e63e9fcf0f7d6c8b2c3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 09:18:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c687e129374be283c3e42952e7daea : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/834b5d3914f8419aa1ec6fd39e739a07 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e7ce5c8c6e340b1b37e6c81034aba92 : SUCCESS in 3m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7830a490d5384892888a84f0d8744828 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f83b69ebaa9045a48cf8beaa2187307f : SUCCESS in 9m 27s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/1aebde8f0e934b1da1c1fb5e9b304a76 : SUCCESS in 5m 06s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54f39117440645c383bd21418f974b16 : SUCCESS in 21m 58s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/3973738ea2a441b59aa6036e0b7d9dea : SUCCESS in 26m 17s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70039ad97521455cb436cf7407e05b31 : SUCCESS in 24m 20s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/a9247e26a5f3414484b06b128236fe69 : FAILURE in 13m 53s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c211d7369a25455d812f05f452bab518 : FAILURE in 9m 54s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/ae405617afc64d969fa5928746ad23ec : SUCCESS in 23m 21s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/a18071f7e18a470a940bad6ef8c49ad5 : FAILURE in 14m 58s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/6cdcd80908b74f3c90c6785770813a23 : SUCCESS in 21m 17s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52ebcbf8fe2041479a256053f53b9074 : SUCCESS in 22m 35s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/a4390710520444618b6717abf543f562 : SUCCESS in 23m 56s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db17440673ab4df1b4ac088fb47f7264 : SUCCESS in 34m 31s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/184889d4e2874b9084857eabbcad1509 : SUCCESS in 37m 06s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f929e8e3cc7945ba90a662b1aa8a421a : SUCCESS in 22m 07s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/6a70a7e41ba44a41aaeb052f23033be6 : SUCCESS in 23m 33s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/d78941a5a0ef4a9fb09efda450e6bec4 : SUCCESS in 23m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c046d49994fa16e6f5901aae079cfa6fe5244457","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-31 12:37:20.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"5beb104b7de37b1b933415a0102e1c9fe6b1ca0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:21:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48370537828f44fdb093d2a4b0072187 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb9689f6c4b245b1b2a355ffc117a623 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be2852f42e0747c2bf1fe34e474079c6 : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e87c090e55b495082bdfd5be44603df : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fdaf8671d544326a047a618406ce173 : SUCCESS in 5m 28s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/b0b4e3527221456b97b39587a5a0fe2a : FAILURE in 4m 10s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fa2dc35560640a588e716650557d514 : SUCCESS in 29m 17s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/ae4970f9a08c4423b8c623de7e63a4f7 : SUCCESS in 26m 45s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a83b432b2d043259070f7635a5408f9 : SUCCESS in 17m 37s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/41374c4f068c4ddeb31027476d5bbcb6 : SUCCESS in 28m 48s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78540d62bc0b4a86ba30b5e6be2f5112 : FAILURE in 11m 11s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/8265b370a861479fa5003130a77d9b61 : SUCCESS in 29m 18s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/239abdf33a58438fb4c762b190bf7cb9 : FAILURE in 12m 13s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/c97d60af68234d49b0092e452843e5fc : SUCCESS in 16m 29s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c81ca418ccb140859d2947ac8b5aed75 : SUCCESS in 20m 21s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/aad1696215e64eaf9540d41630c2e6d0 : SUCCESS in 25m 15s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b75732f5cfd45c8a94f971b7fa2a9b4 : SUCCESS in 38m 35s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/78ffb26ef6d24122b8ccb0cb0254e2a0 : SUCCESS in 38m 37s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/193f820f9e7847a9ab876f0bf4daade5 : SUCCESS in 22m 46s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/cc96f89e34a041afa2da138caf322106 : SUCCESS in 21m 04s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/9446ce27f6114a1192e4099afd49be38 : SUCCESS in 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d2794f2caa86b0ef85953a007721a4fc3ac52d0f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-31 15:50:32.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"8c925b8658ea22194e234545973d5b24b02c5a1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 16:33:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9111c527f93f47a5b92a9080f937b1ac : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9edd87340cb74a72b39024914d8c4da6 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d178bba4ecf64a13ba91f1d5b1544230 : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a020ce49120a47ee8822a6164656ca52 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c78b2f9d81949618b61454ba19164df : SUCCESS in 5m 04s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/725721da15924eabb865c40a808cbfe5 : SUCCESS in 4m 13s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/012b7c03a867401690c950badb1739c9 : SUCCESS in 23m 08s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/9f08cafa0142478d87129f5d32c0c251 : SUCCESS in 25m 55s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcfba75697d14a70948ab1f4fb619778 : SUCCESS in 23m 08s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/f27765ee2b3943b59389247247e03a44 : SUCCESS in 27m 57s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03e4d27d897a46fd935ab3924f656450 : FAILURE in 10m 44s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/8877c81d009f4f47878bca784df8ff46 : SUCCESS in 25m 28s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/25cec1d7794f4c66a02332ed81516e6b : FAILURE in 12m 17s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/f0f8606478494c79893c1fc75c751045 : SUCCESS in 24m 56s (non-voting)\n- bifrost-integration-dhcp-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6b72adc21114d79be6fd84c673a0f11 : SUCCESS in 22m 44s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/805111b1023a42b091d6623c221a73ef : SUCCESS in 25m 56s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5496860be3354de78ce120108337287f : SUCCESS in 38m 10s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/5f2490cf278940ceb1729d0b3f00d527 : SUCCESS in 41m 46s (non-voting)\n- bifrost-cli-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2088c4db183b43ed8aeaf6a39b8d268a : SUCCESS in 23m 53s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/0e25ae7a9cfb4bd9a3e14b8491efe953 : SUCCESS in 24m 01s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/237521c8154d4afaab7fffc3c8042d00 : SUCCESS in 25m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"adab50fcca4d1aabe62cd0c6959cfda3e2d07376","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-09-01 07:31:53.000000000","message":"Patch Set 7: Code-Review+2\n\nThank you Riccardo","accounts_in_message":[],"_revision_number":7},{"id":"d4ccb98e5e5975d2fb10c36cbb2ebd8a3ee8021a","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-09-01 08:23:54.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e4340f5f5a5dd42d5850f1cb630e6618a4a4706c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 08:24:06.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c6a1629cafca424ce23ab2176c7d9b51691e7abc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 09:12:16.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d0e777bd0c04527a66035a0a808c088 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46e2437fdf6a4a989b88688bdcfcfaa9 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a7115b725af40e589ba98c69654a5e0 : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffd0f1897722483da5e350269723ee3f : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d47fff40038841d09d8b484fdb6dd830 : SUCCESS in 7m 08s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/e6be20771fe44095abc361aa50d5a785 : SUCCESS in 6m 58s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e75432dff88d4e259b4cefad773eb944 : SUCCESS in 26m 00s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/56bed69576654041b4ea4f97e513e403 : SUCCESS in 28m 08s\n- bifrost-integration-tinyipa-keystone-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e13bc64cf7a4b32944a4a39f75ec56a : SUCCESS in 27m 53s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/5b1d4d5c206a4382beec2420bfa52b34 : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":7},{"id":"ec1acc7aae8ba1980c74224a2d662f6bdd327ce8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 09:12:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"377a3c81be6f105e6eac287605d2ab7e48dbffa7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 09:15:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32f9e22b72bd4a2e85291789a2c72c84 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2e6ce4165e01e3fb62656b5fb3816ef8688f4aac","revisions":{"725b623ea57dcaa1df82b7d15eb85aee8e523ffe":{"kind":"REWORK","_number":1,"created":"2020-08-28 10:05:35.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/748597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/97/748597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/97/748597/1"}}},"commit":{"parents":[{"commit":"fbdc239b4f3cb11486105fb8651027a99bdaa873","subject":"ansible-lint: fix error 504","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbdc239b4f3cb11486105fb8651027a99bdaa873"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-28 10:04:44.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-28 10:04:44.000000000","tz":120},"subject":"ansible-lint: fix error 602","message":"ansible-lint: fix error 602\n\nDon\u0027t compare to empty string\n\nChange-Id: I3293a554ad4af76b47c73fe820e1809f561aefef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/725b623ea57dcaa1df82b7d15eb85aee8e523ffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/725b623ea57dcaa1df82b7d15eb85aee8e523ffe"}]},"branch":"refs/heads/master"},"ce51c9c2657e1a5741483cb299e4851a647f4298":{"kind":"REWORK","_number":2,"created":"2020-08-28 11:39:13.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/748597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/97/748597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/97/748597/2"}}},"commit":{"parents":[{"commit":"fbdc239b4f3cb11486105fb8651027a99bdaa873","subject":"ansible-lint: fix error 504","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbdc239b4f3cb11486105fb8651027a99bdaa873"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-28 10:04:44.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-28 11:39:04.000000000","tz":120},"subject":"ansible-lint: fix error 602","message":"ansible-lint: fix error 602\n\nDon\u0027t compare to empty string\n\nChange-Id: I3293a554ad4af76b47c73fe820e1809f561aefef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/ce51c9c2657e1a5741483cb299e4851a647f4298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/ce51c9c2657e1a5741483cb299e4851a647f4298"}]},"branch":"refs/heads/master"},"11a495ea595017884fcb220cf98b09297dae6790":{"kind":"REWORK","_number":3,"created":"2020-08-28 12:27:36.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/748597/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/97/748597/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/97/748597/3"}}},"commit":{"parents":[{"commit":"fbdc239b4f3cb11486105fb8651027a99bdaa873","subject":"ansible-lint: fix error 504","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbdc239b4f3cb11486105fb8651027a99bdaa873"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-28 10:04:44.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-28 12:27:26.000000000","tz":120},"subject":"ansible-lint: fix error 602","message":"ansible-lint: fix error 602\n\nDon\u0027t compare to empty string\n\nChange-Id: I3293a554ad4af76b47c73fe820e1809f561aefef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/11a495ea595017884fcb220cf98b09297dae6790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/11a495ea595017884fcb220cf98b09297dae6790"}]},"branch":"refs/heads/master"},"8c4da91d6bfd075000f778723f0d205fa70911d5":{"kind":"REWORK","_number":4,"created":"2020-08-28 13:23:38.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/748597/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/97/748597/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/97/748597/4"}}},"commit":{"parents":[{"commit":"fbdc239b4f3cb11486105fb8651027a99bdaa873","subject":"ansible-lint: fix error 504","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbdc239b4f3cb11486105fb8651027a99bdaa873"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-28 10:04:44.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-28 13:23:27.000000000","tz":120},"subject":"ansible-lint: fix error 602","message":"ansible-lint: fix error 602\n\nDon\u0027t compare to empty string\n\nChange-Id: I3293a554ad4af76b47c73fe820e1809f561aefef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/8c4da91d6bfd075000f778723f0d205fa70911d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/8c4da91d6bfd075000f778723f0d205fa70911d5"}]},"branch":"refs/heads/master"},"55c0ab12981b1fdeaccde7c96797e8996afdea1f":{"kind":"REWORK","_number":5,"created":"2020-08-31 08:39:19.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/748597/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/97/748597/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/97/748597/5"}}},"commit":{"parents":[{"commit":"fbdc239b4f3cb11486105fb8651027a99bdaa873","subject":"ansible-lint: fix error 504","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbdc239b4f3cb11486105fb8651027a99bdaa873"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-28 10:04:44.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-31 08:39:12.000000000","tz":120},"subject":"ansible-lint: fix error 602","message":"ansible-lint: fix error 602\n\nDon\u0027t compare to empty string\n\nChange-Id: I3293a554ad4af76b47c73fe820e1809f561aefef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/55c0ab12981b1fdeaccde7c96797e8996afdea1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/55c0ab12981b1fdeaccde7c96797e8996afdea1f"}]},"branch":"refs/heads/master"},"da98eafad2f62389d65259b8297df09831258109":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-08-31 12:37:20.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/748597/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/97/748597/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/97/748597/6"}}},"commit":{"parents":[{"commit":"5e578b039977247c68b1a6cc0f3d0a717af0df4c","subject":"ansible-lint: fix error 504","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/5e578b039977247c68b1a6cc0f3d0a717af0df4c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-28 10:04:44.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-31 12:37:20.000000000","tz":0},"subject":"ansible-lint: fix error 602","message":"ansible-lint: fix error 602\n\nDon\u0027t compare to empty string\n\nChange-Id: I3293a554ad4af76b47c73fe820e1809f561aefef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/da98eafad2f62389d65259b8297df09831258109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/da98eafad2f62389d65259b8297df09831258109"}]},"branch":"refs/heads/master"},"2e6ce4165e01e3fb62656b5fb3816ef8688f4aac":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-08-31 15:50:32.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/748597/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/97/748597/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/748597/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/97/748597/7"}}},"commit":{"parents":[{"commit":"414053634d145436e383224b689d7b76e09749ba","subject":"ansible-lint: fix error 504","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/414053634d145436e383224b689d7b76e09749ba"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-28 10:04:44.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-31 15:50:32.000000000","tz":0},"subject":"ansible-lint: fix error 602","message":"ansible-lint: fix error 602\n\nDon\u0027t compare to empty string\n\nChange-Id: I3293a554ad4af76b47c73fe820e1809f561aefef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/2e6ce4165e01e3fb62656b5fb3816ef8688f4aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/2e6ce4165e01e3fb62656b5fb3816ef8688f4aac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
