)]}'
{"id":"openstack%2Fbifrost~767384","triplet_id":"openstack%2Fbifrost~master~I76ba1245fcd41ff836ba2cd853bbc4090bd6a5b7","project":"openstack/bifrost","branch":"master","topic":"ramdisk","hashtags":[],"change_id":"I76ba1245fcd41ff836ba2cd853bbc4090bd6a5b7","subject":"Enable ramdisk deploy by default, disable iscsi","status":"MERGED","created":"2020-12-16 17:26:18.000000000","updated":"2021-01-08 18:39:18.000000000","submitted":"2021-01-08 16:17:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"767384-ramdisk","meta_rev_id":"66f67c0658f1fda9cd5af2453779ce1db3ef9b38","_number":767384,"virtual_id_number":767384,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 16:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"all":[{"value":2,"date":"2021-01-08 15:41:25.000000000","_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-08 12:13:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"all":[{"value":1,"date":"2021-01-08 15:42:44.000000000","_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-16 18:12:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-08 12:13:48.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-01-08 15:41:25.000000000","updated_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"reviewer":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"state":"REVIEWER"}],"messages":[{"id":"c99be3da8e913aff09bf278a6526d89316389468","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-12-16 17:26:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa199f391a53ddd56cc25c967e53103d441efccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 18:12:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd799164f6434185a53f2647f587ab1a : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a6b57dba2484add85b10960b9c788be : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5de335a3584841158e8a3141d14afaf2 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99639d92310d4d199ba24a2bc9940120 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23d403d2f6be427daa37542a9aea2cc3 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2394744aa00840a4af5d696f583ec05c : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d7fdd8425194f51b738b2429793791a : SUCCESS in 5m 15s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/37b9273cff994c1e97252d7382986eb4 : SUCCESS in 6m 49s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41c647e7c562447cb80e24557f401525 : FAILURE in 10m 43s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/1b68fa48809e465dbad79dbe08ed6768 : FAILURE in 11m 06s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2921006e0074abbbba46a87400f4b6d : FAILURE in 12m 24s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/743da2fa11a84545b9ec83c658655d14 : FAILURE in 16m 00s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05fe1d3a65334b53a8fb4be56a657e19 : FAILURE in 11m 21s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/c3b010ffc9dc4abe96d4f33b06452ca1 : FAILURE in 10m 29s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/561694a5a370439faddaa870da91a605 : FAILURE in 9m 27s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/d67bf0c6e60b44b38bf06967b10dc458 : FAILURE in 3m 29s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/8201d2e010eb40e9bf401a49d52d4987 : SUCCESS in 34m 33s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/1020581b37304b169d6dcf620408bc58 : FAILURE in 10m 20s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/0807455d38a7434e9dd88eb01f15c888 : SUCCESS in 27m 59s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8dd9d84039640908a4b546fe0e260d1 : FAILURE in 10m 36s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/23568c9ea4c54bd18b493d21c0007a0f : FAILURE in 13m 06s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afc8886986a84cbb9cb5127e0cb8733a : FAILURE in 11m 14s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/988309fb11ed4de8932107ba004c1d44 : FAILURE in 11m 40s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/668e692c0a414c55a7581a27f1443091 : FAILURE in 9m 24s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/858c7eee91dd4923a4fb3ae67f56bbae : FAILURE in 10m 33s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/464535d2dafa424f9af257cf1b3a1862 : FAILURE in 13m 03s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/aeac098d8cd74b6ca62b1001f802738e : FAILURE in 25m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56a5bfea103714bb31fdda2e1c424dc4f6bd11ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-12-16 18:35:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"01fd0162dd48cca26d615d036da6b482380cc41f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 19:18:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70c6d620af0b4c2f9371e31f7bc9c603 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b41561a7cf1f453cb9cb7072127eae63 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f04b71068fca4ec8b53d0d51f91e6e93 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41fa70c7205145af8d9206bde1e2cfb7 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fa53199e2974e6ba41d539963cc3b11 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a9b48c6c0147579d2f15c228bbfce9 : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a67a62f5dcd47d195d08e5ba1225706 : SUCCESS in 4m 28s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/9cc3fc9f04b84cc89ef81b3bde795075 : SUCCESS in 5m 46s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c795cfd9eb047e79f2def1798866b08 : SUCCESS in 23m 42s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/bc7e80a31bcb4b33addf1245ca2abdac : SUCCESS in 24m 08s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95cb9690c11d47feae8ea07821351848 : SUCCESS in 32m 27s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/0011fa9ebbd04a7083bf5037d4bbc21c : SUCCESS in 29m 04s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37459dad806a4f3dab926fccdb96b91c : SUCCESS in 22m 32s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/be44c2049f474426be225873a2f96987 : SUCCESS in 25m 52s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c6d8502afd94a31995d84734b0a98eb : SUCCESS in 23m 17s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/c9c800d36c95472085d0b5a0971a86d4 : FAILURE in 2m 58s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/568d59599a7c4c16a34c85232d8d7863 : SUCCESS in 26m 39s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/b01143782c92478eace2b4215537a21f : SUCCESS in 25m 34s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/12cc8fe8cbb14d53970659651664d310 : SUCCESS in 23m 58s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/609c7955991b44fda3250cfaecbf7766 : SUCCESS in 29m 20s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/d13272d9d36646e1b417f2963a0108bf : SUCCESS in 28m 58s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ab9c5782be545bcba8952f173c81d54 : SUCCESS in 33m 56s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/fea8fe4d03c84caab43eb8f779054cbd : SUCCESS in 33m 20s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b4e81141201495bb67fa8a13bdb58f3 : SUCCESS in 24m 00s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/2ae5499ab1044cc1b4030167b4b7b35a : SUCCESS in 31m 10s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/0ac303c0b24a47168e2b5e375233a08d : SUCCESS in 25m 57s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/9c3519683f464657ab0400d2705d7e61 : SUCCESS in 31m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6fa3e04d9c2111ea78bf79fbbc3504da7fa9013c","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-01-08 12:13:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cc614e5a955694c7651f44c83c8b75775a6bedb9","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2021-01-08 15:41:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLGTM!","accounts_in_message":[],"_revision_number":2},{"id":"1a6d79666673f9baa97d629b34f468c555217f73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 15:41:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"942ee3b3f505e1fb79cade3ca5bd0e70a4622c20","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2021-01-08 15:41:55.000000000","message":"Patch Set 2: -Workflow\n\noh wait. does this need a release note?","accounts_in_message":[],"_revision_number":2},{"id":"1b359d3d27e0466d0fefc13a520971e554a821f7","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2021-01-08 15:42:44.000000000","message":"Patch Set 2: Workflow+1\n\nOh, I missed it. the reno is there.","accounts_in_message":[],"_revision_number":2},{"id":"9657cb2d08a1cc2ba90e01c9aeabebe5c14a7d41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 16:17:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091780d943bc4ae2a417111632443ec1 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82de3ddc5f4f450faecb05301dd3c23a : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e143e2f8ae494126a28f82d722ff9067 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c033363efa0a4279acd55c271ab6e084 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52b4cfeff29242bea95737d5d1445533 : SUCCESS in 3m 50s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/8d89a5561e8b45409a6c2d574f5f9d45 : SUCCESS in 6m 19s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b9684b8d32842a7bc49aa7fd9da5924 : SUCCESS in 23m 42s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/be1823619aaf48a696b38d852a2be9ae : SUCCESS in 24m 42s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91c6af8efd2e461990d7d9ebf7fb652e : SUCCESS in 31m 01s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/cf92ac0d5a4140e797862cb5864bf28d : SUCCESS in 28m 57s","accounts_in_message":[],"_revision_number":2},{"id":"afd81ec930e1d63238daf76ad4e378864a613bd8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 16:17:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8a71a28baef1d0602e67bbb901323c9c746ab99f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 16:18:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c812295211e0487f9cdde97a8f001785 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad8ff3cd691b457c9441f8168412756f : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2},{"id":"66f67c0658f1fda9cd5af2453779ce1db3ef9b38","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-01-08 18:39:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"206149a56b8d07475234048815727c4016fe3271","revisions":{"831c0e0686aedd2e829ffc29cba005a7a4926dbc":{"kind":"REWORK","_number":1,"created":"2020-12-16 17:26:18.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/84/767384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/84/767384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/84/767384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/84/767384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/84/767384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/84/767384/1"}}},"commit":{"parents":[{"commit":"aa727de74d4a88ba4f86137e36c113477592a79b","subject":"Switch to using authentication by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/aa727de74d4a88ba4f86137e36c113477592a79b"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-12-16 17:25:14.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-12-16 17:25:14.000000000","tz":60},"subject":"Enable ramdisk deploy by default, disable iscsi","message":"Enable ramdisk deploy by default, disable iscsi\n\nChange-Id: I76ba1245fcd41ff836ba2cd853bbc4090bd6a5b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/831c0e0686aedd2e829ffc29cba005a7a4926dbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/831c0e0686aedd2e829ffc29cba005a7a4926dbc"}]},"branch":"refs/heads/master"},"206149a56b8d07475234048815727c4016fe3271":{"kind":"REWORK","_number":2,"created":"2020-12-16 18:35:55.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/84/767384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/84/767384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/84/767384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/84/767384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/84/767384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/84/767384/2"}}},"commit":{"parents":[{"commit":"aa727de74d4a88ba4f86137e36c113477592a79b","subject":"Switch to using authentication by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/aa727de74d4a88ba4f86137e36c113477592a79b"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-12-16 17:25:14.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-12-16 18:35:19.000000000","tz":60},"subject":"Enable ramdisk deploy by default, disable iscsi","message":"Enable ramdisk deploy by default, disable iscsi\n\ndosfstools is required for building ESP and thus is moved to\ngeneral dependencies.\n\nChange-Id: I76ba1245fcd41ff836ba2cd853bbc4090bd6a5b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/206149a56b8d07475234048815727c4016fe3271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/206149a56b8d07475234048815727c4016fe3271"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
