)]}'
{"id":"openstack%2Fbifrost~785454","triplet_id":"openstack%2Fbifrost~stable%2Fwallaby~I113f0eae217597ea904e5f94391b7a1317919e90","project":"openstack/bifrost","branch":"stable/wallaby","topic":"fix-which-alias-centos8","hashtags":[],"change_id":"I113f0eae217597ea904e5f94391b7a1317919e90","subject":"Use explicit path for which","status":"MERGED","created":"2021-04-08 18:18:32.000000000","updated":"2021-04-09 10:27:23.000000000","submitted":"2021-04-09 10:25:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785454-fix-which-alias-centos8","meta_rev_id":"68077da1c91dbeddf0eb80df2e8046c11842ad42","_number":785454,"virtual_id_number":785454,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-09 10:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-09 09:20:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-09 09:20:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 19:08:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-09 09:20:49.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"c998da3f694e122ebb17f49aeaaf381b37dda101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-04-08 18:18:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"705f67e31c019e8de229fea76693a5bc568480ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 19:08:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/143a321a0223456cb456dd9cbf323c14 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/115dc4b184144b29aa445f95eff867bb : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaf8ed6b5d3d4b478a7d965378de0936 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/433efc7b927b4c7594ad9402929a9617 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb973f5b293148b5880cd857c0bca6f8 : SUCCESS in 7m 43s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/d14a32298d804c6a8cecb753ce32be33 : SUCCESS in 5m 22s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a61ad7593eb641e1afb5f150db37f15f : SUCCESS in 35m 03s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/b1a2e9b784444393b38a441c5fcf23f6 : SUCCESS in 22m 13s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f63fd6a8fcc84294b8abb4bd82c6872d : SUCCESS in 23m 24s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/71692bb631fd4a2fb83bd0c19518d221 : SUCCESS in 28m 39s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35c25dec0ca84789a16ff1d33214faf1 : SUCCESS in 23m 55s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/1cc9c36f317a46ab883447d1358f48ca : SUCCESS in 27m 26s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0e4bc313fb84458b2a7e914f8b8858b : SUCCESS in 48m 00s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/e74d3f1988a040f9ab3deda8570be734 : FAILURE in 5m 01s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f35193aaf4394f0799058187d0a2566e : SUCCESS in 25m 28s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/488565d294774aa5a144b9d990e9d534 : SUCCESS in 30m 08s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/7f6a8264a8bc481eb9d53892120e0439 : SUCCESS in 29m 37s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/e2d28a1952de48a8b4d741dc066366c7 : SUCCESS in 23m 47s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/7b87b211d3414e5b838be62508f209a3 : SUCCESS in 26m 12s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09aec84d7eb7438a958e745887214966 : SUCCESS in 29m 42s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/40822952019c4139a819d7689f499caf : SUCCESS in 32m 36s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27f28bff1b904801afc08bf92d271dc7 : SUCCESS in 30m 23s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/0aa51d17d859471e83f33605444cf92f : SUCCESS in 32m 25s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79ae75ba0ceb44a2bad405d7a961baaa : SUCCESS in 25m 55s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/34ac3ce80a9a4ec28962f94cb0334cef : SUCCESS in 27m 27s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/836bad8b3f69472a9284078304e876b7 : SUCCESS in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97bc051e698b7c6d57350b1dde8633274c197c8c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-04-08 20:48:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"591e6f0b7e7931839f2d658e56e8fc0a27795e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 21:44:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c65784b86b428990b37e97211b7c77 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5c57a606ec04b7a9b6a28f79505e79b : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e67e9a19875b4e6fab5f1221d6a20d53 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2f22fed359a45fea5dd7f6847c4da5a : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d74f6ada76884c6094017a08978630a2 : SUCCESS in 6m 43s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/da99bed4371042bc915164bab16c0b82 : SUCCESS in 6m 24s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f7c6692916a4847adb3ba3b2025d234 : SUCCESS in 25m 45s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/3bf2c93c960148269ef59ae6ab494de0 : SUCCESS in 30m 37s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41b2c04ab4314180ac7b7802542e7ae1 : SUCCESS in 40m 40s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/c9ce4ff2242248c6a2317ec3ca2bb491 : SUCCESS in 34m 22s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68b984bb3d0742fa975104ed5e09f5ec : SUCCESS in 19m 03s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/9f568f1f94d34fd29c9dabd09f6178e4 : SUCCESS in 37m 23s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03866946d56b40b682e70e7ed387ba0c : SUCCESS in 43m 16s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/188fae0e2bf24911860809339bda38a4 : SUCCESS in 52m 37s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b62bc8423a9d4b3ea6b825d1922c1ece : SUCCESS in 32m 13s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/50e2785b385440e88d06b33c55d6157a : SUCCESS in 28m 00s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/5e2b4972912747c0af7ca02cb377a1aa : SUCCESS in 22m 21s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/0f062dc3bc65444cadcb026f4d16afbd : SUCCESS in 28m 46s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/aba600a89f2e42df80cd8da3867cb0c6 : SUCCESS in 41m 39s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d9d6589f4a9411b8168babb42b5414b : SUCCESS in 28m 37s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/dba1fed89c354e05a50eb3c1dbb3bcef : SUCCESS in 30m 11s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f801495f06634061b7c79e95beac9564 : SUCCESS in 28m 46s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/3f0074d102764b218298a93d1c355acb : SUCCESS in 29m 44s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c22e1b75051430e80a353f85e07f6b0 : SUCCESS in 24m 12s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/f81251cab4e74eef8f8ad0cafe72544c : SUCCESS in 23m 52s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/6e54165cf679465281ba65740c326eaf : SUCCESS in 22m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"879722180c4fcab5aeeb50e458820ed742e32032","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-04-09 09:20:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f0230ff7e3d25c85f4fa52f5ab9c7b393789087","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 09:21:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6052b2d3de563f02ecbdb9dc51c301733e96b4cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 10:25:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e78326a1f52145fd8009ea07df064974 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43afe7bfe35d4805806baeb18d6f614b : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/797c65f54f8f438ab6dbe58aabb0ff77 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e5299182c7416d8edb46c26286f650 : SUCCESS in 8m 12s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/dbd084d579144647904570391681966a : SUCCESS in 6m 31s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecc2078adf2a4403b1ec890b8608be06 : SUCCESS in 28m 11s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/937429bab4324012afbe232e2b463229 : SUCCESS in 29m 13s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ba444b2f5a44fb5a3e9a328afdec043 : SUCCESS in 29m 54s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/2101532f26ed4cc7888d18a46c83ac52 : SUCCESS in 31m 54s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0528c9b0ea445ffb8fe51292c5084d1 : SUCCESS in 43m 45s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/1e27707de54f4d24a3f0267238dd5b2e : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c390040825c1c948d5c216776785e12c65053998","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 10:25:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"68077da1c91dbeddf0eb80df2e8046c11842ad42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 10:27:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72bbe2aeb25447ca1f2692555dac636 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99c1107bced555032ef4b65c32621544480e84da","revisions":{"99c1107bced555032ef4b65c32621544480e84da":{"kind":"REWORK","_number":1,"created":"2021-04-08 18:18:32.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/54/785454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/785454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/785454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/785454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/785454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/785454/1"}}},"commit":{"parents":[{"commit":"9346bfb37490cb065339df44fe75ee0f8597190e","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9346bfb37490cb065339df44fe75ee0f8597190e"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-04-08 18:17:46.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-04-08 18:17:46.000000000","tz":120},"subject":"Use explicit path for which","message":"Use explicit path for which\n\nUse the /usr/bin/which path for which in rpm-based distros to avoid\nissues with any aliases.\nWe need to fix stable branches first because of the upgrade job.\nyay\n\nChange-Id: I113f0eae217597ea904e5f94391b7a1317919e90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/99c1107bced555032ef4b65c32621544480e84da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/99c1107bced555032ef4b65c32621544480e84da"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
